)]}'
{"id":"openstack%2Fkayobe~704612","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Id8cf5f8e50af438e2d8f0077225c38175e38b662","project":"openstack/kayobe","branch":"stable/stein","topic":"wipe-vd-disks-stable/stein","hashtags":[],"change_id":"Id8cf5f8e50af438e2d8f0077225c38175e38b662","subject":"Allow Kayobe to wipe virtualized disks","status":"MERGED","created":"2020-01-28 16:17:17.000000000","updated":"2020-02-07 19:41:03.000000000","submitted":"2020-02-07 19:37:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704612-1581104251567-1e8c484d","meta_rev_id":"881b23ccd452d7992e601f33b0ba071d53a3c533","_number":704612,"virtual_id_number":704612,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-01-29 14:32:08.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 19:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-02-07 17:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-29 17:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-02-07 17:42:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 17:27:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-07 17:42:08.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-07 19:37:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7a634f49c87b3866a51ff188fbd2813ece1d0f7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:17:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ea1a42729e3f89dada7309d04980fe333a1ad5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:10:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da9edec7990b4435927965b8a1123834 : SUCCESS in 5m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4d9c25511a2a42739bd0373c6914cbfb : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89c19da023584e8691384cd289dda65d : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170b814fdf15476b87d95592a6c81214 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87aaec14872d4bfdb455d663139384e0 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a14f951a74b54345a6b7caed18a0a281 : SUCCESS in 4m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7349cf6baf5942edbb752c4e7cb105af : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/632afe282f3341d6a10f556c9b288378 : FAILURE in 8m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2cdda49be5fb49e69af3b0d727cc09ba : SUCCESS in 38m 13s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/911d6f4d54b5475ca0737b82344e978e : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a7009bc133e74ca9ad73514e483740da : SUCCESS in 1h 43m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4c80748b13e24609a33e01826efdc4b0 : SUCCESS in 20m 32s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/63ca2bfe999440a090374f825b5f8d2a : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2db458a11a193ba244b798f1f8e70774df4fe428","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 14:32:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5bc4c5a1da796652f13b3157285ae25a04868e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 16:57:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7c77d98feec42f88ae26b56e36fb284 : SUCCESS in 5m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/81d5512248d44fa1b7a629bda6ea8818 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/353e491175944a4ead6e86e4bbc18e4f : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b7c8d67b4a74f3a891a5bd4fd369c2e : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18c667ad5ebe4f759973e0543ca60020 : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4b7071e4ba49b889b73286c61cf8d6 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/32825710428143a680bfef3bc0b334c9 : SUCCESS in 12m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c05a1bebb2348f28d630a9c8cf498de : SUCCESS in 26m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4feff5d1847f4415913d007dc1c43e03 : SUCCESS in 39m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/723eaa578cfc4071889ede786bd0b8ad : SUCCESS in 1h 08m 09s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/039b480951eb4d169ad85597cc939b0c : SUCCESS in 2h 01m 31s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6a70cc262c724b1cb8cdbfc2790efaa9 : SUCCESS in 25m 34s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a790ac81f6e74bcbb292680009bf55bc : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":1},{"id":"40b64ea6bcd5142043e22047089a38a869ab8099","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-29 17:27:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f68a0f4bdf3810d22e57061c24bfe7286abf509","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-07 17:42:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea0ae8c27a61e796844d9144934604a835b20d15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 17:42:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c68e8954c4005bf448435505a0fec6ecbdbeaef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:37:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c4939fa23f9f4ac2b10780a40e5ceeed : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4beea9f0ce6d4290a3fd5dc8ab1274d3 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bd730b1495422f8b425d90b331528b : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e02a77bd07b4dada61e6f3f3bdf312b : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572f2507c15841dc8b13ce04c350dc9e : SUCCESS in 4m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19ccb3e000774a54a52e1104c4b03a9b : SUCCESS in 8m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8edec38c1dbe46a192471de66fd07ec9 : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/558e714c3f054e6d942b7bdfa0a3e758 : SUCCESS in 40m 10s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a68326a837fa41f39bc6a6a5aff4ac6b : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ca4d489b1b6b44209e87a7aea66d6c80 : SUCCESS in 1h 43m 58s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/52f8c9b3b34d46af84bcd2f6fe18fe13 : SUCCESS in 24m 55s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e08f9f138e7341a48b5e1539370fdd57 : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":1},{"id":"434f57e4c739758f74cbb5d50818e677805491f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:37:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58d00188e8c2f27d1016d3b30f28ca26b2f804f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:41:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d8aaa9e8431470c9b903f3dba98e9d5 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57a6d42c100d2aa6006bfdfe707da700d24ab3f6","revisions":{"57a6d42c100d2aa6006bfdfe707da700d24ab3f6":{"kind":"REWORK","_number":1,"created":"2020-01-28 16:17:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/12/704612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/12/704612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/704612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/704612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/12/704612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/12/704612/1"}}},"commit":{"parents":[{"commit":"0b9a7942604192f767b4c3c14871e5b2cce5bbc5","subject":"Add support for custom Elasticsearch config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b9a7942604192f767b4c3c14871e5b2cce5bbc5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-09-17 17:13:56.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-28 16:17:17.000000000","tz":0},"subject":"Allow Kayobe to wipe virtualized disks","message":"Allow Kayobe to wipe virtualized disks\n\nChange-Id: Id8cf5f8e50af438e2d8f0077225c38175e38b662\nStory: 2002046\nTask: 19692\n(cherry picked from commit ea7fd93692a92b3958136673c79f0c2b54bd8fc7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57a6d42c100d2aa6006bfdfe707da700d24ab3f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57a6d42c100d2aa6006bfdfe707da700d24ab3f6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
