)]}'
{"id":"openstack%2Fkayobe~704614","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ia2c7c151aca75aadc1efc406383b3966d77d4b9e","project":"openstack/kayobe","branch":"stable/stein","topic":"seed-hypervisor-kolla-venv-stable/stein","hashtags":[],"change_id":"Ia2c7c151aca75aadc1efc406383b3966d77d4b9e","subject":"Stop running kolla-target-venv in seed hypervisor host upgrade","status":"MERGED","created":"2020-01-28 16:18:42.000000000","updated":"2020-02-07 19:30:47.000000000","submitted":"2020-02-07 19:26:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704614-1581103607816-a46f390a","meta_rev_id":"8a7b44f44d3cead0a6d1500db70862eb092207af","_number":704614,"virtual_id_number":704614,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-01-29 14:31:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 19:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-02-07 17:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-29 17:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-02-07 17:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 17:28:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-07 17:35:22.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-07 19:26:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c672453ca1dac4170c3b7459556f8a7e34256121","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:18:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b0d5092701e60153ae148157f0cdfd97d96d1815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:35:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfac5767ab8b4d92ba3b40959f3caacd : SUCCESS in 5m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4deff78e10714e1c8e6dd7bbaa9a9c18 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c4b66fc9db4463cb630f337f3275c3c : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699aeeaccbb145d6aadec1230c9140be : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b61407bdcfc14c439cc0d0f1e6e34d6b : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8befed9a57034bba80ecd55e46fa8314 : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/87c3f5407d3a477faeed80e62550671c : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5129a5bb5e06447fb42f5cd0bf2cbaf9 : FAILURE in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0370fc9f9c184569910b2e9591ad1ad8 : SUCCESS in 30m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/681ec945844f402da95c14dbb27493e2 : SUCCESS in 57m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4fe3af7f6398468a921b5a51692f4f3c : SUCCESS in 2h 01m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f2f560f7c98b4ebc9fa91dd8153fbff5 : SUCCESS in 21m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/24c52e5657204d28aa568a98fc023a58 : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"adf336031eb710635be70efc54f723cc02a5f9a7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 14:31:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cf24afc48547d5e3ba7f5123291a457eda4e7c87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 17:02:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5053d679a93b4c188db7736bc7d4e263 : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/776353e9c17d4c77a70728c696efb4bc : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb79ad16c9cf4d7ca3b3e6cd0a541243 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/877e017d26574f6daf2ad07116c24059 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/672d0b98390d43aa95294fd5846d085f : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5746eb26d74545849854c8582da50a55 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ccc20c57a4264bca936a09dd3edbe71b : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d10abf5dfc204c26a3584c25e9ab4c58 : SUCCESS in 14m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7a8f4463e344797ad01c9e09d8464fb : SUCCESS in 31m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6af07c8beb1f49b1852f2ef9f3c26072 : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a5da2db46581462b926f0618365d015d : SUCCESS in 2h 07m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c75281ca0a6f4410920fc7429d7288d0 : SUCCESS in 51m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ac16b4d096cf441c9f8091121bf6d9b0 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a9d9076a2f29dbe9c4bf2ec4afd466d3eeb97ab6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-29 17:28:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1120cefed32555447874b0e01e1c339fb2b2df3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-07 17:35:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"311169985b6c4bf733f06aa5e2f559d8ea02c0b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 17:35:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"734458d27a2fcab425bbcc8bfddee132bc13b1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:26:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7a20b522092147e09165a9f61c4c610d : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faba22000a1e4d23a23ec8b86a13c7bd : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a011818d727410ea7715b9f3bea447d : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c7731f1f48048e4bc50f146e1106bd7 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec10528fd5c403894bfa5970d146d00 : SUCCESS in 5m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d252af2b49ad4cabae81314be02229d7 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e53ece859e5e480287a7c076a98421db : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f751807ce9a840599d4f959be6dbed81 : SUCCESS in 29m 53s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1458db146fd743ec967b12bdc5d600ab : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4baf8707a7d04143bcc5bec16d258ea1 : SUCCESS in 1h 38m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/93df87d758a3410a8df973b7821e32a6 : SUCCESS in 22m 55s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8ee81710e45442a48b781cb14d929cb5 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":1},{"id":"00b4c15b0e93e74eb8c2a186a8ded764addb6fb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:26:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"15c94dbbdfdbd47f0b755929eade5d9d783a2499","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:30:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b644eb4d786b4ec4a90a24077a260ac6 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8009d970bc4c7b8651c80cb142a27e30b3164415","revisions":{"8009d970bc4c7b8651c80cb142a27e30b3164415":{"kind":"REWORK","_number":1,"created":"2020-01-28 16:18:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/704614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/14/704614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/704614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/704614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/704614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/14/704614/1"}}},"commit":{"parents":[{"commit":"0b9a7942604192f767b4c3c14871e5b2cce5bbc5","subject":"Add support for custom Elasticsearch config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b9a7942604192f767b4c3c14871e5b2cce5bbc5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-11-05 11:28:37.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-28 16:18:42.000000000","tz":0},"subject":"Stop running kolla-target-venv in seed hypervisor host upgrade","message":"Stop running kolla-target-venv in seed hypervisor host upgrade\n\nWe don\u0027t run kolla-target-venv during `seed hypervisor host configure`.\nThis playbook is limited to seed:overcloud anyway.\n\nChange-Id: Ia2c7c151aca75aadc1efc406383b3966d77d4b9e\n(cherry picked from commit 0b0b921a62f82bbb50bc06dbe05880e57d364028)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8009d970bc4c7b8651c80cb142a27e30b3164415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8009d970bc4c7b8651c80cb142a27e30b3164415"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
