)]}'
{"id":"openstack%2Fkayobe~704615","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Ia2c7c151aca75aadc1efc406383b3966d77d4b9e","project":"openstack/kayobe","branch":"stable/rocky","topic":"seed-hypervisor-kolla-venv-stable/rocky","hashtags":[],"change_id":"Ia2c7c151aca75aadc1efc406383b3966d77d4b9e","subject":"Stop running kolla-target-venv in seed hypervisor host upgrade","status":"MERGED","created":"2020-01-28 16:18:51.000000000","updated":"2020-02-10 13:59:02.000000000","submitted":"2020-02-10 13:59:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704615-1581343142713-86a07e3f","meta_rev_id":"7b146e0913f39dd0cd2a4666b5afabb225ab4e93","_number":704615,"virtual_id_number":704615,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 13:59:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-02-10 12:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-10 12:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-02-10 12:11:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 12:08:01.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-10 13:59:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a30153e884a6d2c85a3b2ecd94c4a7d35d1e10c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:18:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1037428d26a213a81bb3272d00d732cca01189c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:12:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9f51c0531349495da7fd8f3ae705d610 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7132d9e87cd14f2091256bfbc19ddb86 : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab6e2a7b8f1643f3abd9f28af8d3ffe8 : SUCCESS in 6m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e3b76b362a294cfe808ddf48c774d872 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e44684b3544bfbab8da38b08bde45c : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e78ef7ecf15b4bdfa9f2d5a1930e03f1 : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/83c38679836249efb5dc776f7f7b1fde : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b5d8b9ff34e3499094f278d98233599a : FAILURE in 8m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78d906bbd46b45cb84fbd47f04c6162d : SUCCESS in 26m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/61b9ddf9a601401d8a0075750b512536 : FAILURE in 11m 09s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b88cf053a0864bf188b86143cc5f640e : SUCCESS in 1h 31m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/61bff479e88b42c48a4a8dad037285ae : FAILURE in 11m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e2f5c63bfe204ba1a074db18c47b945d : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5e190bf6fcd4e6737502ab489ff1db4af7f3c00e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-07 21:51:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1d8672cd729b98f7873109d4c821336de2dc31ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 23:29:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5e3ab46f921f4a7e9061340fa5947a06 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f5f8efb4ea64013b3ff8b9e6700b7bd : SUCCESS in 3m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25fcbbf6ef1b41998b5c6bf2360efa63 : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bc18a0e64625480fabc8c34b61dbc7b7 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f9dd570788476c8d381dd98ef3cc19 : SUCCESS in 7m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb5b23f1466948f18b18fe4fbde9c512 : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a9339ccfda934f73be5394d69a39bd1a : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36f7967a3a984ec9af0b1ff1916cd49f : SUCCESS in 13m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73f072a4e0c64c488b135daf30e906eb : SUCCESS in 35m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7efea82d67874dcdbea1950dfbaef8d5 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1b6d6389d68a4e24b2112aed5abb144c : SUCCESS in 1h 36m 53s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dfde3e2929dd4953af838c6867c35408 : SUCCESS in 20m 21s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/248dfbbc308441a6b1ddcec35b9845ac : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":2},{"id":"f35771704eaac85b2a23bfdd3cd1e81af242040e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-10 12:08:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"28e506b89e4ba7d2f4307cab5587bafa1ea3a3bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 12:11:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fd49a078f0811162d56c1dfd436104eaac21dec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 12:11:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c76c3b454f32ebcb80bd44ff070822457e21dfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:59:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f02dfb2705f414d644bee733a3549cdb121e2874","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:59:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d477adec177d4c2d8fc1b836786c6ab6 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d334f5dfba442b957cb62a9f28e150 : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d6ccc124e2aa445f8b6765a404d4996a : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffd5e4a9c81406db1573d8ceb96a47c : SUCCESS in 7m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/508d3cf653db4b8badbbca4947b3ec03 : SUCCESS in 6m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4f5eb0e91f840b496015c8c47cab133 : SUCCESS in 8m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1408cf70895647be8e92b04ed605febe : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9098cabf55c2421eab9ef945af90f618 : SUCCESS in 31m 20s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/fa0d044f54d1426fb0c6737bdbe32222 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3cd5575ff013449b90c24eeeb69ddde4 : SUCCESS in 1h 42m 48s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/496e9adaa7ec4a98b44a4702f08d9ec0 : SUCCESS in 27m 06s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/fceee670d1da44859415b7ba541437a1 : SUCCESS in 39m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e72d7b3d12ed99808c45b0b8ee5c4e7f98e1844d","revisions":{"d2f7ddf8a86827d3fcf9bada0c6155ea01ac9608":{"kind":"REWORK","_number":1,"created":"2020-01-28 16:18:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/704615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/704615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/704615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/704615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/704615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/704615/1"}}},"commit":{"parents":[{"commit":"4652d3aae3c9c7790542283036292622e91b7cf3","subject":"Add support for custom Elasticsearch config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4652d3aae3c9c7790542283036292622e91b7cf3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-11-05 11:28:37.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-28 16:18:51.000000000","tz":0},"subject":"Stop running kolla-target-venv in seed hypervisor host upgrade","message":"Stop running kolla-target-venv in seed hypervisor host upgrade\n\nWe don\u0027t run kolla-target-venv during `seed hypervisor host configure`.\nThis playbook is limited to seed:overcloud anyway.\n\nChange-Id: Ia2c7c151aca75aadc1efc406383b3966d77d4b9e\n(cherry picked from commit 0b0b921a62f82bbb50bc06dbe05880e57d364028)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2f7ddf8a86827d3fcf9bada0c6155ea01ac9608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2f7ddf8a86827d3fcf9bada0c6155ea01ac9608"}]},"branch":"refs/heads/stable/rocky"},"e72d7b3d12ed99808c45b0b8ee5c4e7f98e1844d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-07 21:51:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/15/704615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/704615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/704615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/704615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/704615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/704615/2"}}},"commit":{"parents":[{"commit":"b57ff069d27b9fac4b8bcb9a7454205fca1c63aa","subject":"Fix multiple CI failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57ff069d27b9fac4b8bcb9a7454205fca1c63aa"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-11-05 11:28:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-07 21:51:21.000000000","tz":0},"subject":"Stop running kolla-target-venv in seed hypervisor host upgrade","message":"Stop running kolla-target-venv in seed hypervisor host upgrade\n\nWe don\u0027t run kolla-target-venv during `seed hypervisor host configure`.\nThis playbook is limited to seed:overcloud anyway.\n\nChange-Id: Ia2c7c151aca75aadc1efc406383b3966d77d4b9e\n(cherry picked from commit 0b0b921a62f82bbb50bc06dbe05880e57d364028)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e72d7b3d12ed99808c45b0b8ee5c4e7f98e1844d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e72d7b3d12ed99808c45b0b8ee5c4e7f98e1844d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
