)]}'
{"id":"openstack%2Fkayobe~704618","triplet_id":"openstack%2Fkayobe~stable%2Fstein~If73cebb95c7d0596dc8087112fc19398a113f0cc","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"If73cebb95c7d0596dc8087112fc19398a113f0cc","subject":"Fix use of IPA images in Bifrost without a checksum","status":"MERGED","created":"2020-01-28 16:20:02.000000000","updated":"2020-02-07 20:15:14.000000000","submitted":"2020-02-07 20:10:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704618-1581106229404-b4850ae3","meta_rev_id":"9670183c86a0004efc67102df20a9314df4272b1","_number":704618,"virtual_id_number":704618,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-01-29 14:31:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 20:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-02-07 17:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-29 17:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-02-07 17:46:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 17:24:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-07 17:46:21.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-07 20:10:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"547097d8488ac9f494743aeabf2f9997b5f1f89e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:20:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8d25d84743a356188355dce5787436c47e7fa8b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:37:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b84ba48c7ffc4c9b9b3d31c59bd1f8f0 : SUCCESS in 5m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d1fa373a34324686988420be162f261f : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6661ee6b9f674e7a8ee842cb2c504816 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105c6744bbeb4ab5bc0dd5c5004b52aa : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/41fd446087f840d39664257741171053 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d4cc55dadd442c78626692783390b44 : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/88063b132e974c43965f03822a2ac3a3 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0c335dca421405498e4495838f098ab : FAILURE in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/858e502fe666437ba0d9a331beb78483 : FAILURE in 8m 23s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d3dabf1ea1994fc0b48e1eff2fd04e7c : SUCCESS in 1h 10m 09s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e4cf170602d348a29532baec6aeb72b6 : SUCCESS in 1h 46m 21s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9ccd6d2e1ec648a18959dc5122cec875 : SUCCESS in 27m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/acd8638daa904deca3814903423b3252 : FAILURE in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"291c4ed901b7fb89994e8c29f4017f1ed6160a4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 14:31:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2858588e3385546194b73310fc34b5d398571821","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 16:18:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/987b7fe2ee194e5fab7bd726f3be6d7c : SUCCESS in 7m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bfcf68b9fb544571a2aa62ff0f502b3d : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4214580c7c64e8592d9e02c4f4a3216 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a21dbfd8a4ec4420ad1bd993032d2214 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a02c7222da3d4d75a57df1b353329bb3 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a655ee2385438ca273312886dcbab5 : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9ea11b8894934c57828cc431baa98ae1 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/540e29f1ca4740d283851046397781fe : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9afe88dcff4c4bc2a479fdc75129daa6 : SUCCESS in 30m 09s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/31c78e00804b4566915588d57e025897 : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3bda00b183354267abda87e2d99169b2 : SUCCESS in 1h 45m 00s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3b2a96fa90af41a8bcedbbb6a642e6c4 : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/245293918f3c4424af9f6c9bcb1438f5 : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7434c22b3e25b32e57d7c3690c1827a4cc4e5189","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-29 17:24:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c0c76989d85c66d7121c015528cbcbac4edfa91","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-07 17:46:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf3ed6d92f08eb562e4e369f0e96c090dd9f53ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 17:46:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50016f6c1d753acef27751a32645395d91acc766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 20:10:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90948b854d924e19895db280da0049f1 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa9f6082f01a40e08240c11829e60199 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06cfd601a594774a0bcb4403b8f073c : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/118ee0739d1448b987a5386ac6cb908f : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d83ad22a5e4b80abd533a667bb04c5 : SUCCESS in 4m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c4394e7de7a048dda7ed47c93e7741f2 : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f691788c0641492b9875cbe0d36d7b9c : SUCCESS in 8m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b4888a8c6643435a93cec8f9a1e48100 : SUCCESS in 37m 36s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/827f0cb3d4434325b273e52e81d89b71 : SUCCESS in 1h 12m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a79d139a5dbd4c598f913add8732a624 : SUCCESS in 1h 50m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1802e895b47c4e9398f7887325e67481 : SUCCESS in 23m 06s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/48b01df90ef245f3af3eee9216c9f8b0 : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":1},{"id":"90ecb2eca73f3f2ddac1f63a8edf0a20b0986aaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 20:10:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9670183c86a0004efc67102df20a9314df4272b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 20:15:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508078d83077430eac0343c53acfb974 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e26e6ba8db06fd9c43f2171f7c8a4681f575259","revisions":{"6e26e6ba8db06fd9c43f2171f7c8a4681f575259":{"kind":"REWORK","_number":1,"created":"2020-01-28 16:20:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/704618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/704618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/704618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/704618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/704618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/704618/1"}}},"commit":{"parents":[{"commit":"0b9a7942604192f767b4c3c14871e5b2cce5bbc5","subject":"Add support for custom Elasticsearch config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b9a7942604192f767b4c3c14871e5b2cce5bbc5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 17:54:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-28 16:20:02.000000000","tz":0},"subject":"Fix use of IPA images in Bifrost without a checksum","message":"Fix use of IPA images in Bifrost without a checksum\n\nIronic now provides DIB-built CentOS images at\nhttps://tarballs.openstack.org/ironic-python-agent/dib/files. However,\nif Bifrost is configured to use these, then the \u0027kayobe seed service\ndeploy\u0027 command fails because the checksum files are not available.\n\nThis change fixes the issue by allowing\n\u0027kolla_bifrost_ipa_kernel_checksum_url\u0027 and\n\u0027kolla_bifrost_ipa_ramdisk_checksum_url\u0027 to be overridden to empty\nstrings, indicating that there is no checksum.\n\nVery similar to the issue fixed in https://review.opendev.org/688359 for\nthe overcloud.\n\nChange-Id: If73cebb95c7d0596dc8087112fc19398a113f0cc\nStory: 2006715\nTask: 37407\n(cherry picked from commit 170ebad1dbaff5114873990d53236e2515b1303e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e26e6ba8db06fd9c43f2171f7c8a4681f575259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e26e6ba8db06fd9c43f2171f7c8a4681f575259"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
