)]}'
{"id":"openstack%2Fkayobe~704619","triplet_id":"openstack%2Fkayobe~stable%2Frocky~If73cebb95c7d0596dc8087112fc19398a113f0cc","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"If73cebb95c7d0596dc8087112fc19398a113f0cc","subject":"Fix use of IPA images in Bifrost without a checksum","status":"MERGED","created":"2020-01-28 16:20:12.000000000","updated":"2020-02-10 14:13:17.000000000","submitted":"2020-02-10 14:13:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704619-1581343997554-71032318","meta_rev_id":"a846aec934863915c3bf7e3a95536e38d5377789","_number":704619,"virtual_id_number":704619,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 14:13:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-02-10 12:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-10 12:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-02-10 12:11:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 12:08:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-10 14:13:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8544b2e79153a194468c528157b69d704a6bf33d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:20:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eb0f9b8aefb0110e4d08e53952b9bea2f803c03f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:30:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b6a7e934d68948b3a700c52c32da2d3c : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd84d6884e084a3ea7715f936d1568bf : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ae5c6444de34a4391008fe92108603a : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2cbc121715742739be3b1948aa15c81 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62002c089b374c5581bbca08e221a489 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71290cb2da334a15b280304f8b9c3dcf : SUCCESS in 7m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/323b81add8e84f70adcb1d290e3c6276 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2edfd57a9c0240aea5e34e0e5bd0a77e : FAILURE in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/823f0a24dfb9418c95e9b857c0e4f9ad : SUCCESS in 35m 05s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4806b3ae684f4179bde5935c64abf3e3 : FAILURE in 11m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f33bdea5bffb42439adea2596a35c9f0 : SUCCESS in 1h 30m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/bd911eb5043044f991b4e189f4d5c068 : FAILURE in 10m 06s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/21c814f434b5491a86569b4e04495e75 : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2f9fcd4cd8f00b2112313857ca59e266d5d0a04c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-07 21:52:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c34e588d0412e8b38f1259e5646e3bba7518e56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 23:29:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/908dc0ce73cf412eb8db00a7164b8d2f : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a4390c53c046cb9cf69f947639ee4f : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bc4064b1649429b8b8f9aa64426d981 : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eeccd660e67a4e4ea2070af355a19f3c : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2457b932e812478882e9bb31984be813 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0892832323c54330b0a7524a4d32ff55 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c9acec52aae4bb385b4dd28c42c7861 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5db4e15c71374fa38f8865e181dac561 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c4d3d621989641d7afd96a152211b131 : SUCCESS in 26m 32s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f871fde78150414cbbe5ad9a9ab3d029 : SUCCESS in 59m 11s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f14ba023ccbd427abeb79b27b37ef374 : SUCCESS in 1h 36m 29s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/60367a21c77948959b323f8f15384052 : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/2c625d856dad415fb3b2b61a646820af : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0174dcab9c4a2441b8ffb28418a25859762b9ae4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-10 12:08:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81d9038843381eaffb65824b3b2c6db812524671","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 12:11:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"862defc64c0466f10df752f5031ff0222a2f2740","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 12:11:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac15a432e729940655b918eda29837e82c27c7ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:13:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b5e9b2bff5f7435aa719762333fd7e89 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772c5d6945464f1f9d229850118f1e7f : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3af6ad639fd24f75b934d7b07f0f30f3 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/411bac4a21784fa9bf80ee6ce5359f92 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ebbdf58bc8844c15aa90860880f0b298 : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f9691acc149446f49677bc56c6c3a95b : SUCCESS in 8m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e361202389e04f2ca9884c5a507e1e8e : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7bf114a8a946473ba650d3f549d2f109 : SUCCESS in 35m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ef5f331d11054f9384a3ca59df4ec4ee : SUCCESS in 58m 55s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a6d54e748e454ed8b892ad08e465d4d2 : SUCCESS in 1h 56m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/72d33ca752824cd0a2c55320c47dfad3 : SUCCESS in 19m 46s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f778c75919bb470a8bb9f12dd35a5cd4 : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":2},{"id":"13eee723366dc06536d257f9455c4d57adb2b5b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:13:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"244652e55563c442f1f39f69a08815c2058cad55","revisions":{"ede881d09ebd8ec0025ef9cafd14fb05b10eecf8":{"kind":"REWORK","_number":1,"created":"2020-01-28 16:20:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/704619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/704619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/704619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/704619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/704619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/704619/1"}}},"commit":{"parents":[{"commit":"4652d3aae3c9c7790542283036292622e91b7cf3","subject":"Add support for custom Elasticsearch config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4652d3aae3c9c7790542283036292622e91b7cf3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 17:54:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-28 16:20:12.000000000","tz":0},"subject":"Fix use of IPA images in Bifrost without a checksum","message":"Fix use of IPA images in Bifrost without a checksum\n\nIronic now provides DIB-built CentOS images at\nhttps://tarballs.openstack.org/ironic-python-agent/dib/files. However,\nif Bifrost is configured to use these, then the \u0027kayobe seed service\ndeploy\u0027 command fails because the checksum files are not available.\n\nThis change fixes the issue by allowing\n\u0027kolla_bifrost_ipa_kernel_checksum_url\u0027 and\n\u0027kolla_bifrost_ipa_ramdisk_checksum_url\u0027 to be overridden to empty\nstrings, indicating that there is no checksum.\n\nVery similar to the issue fixed in https://review.opendev.org/688359 for\nthe overcloud.\n\nChange-Id: If73cebb95c7d0596dc8087112fc19398a113f0cc\nStory: 2006715\nTask: 37407\n(cherry picked from commit 170ebad1dbaff5114873990d53236e2515b1303e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ede881d09ebd8ec0025ef9cafd14fb05b10eecf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ede881d09ebd8ec0025ef9cafd14fb05b10eecf8"}]},"branch":"refs/heads/stable/rocky"},"244652e55563c442f1f39f69a08815c2058cad55":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-07 21:52:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/704619/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/704619/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/704619/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/704619/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/704619/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/704619/2"}}},"commit":{"parents":[{"commit":"b57ff069d27b9fac4b8bcb9a7454205fca1c63aa","subject":"Fix multiple CI failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57ff069d27b9fac4b8bcb9a7454205fca1c63aa"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 17:54:00.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-07 21:52:10.000000000","tz":0},"subject":"Fix use of IPA images in Bifrost without a checksum","message":"Fix use of IPA images in Bifrost without a checksum\n\nIronic now provides DIB-built CentOS images at\nhttps://tarballs.openstack.org/ironic-python-agent/dib/files. However,\nif Bifrost is configured to use these, then the \u0027kayobe seed service\ndeploy\u0027 command fails because the checksum files are not available.\n\nThis change fixes the issue by allowing\n\u0027kolla_bifrost_ipa_kernel_checksum_url\u0027 and\n\u0027kolla_bifrost_ipa_ramdisk_checksum_url\u0027 to be overridden to empty\nstrings, indicating that there is no checksum.\n\nVery similar to the issue fixed in https://review.opendev.org/688359 for\nthe overcloud.\n\nChange-Id: If73cebb95c7d0596dc8087112fc19398a113f0cc\nStory: 2006715\nTask: 37407\n(cherry picked from commit 170ebad1dbaff5114873990d53236e2515b1303e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/244652e55563c442f1f39f69a08815c2058cad55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/244652e55563c442f1f39f69a08815c2058cad55"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
