)]}'
{"id":"openstack%2Fkayobe~704620","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ifa766d2cc3ed7077f03c571398072ad5117701c6","project":"openstack/kayobe","branch":"stable/stein","topic":"story/2006814-stable/stein","hashtags":[],"change_id":"Ifa766d2cc3ed7077f03c571398072ad5117701c6","subject":"Use internal API endpoints in overcloud API interaction","status":"MERGED","created":"2020-01-28 16:21:12.000000000","updated":"2020-02-07 19:28:14.000000000","submitted":"2020-02-07 19:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704620-1581103604025-85323f19","meta_rev_id":"ee9bb143969bbf7e297d46969d88ddaed57786cc","_number":704620,"virtual_id_number":704620,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-01-29 17:19:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":31243,"name":"Michael Senizaiz","email":"michael@r-hpc.com","username":"rsys-michael"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 19:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-02-07 17:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":31243,"name":"Michael Senizaiz","email":"michael@r-hpc.com","username":"rsys-michael"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-30 06:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-02-07 17:34:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":31243,"name":"Michael Senizaiz","email":"michael@r-hpc.com","username":"rsys-michael"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":31243,"name":"Michael Senizaiz","email":"michael@r-hpc.com","username":"rsys-michael"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-01-28 16:21:12.000000000","_account_id":31243,"name":"Michael Senizaiz","email":"michael@r-hpc.com","username":"rsys-michael"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31243,"name":"Michael Senizaiz","email":"michael@r-hpc.com","username":"rsys-michael"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 16:21:12.000000000","updated_by":{"_account_id":31243,"name":"Michael Senizaiz","email":"michael@r-hpc.com","username":"rsys-michael"},"reviewer":{"_account_id":31243,"name":"Michael Senizaiz","email":"michael@r-hpc.com","username":"rsys-michael"},"state":"REVIEWER"},{"updated":"2020-01-30 06:37:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-07 17:34:15.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-07 19:26:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29e038aad16905ce1e5961e79c6edd77aba3cc54","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:21:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0cbbb2e1b3f1bb53b73632b2ee94600d01869b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 19:29:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78da6898cd774d27bbc814bdc0710f80 : SUCCESS in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/68b5c85ebdf9408c876b518e8be680f8 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cf53e064d034fa9bdfcf00e0ccc300c : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b82d3cae0824912a0d8df01a3124e9d : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43fa858007ea4462ab8e40fef8fd14a7 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c894e79504ab4487a71d8d4cadfdeba0 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce682613747245d8ac569dbae38410a3 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92454a60957a45f4bd049b03a818f1f9 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e0b138246e94cb6a04521939438301a : FAILURE in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db092dbce23a419ba2f4e95b28504064 : SUCCESS in 36m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b3f041fab4474622af7dfc2b311f7d9d : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d133bea2b88a46fbbc56d7f68f054a43 : SUCCESS in 1h 53m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f059588c0418404795e7be0e0812610f : SUCCESS in 25m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f1b45f2241774132ac714f568d16559f : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":1},{"id":"07b88410633d8c4d738390916fe44be53badecf0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 14:31:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4bb30eb92b1297e2258978a6883c87ae537b6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 17:09:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e50404bf193d4df9a2d5b03d04539b82 : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a0c7fec13a84489485b694f0d1019cf1 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f60be5e6fd94d72b96bfcd159627b6d : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f09d6565acf4fb189f06b6263e05fd3 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c7d21be777744f082400bce90fa8dd2 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da86b5d41f24220b6dace8910c4cdd9 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a8d2e4e4acf4d0ca725c385ad15c47c : SUCCESS in 5m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a1630d90b09470c80d6ece22ecbefa8 : SUCCESS in 6m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e083ec10ad3d4bae8635abc053c403a3 : SUCCESS in 13m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/23a6c174188e4d6d92292b77f0999b40 : SUCCESS in 44m 07s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9cf5dcb7f5424ca09611e249605d4100 : SUCCESS in 1h 56m 46s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2474da212d7546deb1261298097343fd : TIMED_OUT in 2h 32m 38s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/80cb316df3d3471b866423fb6d396eda : SUCCESS in 48m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/17d3067a414c4690848b3440d5743ccf : SUCCESS in 1h 17m 45s","accounts_in_message":[],"_revision_number":1},{"id":"63e79ba94db1c18a60dda925b0ef8c1921be849b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 17:19:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3431283ca29c7bf94c469fe6a5b403a59320c795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 19:04:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86bfd05a2ecc48199aeb1cfa63faa4e9 : SUCCESS in 3m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f5fba65d01d841079600df065f880ed5 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c54d9d4f266479badc4946ac3987180 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6e1e20ec4c406099c6b26c937657f0 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7d75cf603044fada5fc3d4c704310a5 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741284d3ccad46dfac6bbcc3f76dd1ef : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d11446fb20744829affb38e9b99f63b : SUCCESS in 5m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8c9427547a04e55b57280cbaed75cb0 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e6d2d937ee74f588632689249f3cb3b : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/61f90fb4f83b483684cce3544042b905 : SUCCESS in 41m 41s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f2b5e087b80e40489c8fe729e9405790 : SUCCESS in 1h 02m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/467826d1242d45a0a04481abc146de13 : SUCCESS in 1h 41m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d9e10041ce4944ac80fce930a7bb3108 : SUCCESS in 22m 33s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f11819c0300942e8b72f3ff25fc81ccc : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"3f873d68292e35d1abb93f107c0b2dee4410b7eb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-30 06:37:52.000000000","message":"Patch Set 1: Code-Review+2\n\nI checked that this stackhpc.os-networks was not affected by bugs, looks OK to me.","accounts_in_message":[],"_revision_number":1},{"id":"78ef28486e5c5a9828e6271105de885ee5e326a1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-07 17:34:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3c10e2f447809376eadf0fd03e48d6aa6265089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 17:34:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16c08da1ec89d25d8c2f0e3c91f5a4e0f11b3fb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:26:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2ac0df3ea7834db68871330d6ccf842a : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55d3b5e98b8d43879d54592c9b2db132 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60704364328c42649683d41d3412da98 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1a79395ab334aaaa5a2ed2c696e2838 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82e2f54415ee467e8e0c455d51d8702d : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81d99926cb424c70b1dc7d40f6f25731 : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9cdfdbf03eba4e37bc8a55245f178cb8 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7c47ff0c8a854b09bcef8d7a2599aa7e : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aef7d16b789f4c3db2a969697460e44a : SUCCESS in 32m 47s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/62360293a0474194a1bb744c6fcc840c : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/679cd42bc8b0414ea4367c48e11a7df6 : SUCCESS in 1h 42m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6cc249889c444f819f0810c460c08165 : SUCCESS in 23m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/68d0cbf068224c26ac26ec01723a673d : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f1b20f9e5c24da79106697e742f682d232197147","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:26:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"47123930059062a3a722b1b3213a0089dbd5da12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 19:28:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c183115aa7641eba82f87a793710368 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e4b63b82d9c447395b95b85dc6ab329 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a1a6543b65ae70e3237c1183a7c94e8d6babdeb","revisions":{"4a1a6543b65ae70e3237c1183a7c94e8d6babdeb":{"kind":"REWORK","_number":1,"created":"2020-01-28 16:21:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/704620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/704620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/704620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/704620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/704620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/704620/1"}}},"commit":{"parents":[{"commit":"0b9a7942604192f767b4c3c14871e5b2cce5bbc5","subject":"Add support for custom Elasticsearch config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b9a7942604192f767b4c3c14871e5b2cce5bbc5"}]}],"author":{"name":"Kayobe deployment user","email":"michael@r-hpc.com","date":"2019-11-05 22:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-28 16:21:12.000000000","tz":0},"subject":"Use internal API endpoints in overcloud API interaction","message":"Use internal API endpoints in overcloud API interaction\n\nConnections default to \u0027public\u0027 interface when attempting to run\nopenstack commands.  This breaks setups where the network hosts are\nseparate from the controllers.\n\nThis change adds an ``openstack_interface`` variable to select the\ncorrect endpoint to use, which defaults to ``internal``.\n\nCo-Authored-By: Michael Senizaiz \u003cmichael@r-hpc.com\u003e\n\nChange-Id: Ifa766d2cc3ed7077f03c571398072ad5117701c6\nStory: 2006814\nTask: 37374\n(cherry picked from commit 84d17b055eda257395aa5fd38487382d1ac2105a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a1a6543b65ae70e3237c1183a7c94e8d6babdeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a1a6543b65ae70e3237c1183a7c94e8d6babdeb"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
