)]}'
{"id":"openstack%2Fkayobe~705037","triplet_id":"openstack%2Fkayobe~master~I2172a74f4719605f6ec81fadec46ce49f8310a96","project":"openstack/kayobe","branch":"master","topic":"story/2007364","hashtags":[],"change_id":"I2172a74f4719605f6ec81fadec46ce49f8310a96","subject":"Allow OVS bridges to connect directly to interface","status":"MERGED","created":"2020-01-30 14:56:39.000000000","updated":"2020-05-18 11:26:39.000000000","submitted":"2020-05-18 11:23:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705037-1589801002017-ed83af0c","meta_rev_id":"b05b9a82228c7ef1cac7d1f4f6821b9793bdc879","_number":705037,"virtual_id_number":705037,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 11:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-05-18 09:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-05 15:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-05-18 09:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-05-14 09:01:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 17:25:16.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2020-05-05 15:27:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-05-18 09:18:07.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-05-18 11:23:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c50cecff68cff335966421d7f055bd57df4c1b61","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 14:56:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"932a342c0f8b617cf1f308b8be1242f9c68b3be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 16:26:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1168581df908419b88477282e200a491 : SUCCESS in 4m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/15609e3183b445bca681c9a32a5913c4 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af8b908c19e6455aa20b849f58206905 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d9463c13e14ccfa14d3d1ece50d8cb : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9bb9fb2782ef4ebc926adbe5aec26328 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642a6055390b4021b79c8e0f0114f3a4 : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4e86af7a054420f87644edf79814e5a : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39422f66b4cc4e7893c60d1067e64258 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1cb2b5d95687464d8ffcb7f2be2384ec : SUCCESS in 33m 52s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/325c8f7ccac34181a06acb3b82af6729 : FAILURE in 10m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/21fb43301acb40c59ee80142be149b0f : FAILURE in 1h 25m 25s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/40e066476bf24cf496c3147028511400 : FAILURE in 8m 30s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f23415be3a3344e59a70a82f1b288aaf : FAILURE in 29m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f6355f28d48ec6fdf3fd0d0b5f860f2a0b322108","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 17:41:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f23d19829fba7fba0ab6218f918f79732dfff288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 20:01:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30c3f07859ed4208ace8ddc28f33822a : SUCCESS in 8m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c84d16e60001454abddbe38ae094e971 : FAILURE in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa96b164c3264920bf45b5e83585358e : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f613cd844b545369074e2b87e968c74 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e9792931bb5413e84a45386b9cc117c : FAILURE in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd560fd706554706b4d09dd83af03810 : SUCCESS in 9m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b480777ca7b4bbba1f648e9aa86730c : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e674466385434471aa958231f6f0865c : FAILURE in 6m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/edc1c9a551f645378dfe174668cea162 : FAILURE in 6m 32s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/47053706d64041f794bb15a7e24bf4a2 : SUCCESS in 1h 20m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8333f749212746018b94a56f2f2fb558 : SUCCESS in 2h 03m 26s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1ece6e62fbe9412fb4dbab895f7a6b40 : SUCCESS in 29m 27s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0bb1902ec2fa44eeb277618d89c3580f : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":2},{"id":"febf840fcca71c6a1ce4471f3b3652a8c40f2f9a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 17:42:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"339e0a1a0bb72a07e894d004bf9a90deedcd0b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 20:14:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/600f3c373b044c20a1d697d8d484d1d3 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1de68179a2249d49f28b1dd86e12220 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b965f5b861c14761b984dd4dd1b2fc7d : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1fa8ed32c984b7fb7a46486e521e7d8 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a24bc895c92d4531975ef83d0b31d863 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6949bfa2d10c457c88db80ca4fd40927 : SUCCESS in 10m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5eef155371374eb98a370030d0b78fa4 : SUCCESS in 8m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aced5e2d56c64808a579df9cf616e4ee : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/27a3da3eca254114aeeab4fe17ad665b : SUCCESS in 37m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bcb73d1e1a454c7ea94cffd27e01802c : SUCCESS in 1h 13m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/410399fda6854695994580811bf4f111 : SUCCESS in 2h 10m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2f11cd020b3d452a89e7079eb959bbc0 : SUCCESS in 27m 07s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a801209b04544a33925106bc45d9b73d : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":3},{"id":"c6cca85670ac193d4f7ed5254310535da78899dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 12:53:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d2f9517874757e26323c3ef8040e1a4efde493ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 18:08:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28f3945d2bdb426aa74f135d926ba2c1 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430014f364e7455f95bca3db17faa66d : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66911d3c7d8f4251b11581afff9e40bb : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d1f9e70f78d42979524371825c52ffb : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/505f8e44a70f46eab6b0c51eaa126ff9 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fb91efedc044f4a359592993207a91 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7c323e6103947f99168f9fb3e735f20 : SUCCESS in 6m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee97bbb1614e46f5bbbfabab64b5c2b6 : SUCCESS in 8m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/328b95633dbc4291b93622755db2ee1d : SUCCESS in 13m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7a5da0b83924b47bd559b5a937497ef : SUCCESS in 36m 19s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d0ee67a6f1d74f16a635d489db838491 : SUCCESS in 1h 19m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1e02325ef5b84e13a1d3b3f34bea575b : SUCCESS in 2h 13m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0c8f361bace94147933395abf7f47d56 : SUCCESS in 30m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/66a7203f745d4c068799e1adae178706 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":4},{"id":"2935299650d6efc5facf0cefc433ec40cfd904ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 18:20:52.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9f4bff01b8e3a3e5e6f5de7839273f53e070d07d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 18:22:26.000000000","message":"Topic changed from allow-ovs-without-linuxbridge to story/2007364","accounts_in_message":[],"_revision_number":5},{"id":"b8e5f6dc5f61fddfafdcf8bfdd97b82e55e964a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:08:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeef03bdde50413c92d3e41d01fde52c : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/040c247205774b5489bfabb5db4443d6 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d636aad6d7545079594fa7e6329b19e : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb099dc441ed4834bd8a3d939ad83f18 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2bf80f00de742be96b5057cc33efa03 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2167b1ab14744b0a9641b2569ba68537 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/034ae9d88e8b44a48edb3b0b32bdcad0 : SUCCESS in 5m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/159881883e2d48ecb8cded94ece24210 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ec960f8f3e443af90c619b4243450c8 : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bf5c99ab1e37463ca7c5118a0afb02b0 : SUCCESS in 41m 11s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/00eb4385c5a54803b2101abc452cb8ca : SUCCESS in 1h 19m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b181a2f66bdd4bcb92a237ee520e40e2 : SUCCESS in 2h 10m 04s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8393e5800a2d40618d6984580edbfda3 : SUCCESS in 27m 45s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8f182b6f0ca34effa6d86ba453209e68 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":5},{"id":"d403663002eb5c4844bcfbae405e98c961f14344","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2020-03-05 17:25:16.000000000","message":"Patch Set 5: Code-Review+1\n\nI tested this with a bond interface in place of linuxbridge and it appears to work so far.","accounts_in_message":[],"_revision_number":5},{"id":"f07ee273fbf778aac9e0f8fa642cc61fd4dcdae9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 16:18:17.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"e5206c993fe446253a1d2ed2fcb9cbad405a9d5f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-17 09:51:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c9ebdc7e8bc6570f006d15894e949750e93dc31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 11:48:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b1d747d3b84428ba5cc0262f519ee2d : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8227299dca46409787c63625f7ae2865 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ab57c9d7ee849d2a4102c065538319e : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95d4a58e70e44ee1b5655714715be51e : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19107140756c4bfcb30c509df90fde9a : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/829b8d5241774dd9b544e9ffc3ea3589 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0491f198a3fc45c6bcbff6accea525a9 : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c3addba388b24ba3aeb293b77e484405 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f3590a983a34da99ded261c547425a9 : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f89ee1f33e14ce6a7617090aa58efe4 : SUCCESS in 39m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7122c3f8ee814fa8870d617b82f838a7 : SUCCESS in 1h 08m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/95ba8f0ba68f4ebdb3742d4b39688fdd : SUCCESS in 1h 55m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2f4c32a2d030458ba449c0a256ec52fd : SUCCESS in 19m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/89fac7bfaa6c43f48f9b973552efde4c : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":6},{"id":"de6225320f7c3d2bb54e721ded60b6d5dfc8465d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-05 15:27:04.000000000","message":"Patch Set 6: Code-Review+2\n\nLooks OK. I would quite like to try it for real…","accounts_in_message":[],"_revision_number":6},{"id":"67f5c70c8b8e896a146fee1381630483ec9a3b2a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-14 09:01:18.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"a70bbca432d53ff3f8121c9f90209f3d45cb3639","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-18 09:18:07.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"99088337733dd3609ecaf08e0e8547cd38fbed79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 09:18:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1e5a5cce1cfe40f3e832d63485c7318dad2ecdaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 11:23:21.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e78b3f10774c7b96eb2cd844a80c94 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf26b8feba9c4679bb230c3e5b559de7 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/201ce84bb9cb4098bdc35479fcb47ec4 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f2531841204bdb9792a911fdc88ac4 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3516c962e01b4829801e85f0f64e47ac : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9145dfbfa7f54470906892378f229dd0 : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/13201d35a8864d56817716620562ecaa : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0814ecf06a8e4beb90064ac852f114f3 : SUCCESS in 33m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7d41401c23b144d49864b15ff2d6aaa7 : SUCCESS in 1h 13m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b2a06139c3524f548c34a9db4e1609ac : SUCCESS in 17m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d3b8000d18e3469c897f2954b160762a : SUCCESS in 2h 01m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d952f95ae3a34066a6fd2ec132527d36 : SUCCESS in 20m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fe6d97bb8c444578b41ab7e503037a34 : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":6},{"id":"094f19fd9fe1de52ca837d0a0273e3f0210f84c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 11:23:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b05b9a82228c7ef1cac7d1f4f6821b9793bdc879","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 11:26:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99fb7f2669124716833eb48cc7024090 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac56010c8d65434ebec8a62175666414 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"073499f322386287fa497a63f63596702b3e80ba","revisions":{"12d74f22f86395900f6243e1aa3731c143a7a5ad":{"kind":"REWORK","_number":1,"created":"2020-01-30 14:56:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/705037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/705037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/705037/1"}}},"commit":{"parents":[{"commit":"2056983a565bc92c3943f93b89124e8e70b6afeb","subject":"Merge \"Fix overcloud provisioning due to Bifrost IPA rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2056983a565bc92c3943f93b89124e8e70b6afeb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 14:55:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 14:55:51.000000000","tz":0},"subject":"WIP: Allow OVS bridges to connect directly to interface","message":"WIP: Allow OVS bridges to connect directly to interface\n\nCurrently we require a Linux bridge to exist between OVS and the\nphysical interface. This is necessary if you want to set an IP on the\nnative VLAN of that interface, but that is not always the case.\n\nThis change allows the physical interface (or any non-bridge interface)\nto be plugged into OVS.\n\nChange-Id: I2172a74f4719605f6ec81fadec46ce49f8310a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12d74f22f86395900f6243e1aa3731c143a7a5ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12d74f22f86395900f6243e1aa3731c143a7a5ad"}]},"branch":"refs/heads/master"},"09bf028963233dc2e80c87466dfc64d041ffd0ba":{"kind":"REWORK","_number":2,"created":"2020-02-10 17:41:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/705037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/705037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/705037/2"}}},"commit":{"parents":[{"commit":"2056983a565bc92c3943f93b89124e8e70b6afeb","subject":"Merge \"Fix overcloud provisioning due to Bifrost IPA rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2056983a565bc92c3943f93b89124e8e70b6afeb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 14:55:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 17:41:35.000000000","tz":0},"subject":"WIP: Allow OVS bridges to connect directly to interface","message":"WIP: Allow OVS bridges to connect directly to interface\n\nCurrently we require a Linux bridge to exist between OVS and the\nphysical interface. This is necessary if you want to set an IP on the\nnative VLAN of that interface, but that is not always the case.\n\nThis change allows the physical interface (or any non-bridge interface)\nto be plugged into OVS.\n\nChange-Id: I2172a74f4719605f6ec81fadec46ce49f8310a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09bf028963233dc2e80c87466dfc64d041ffd0ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09bf028963233dc2e80c87466dfc64d041ffd0ba"}]},"branch":"refs/heads/master"},"3db4f3ee30d98bfdf2e9d074409ca5e33e789eab":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-13 17:42:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/705037/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/705037/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/705037/3"}}},"commit":{"parents":[{"commit":"2fcdff101f888ac310969e4e5e0f2e409288a71d","subject":"Fix sphinx8 script used by pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fcdff101f888ac310969e4e5e0f2e409288a71d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 14:55:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 17:42:50.000000000","tz":0},"subject":"WIP: Allow OVS bridges to connect directly to interface","message":"WIP: Allow OVS bridges to connect directly to interface\n\nCurrently we require a Linux bridge to exist between OVS and the\nphysical interface. This is necessary if you want to set an IP on the\nnative VLAN of that interface, but that is not always the case.\n\nThis change allows the physical interface (or any non-bridge interface)\nto be plugged into OVS.\n\nChange-Id: I2172a74f4719605f6ec81fadec46ce49f8310a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3db4f3ee30d98bfdf2e9d074409ca5e33e789eab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3db4f3ee30d98bfdf2e9d074409ca5e33e789eab"}]},"branch":"refs/heads/master"},"286766f6b2b3ac887f28cd82af8a99df9461f7d7":{"kind":"REWORK","_number":4,"created":"2020-03-03 12:53:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/705037/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/705037/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/705037/4"}}},"commit":{"parents":[{"commit":"208c46e72c9be38bbb87dd44060a641c54163eb7","subject":"[ussuri][goal] More cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/208c46e72c9be38bbb87dd44060a641c54163eb7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 14:55:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 12:52:21.000000000","tz":0},"subject":"WIP: Allow OVS bridges to connect directly to interface","message":"WIP: Allow OVS bridges to connect directly to interface\n\nCurrently we require a Linux bridge to exist between OVS and the\nphysical interface. This is necessary if you want to set an IP on the\nnative VLAN of that interface, but that is not always the case.\n\nThis change allows the physical interface (or any non-bridge interface)\nto be plugged into OVS.\n\nChange-Id: I2172a74f4719605f6ec81fadec46ce49f8310a96\nStory: 2007364\nTask: 38920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/286766f6b2b3ac887f28cd82af8a99df9461f7d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/286766f6b2b3ac887f28cd82af8a99df9461f7d7"}]},"branch":"refs/heads/master"},"d9eb9eec2652f4e12cc0208df77338b0cf37ff37":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-03-03 18:20:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/705037/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/705037/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/705037/5"}}},"commit":{"parents":[{"commit":"208c46e72c9be38bbb87dd44060a641c54163eb7","subject":"[ussuri][goal] More cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/208c46e72c9be38bbb87dd44060a641c54163eb7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 14:55:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 18:20:49.000000000","tz":0},"subject":"Allow OVS bridges to connect directly to interface","message":"Allow OVS bridges to connect directly to interface\n\nCurrently we require a Linux bridge to exist between OVS and the\nphysical interface. This is necessary if you want to set an IP on the\nnative VLAN of that interface, but that is not always the case.\n\nThis change allows the physical interface (or any non-bridge interface)\nto be plugged into OVS.\n\nChange-Id: I2172a74f4719605f6ec81fadec46ce49f8310a96\nStory: 2007364\nTask: 38920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9eb9eec2652f4e12cc0208df77338b0cf37ff37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9eb9eec2652f4e12cc0208df77338b0cf37ff37"}]},"branch":"refs/heads/master"},"073499f322386287fa497a63f63596702b3e80ba":{"kind":"REWORK","_number":6,"created":"2020-04-17 09:51:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/705037/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/705037/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/705037/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/705037/6"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 14:55:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 09:51:13.000000000","tz":60},"subject":"Allow OVS bridges to connect directly to interface","message":"Allow OVS bridges to connect directly to interface\n\nCurrently we require a Linux bridge to exist between OVS and the\nphysical interface. This is necessary if you want to set an IP on the\nnative VLAN of that interface, but that is not always the case.\n\nThis change allows the physical interface (or any non-bridge interface)\nto be plugged into OVS.\n\nChange-Id: I2172a74f4719605f6ec81fadec46ce49f8310a96\nStory: 2007364\nTask: 38920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/073499f322386287fa497a63f63596702b3e80ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/073499f322386287fa497a63f63596702b3e80ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
