)]}'
{"id":"openstack%2Fkayobe~705173","triplet_id":"openstack%2Fkayobe~master~I9164bbacef4b6c7dfc60e91a0f040b6a525067e4","project":"openstack/kayobe","branch":"master","topic":"fix-requirements","hashtags":[],"change_id":"I9164bbacef4b6c7dfc60e91a0f040b6a525067e4","subject":"Cleanup tox.ini deps","status":"ABANDONED","created":"2020-01-31 08:24:15.000000000","updated":"2020-02-25 13:57:19.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9356eb1b841002034e0cfc22bbb7b28fa8ff26e3","_number":705173,"virtual_id_number":705173,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-02-25 13:49:04.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-23 19:45:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":-1,"date":"2020-02-25 13:46:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-23 17:57:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-23 19:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-25 13:46:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8aa7490af1908aad79f11ad4e285e9cd65ccd187","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-31 08:24:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f338041a525166435ce132168451998bbc4f8c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 09:07:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b845825a4fac409eabc722e71bf4fee1 : FAILURE in 3m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/87fb0b514c5f47c88da171eb6f747086 : FAILURE in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3543f10cbb84f4ba6bf186bde9f2749 : FAILURE in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56cb13006f1944718cf239655c1de946 : FAILURE in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f057f8441c6c49dab24288809d2341e5 : FAILURE in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2dcf510c4b443d393624d264c187e5e : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b4839ff61e243d080c69719ff422766 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84d1ceb5ba834d46ab3e19fbc3139c50 : FAILURE in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff6c09fa929f40559d57768df8a09b6a : FAILURE in 4m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd76a48c08dd4159ab10af123af99c0a : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0f999454445bd4312abb2b03cae12480d788eed8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-31 09:24:39.000000000","message":"Patch Set 1:\n\nFailing since it uses upper-constraints.txt from master while the rest here expects train. This works locally. Let\u0027s wait until the repo is in good shape again.","accounts_in_message":[],"_revision_number":1},{"id":"a422682c5d9fc73d8e1b221207db157a282f00d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 16:42:51.000000000","message":"Patch Set 1:\n\nThanks. Kayobe needs a little TLC due to changes in kolla ansible for py2 drop. I\u0027ll get to it next week.","accounts_in_message":[],"_revision_number":1},{"id":"fc00cc5684e9647f956ea6f25050e9f0caa47c22","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-13 19:51:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae3c9253673b1844181a1011f229aee4182693b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 20:32:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05d0f9be8cfd4b0482395afe33044372 : FAILURE in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec0e45fe8d174c1a99bebe83cc433b39 : FAILURE in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fa209af2f4b4b24b52a6233bc0aae88 : FAILURE in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55d4739843d24d54bda770c177bfe20e : FAILURE in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2d8c904d159446d92f7d42105eeecd4 : FAILURE in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f9159b50b6042f2806128768e4eff1c : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b45fff9a41814a0192f5b84272b9cdda : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65a8cd57d35348b5acab5a58c91019ff : FAILURE in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d3f1e3ca41f4f7097ba12d1841b2428 : FAILURE in 5m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/636873ceb8284198b2fd88d0c19963b7 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d16e75d5d16d956562d116fff0017f81ceefeaae","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-18 07:02:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3284ba3b0df3ee6fa04b9672ba8d65483711e9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 07:48:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7639cb50c6d4858a1795f69c25cfe10 : FAILURE in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5282aaa11c47deb265c9b6455df1f4 : FAILURE in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55eccdce0b5d49c1806b3f013a73b59f : FAILURE in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f297e6723c974819a7407068efcdd044 : FAILURE in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/548d0f680d444458be87e1142354e3b5 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740cdd73ad284424b6842b1d344cc518 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9465ed092db64863ab468beaf9b81fef : SUCCESS in 5m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e71f283e6c949e78b0d478023613c0f : FAILURE in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45d21d87c34f4f54a92a8ed96c818d02 : FAILURE in 5m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/659b27aa358243d992e539ff2c11862e : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":3},{"id":"5fd0b11a4eee4ed54a9a0255e798d6e38272d50d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-20 05:51:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7aca3f33aa8acb8c6a7b7f83c6f238b8a803bfbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 06:25:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afbbc39da90044289e396370502c66d4 : FAILURE in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79fb5e3eb3894316931ccb9b7dc3d4d6 : FAILURE in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb737b7c31fd42079fa920af7a5aa175 : FAILURE in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6649dfcd797e45f4b7c492f69abb15f9 : FAILURE in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ecd23a23bea43dd9fc0aad2903f83c4 : FAILURE in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ea657201f041fe9fbc7e4265dc2a1e : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d510be5bffb34693bf91df79859e9024 : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d1f634d884f41c0881161f22fac641f : FAILURE in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af42784804f44b868c7427bf1e2b897f : FAILURE in 4m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/74d8ebff63b84be893475511bc7bb8d5 : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":3},{"id":"23a07a252fd1d54ed3ff7e42972a2dec6916d726","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 11:37:18.000000000","message":"Patch Set 3:\n\nLooks like we need to bump our cliff requirement. I\u0027m not sure why it\u0027s affecting this patch though. I\u0027ll look into it.","accounts_in_message":[],"_revision_number":3},{"id":"73504f03def87204afe576248937fbba8f53ae10","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-23 17:57:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\n-1 for venv comment; pip one needs investigating","accounts_in_message":[],"_revision_number":3},{"id":"f5a0c03458e8cfc69fbcbd1849ec845e7ed93a10","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 19:08:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"64a79c9cc244d5783f05935bab0e62fe0f21ea7e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 19:08:25.000000000","message":"Patch Set 3:\n\nlet me readd install_command...","accounts_in_message":[],"_revision_number":3},{"id":"7bf60d0d13f49ff6192c278244616b9ecfc63bd2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 19:09:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a70f73df704590f3752da8cedb11722b4f8682bb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-23 19:11:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"07c9c6d7d5e4761dab7e1778997b0a61cf3a1aa4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 19:17:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a4093e01f02d459ebe3a599143a2768fbab247c0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-23 19:19:48.000000000","message":"Patch Set 4:\n\nYes, this is wrong practice here.","accounts_in_message":[],"_revision_number":4},{"id":"4562eb7d65fd41a8673bbf42f778324c8b069bcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 19:45:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00ae6782ae2a448a8874d5a30a171b7f : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb03eb00a7ce4152bc8afe96c4964af9 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a89f76c1330f4719b58ef025e1c36cf5 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25f58d9b70bf4fa29d9c2037edcc7a8e : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa1280a60faa45e08244606ad1102554 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1ffa811389e45d4ae26488fa3ee11e5 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e2a100c7ffb436e8b2b79d70dcfe796 : SUCCESS in 5m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/679a9663be4b4633a4b96a92a9d0b8ad : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f19b311f3e4a4ba295150d6bce1cf80d : SUCCESS in 10m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2dd186bc6b32493a8098d0f525ed2b85 : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":4},{"id":"9438b57a65f27cff7a4c7b3f3b9fb200de4dfbc1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 13:46:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"329420e34b3a52c6461bb54ec72a49476dfe3391","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-25 13:49:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"831b64104a00d4523b2822ed8d47a479e06728c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 13:52:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bc4c91012663adbe518633148519d9fea98f2f35","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 13:53:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"56cdff7824df29bf700c8f4546529bff7d7d315b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-25 13:57:19.000000000","message":"Abandoned\n\nLet\u0027s abandon, not worth the remaining bits. Thanks for feedback!","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e48a94d8ef59fba86a919d9fcf79597bea5e0763","revisions":{"22a8307ec38d725055b9633132bd4607ad4f5948":{"kind":"REWORK","_number":1,"created":"2020-01-31 08:24:15.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/73/705173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/705173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/705173/1"}}},"commit":{"parents":[{"commit":"d5a3c74b1628aa77182a57c974d1fabb00b3f1a2","subject":"Update docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5a3c74b1628aa77182a57c974d1fabb00b3f1a2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-31 08:20:18.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-31 08:24:05.000000000","tz":60},"subject":"Cleanup tox.ini deps","message":"Cleanup tox.ini deps\n\ndocs does not need the normal requirements file - and some environments\ndon\u0027t need docs. Remove the extra lines.\n\nUse docs deps in releasenotes build, no need to duplicate the\ndefinition.\n\nRemove venv deps, the testenv ones are fine.\n\nremove install_command, this is the default.\n\nChange-Id: I9164bbacef4b6c7dfc60e91a0f040b6a525067e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22a8307ec38d725055b9633132bd4607ad4f5948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22a8307ec38d725055b9633132bd4607ad4f5948"}]},"branch":"refs/heads/master"},"e7f649e04e406244ae8456e936010a2b49089717":{"kind":"REWORK","_number":2,"created":"2020-02-13 19:51:20.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/73/705173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/705173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/705173/2"}}},"commit":{"parents":[{"commit":"030ede06e8a99eeb51ae9f0386601849948530b5","subject":"Fix error logging of dev/functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030ede06e8a99eeb51ae9f0386601849948530b5"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-31 08:20:18.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-13 19:50:40.000000000","tz":60},"subject":"Cleanup tox.ini deps","message":"Cleanup tox.ini deps\n\ndocs does not need the normal requirements file - and some environments\ndon\u0027t need docs. Remove the extra lines.\n\nUse docs deps in releasenotes build, no need to duplicate the\ndefinition.\n\nRemove venv deps, the testenv ones are fine.\n\nremove install_command, this is the default.\n\nChange-Id: I9164bbacef4b6c7dfc60e91a0f040b6a525067e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7f649e04e406244ae8456e936010a2b49089717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7f649e04e406244ae8456e936010a2b49089717"}]},"branch":"refs/heads/master"},"5bdb6cbf369f1ecf88f5d70cb5f7d4fb4dbf2586":{"kind":"REWORK","_number":3,"created":"2020-02-18 07:02:20.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/73/705173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/705173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/705173/3"}}},"commit":{"parents":[{"commit":"c9474bf9b00a9feb00d7339e4ccb873cf699658f","subject":"Revert \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9474bf9b00a9feb00d7339e4ccb873cf699658f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-31 08:20:18.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-18 07:02:05.000000000","tz":60},"subject":"Cleanup tox.ini deps","message":"Cleanup tox.ini deps\n\ndocs does not need the normal requirements file - and some environments\ndon\u0027t need docs. Remove the extra lines.\n\nUse docs deps in releasenotes build, no need to duplicate the\ndefinition.\n\nRemove venv deps, the testenv ones are fine.\n\nremove install_command, this is the default.\n\nChange-Id: I9164bbacef4b6c7dfc60e91a0f040b6a525067e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bdb6cbf369f1ecf88f5d70cb5f7d4fb4dbf2586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5bdb6cbf369f1ecf88f5d70cb5f7d4fb4dbf2586"}]},"branch":"refs/heads/master"},"e48a94d8ef59fba86a919d9fcf79597bea5e0763":{"kind":"REWORK","_number":4,"created":"2020-02-23 19:09:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/73/705173/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/705173/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/705173/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/705173/4"}}},"commit":{"parents":[{"commit":"c9474bf9b00a9feb00d7339e4ccb873cf699658f","subject":"Revert \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9474bf9b00a9feb00d7339e4ccb873cf699658f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-31 08:20:18.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-02-23 19:09:14.000000000","tz":60},"subject":"Cleanup tox.ini deps","message":"Cleanup tox.ini deps\n\ndocs does not need the normal requirements file - and some environments\ndon\u0027t need docs. Remove the extra lines.\n\nUse docs deps in releasenotes build, no need to duplicate the\ndefinition.\n\nRemove venv deps, the testenv ones are fine.\n\nChange-Id: I9164bbacef4b6c7dfc60e91a0f040b6a525067e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e48a94d8ef59fba86a919d9fcf79597bea5e0763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e48a94d8ef59fba86a919d9fcf79597bea5e0763"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
