)]}'
{"id":"openstack%2Fkayobe~705180","triplet_id":"openstack%2Fkayobe~master~Ieef73950f89e4dfb727ddc59ef2750d9b81f3c58","project":"openstack/kayobe","branch":"master","topic":"logdir","hashtags":[],"change_id":"Ieef73950f89e4dfb727ddc59ef2750d9b81f3c58","subject":"Fix error logging of dev/functions","status":"MERGED","created":"2020-01-31 09:37:34.000000000","updated":"2020-02-13 19:29:35.000000000","submitted":"2020-02-13 19:27:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705180-1581622074314-86a07e3f","meta_rev_id":"8c170b039b12be9bc0db78165fd8db187632ab61","_number":705180,"virtual_id_number":705180,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 19:27:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-02-13 15:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-02-13 15:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-13 15:32:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-02-13 15:32:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 10:09:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-07 17:48:05.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-13 15:32:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-13 19:27:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa03f6cac2ebd7a442b69c0c21bcc2be8e419d23","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-31 09:37:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5699a1d4ea46ea6bb54ab39967e8909148754934","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 10:09:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"147ddd5bd49f2cc87444f85fdade8b05776aa554","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 11:25:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ea8ab9c9a1247c39d53bdfcf20c8f29 : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c62d088a69174ac5aac4de999c7bd345 : FAILURE in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a7afc0956be49839af5b34237e3e676 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e8956e7ea547eb9c3605ba92cb4e75 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e477e7cc1924104aa25842badd7bbdd : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c907dc91f54fcd9f8c0eec6a0937aa : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/637802b74b4446fd861e48a43fdba5d4 : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ccb855628fb43d7898dea33df418cc7 : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/793c50bbd8a648ff9f1f2d19edff8e44 : SUCCESS in 31m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/98c9d189780445b79f614ff672f75f46 : SUCCESS in 1h 16m 11s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/74656fb4052547ef9fca58792c34157c : SUCCESS in 1h 45m 29s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/85c7ed0a0fd347f086a1072882353572 : SUCCESS in 24m 21s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/26576ccf38cd4f13bf80ef2d02c58441 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5e1656366873511875d4f5a9e3b76f4b2b417e9d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 14:17:01.000000000","message":"Patch Set 1:\n\nrecheck\n\nLooks like a transient mirror issue, py35 passed locally.","accounts_in_message":[],"_revision_number":1},{"id":"e4f7abb72ad9add84459c9b9f74f2304b964f684","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 14:19:32.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks for fixing.","accounts_in_message":[],"_revision_number":1},{"id":"6b0e00943a8e8457157067bb5a37346ce565b543","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 15:10:31.000000000","message":"Patch Set 1:\n\ntransient u-c issue, fixing: https://review.opendev.org/705184","accounts_in_message":[],"_revision_number":1},{"id":"ef08213295168c0c87bb7b2a0d2587fcd74df8b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 16:24:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ed8ba5e977b4146b4b998b8609054bd : SUCCESS in 5m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/98ee788a64d243c4a0a44fdc4dd2fb3b : FAILURE in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3d51dcd5f44c2f8acf8cee51ddaa5b : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c3bf7f92d1246fd8ea5184245eb54ce : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c02664d5ab641f88773122f81dd3cb0 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559dfe7c62fa49afb4824ab78ea1beee : SUCCESS in 6m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67204eb7c5594a6b8187587a83802a09 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/023c444fcf594aa8993ca9be7eb61281 : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d3a04456974445309043d3274e5ee02b : SUCCESS in 31m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7b8d7b36a55043d2bcd4068e78b04a1b : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/68cd1f7a964b479eb40fe4ff14863789 : SUCCESS in 2h 02m 22s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dcf7c9b27ffb473f83c383f1c399fb48 : SUCCESS in 25m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ff59a88ad89f4f1b8afee7702bd24864 : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":1},{"id":"51e02c9a750e385d40a5b30ca8ad80e53c511c00","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-31 18:08:11.000000000","message":"Patch Set 1:\n\nrecheck - zipp fixed","accounts_in_message":[],"_revision_number":1},{"id":"4f9f974a3067652fc22aa20b058d701b270d97dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 20:11:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d620199271c4180b1eec3d203d8863e : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66cbb4b3aadc437780dfab866965e074 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c40b202daec548eeaa12144b20717906 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0e0e92ede94ee38e8ac043457fcf1c : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82a26124ad1a49caa9e8f8f2cacde58a : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0f27e5c5394ab59bab4f53567af4a2 : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d72ef4732af45da8abf99765821e11c : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd9971be8f494c3f9f7b2e175c3c44ba : SUCCESS in 10m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4a89987a60c4b8ea68dd40132629de3 : SUCCESS in 38m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7b23956d7ac547d3835892e2a115972c : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2a02ff83f1d947ecb0c58d682d23ba4f : SUCCESS in 1h 58m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2959fe6c5f464bdb90f65d05e19ff29e : SUCCESS in 23m 41s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/26581f72cc3b44dab0d9f87a142f0e08 : SUCCESS in 43m 29s","accounts_in_message":[],"_revision_number":1},{"id":"aa71a37d935a0c730001b40b2ad13c1b34ecbcda","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-07 17:48:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4ca41e471938e57331d39412138889e71b19944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 17:48:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7001b8f42496a8ca05af0420a89782290d21bb17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 20:10:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e57dc495e0341c2a2e2b796d688b6fd : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa8b81b48ec7449eb2b4b49e444a2cb3 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe1400f037e4ed998c698cc3cd8bd7d : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/041129cfd68f460189867b1cd35a67b3 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/855d79a560154d30bb26e45c1224c7b0 : SUCCESS in 11m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3458a91af8c4c98b55f9488c88d9c20 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1ed6a9f974a486498cb5fcb3880689d : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5fb4c6efd6ee4190a170b9d975f2ada4 : FAILURE in 5m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/be2d0a07668b49a0add00c9b96ccef4a : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3a495f81263d481684e251aa432b0420 : SUCCESS in 1h 45m 54s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/25b999c7520c4eb893f4dd7f018d6db1 : SUCCESS in 22m 58s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/cf38fd359d99412b981089a607530cd2 : SUCCESS in 36m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d969a4a2d28bec7d4aa8b64fcc9ab2dee7fd718e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-07 20:30:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb71b44e3a6373283010ab10f253fc74881222b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 22:18:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc454f6a90214e4cb59c620258be3e1c : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3c12f2c82d1c482b9da87c87e91edbe0 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/441f5b4664f94cdf849cd7cc866489eb : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90642f073e4f4a8f9632e6562a8b7836 : SUCCESS in 7m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02119d530a24444587ffc5725ffab177 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a1a000ec8745a0a6cc7f9a3e6f8f20 : SUCCESS in 8m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e217a80f731f41528ef256bec65ca47c : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c2a31dbc07e4b4aac31b21c0f2f2309 : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b909102d1df24d16a7ae74053b4f6740 : FAILURE in 5m 16s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8f9bed53f78040929cef7fdb873e25f3 : SUCCESS in 1h 12m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/af1929d9f6e74e098f141ed3eef21dba : SUCCESS in 1h 45m 42s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9b9c3270ca004606b953f2cfda02af48 : SUCCESS in 26m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e78d2f242784452eb33baef94ed57fec : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c1a1e770a434a620dbf1b9ae6ef5db46399f7b15","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-08 09:09:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae3324f9f3b7fa8e8d059470f17eaea227adc41d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 10:57:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99801af8998f40a099a5f92c06246a10 : SUCCESS in 3m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/28daa502962c412db713d1b4a7ba8b5c : FAILURE in 2m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69b7bb2b84e44522abcdc9814db0a599 : FAILURE in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/955e6b1358bc4cf4a5a41d8e1315c86e : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dffb216d33c4428b8bcad7cbb7b364dc : FAILURE in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0f09bc838d407b9aa84b5bbe84b211 : SUCCESS in 5m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c7694fc604649869baa15bfc1b3c67b : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8326605808e34568b7bc2e4d8e6f06a9 : FAILURE in 3m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2c517753d7b4d4f938600524fad28b1 : FAILURE in 3m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/44e697b1caf84fa1ac1ab928accaa2cf : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5f62deaaeeee4a47810731f1631396ac : SUCCESS in 1h 44m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7b8e20d3514c45e894043a7a0197a4a9 : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/91ab4b980c3d42b29ce018c7fdfef7d8 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4a2703840afa8489e7909dc11eecf844169fa050","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-08 15:58:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7f3a728abce13e7b68dcdbc622c9b21e429c928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-08 17:46:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d9f736676134e37a9ebe5e55199ab91 : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e5ae24b590e94701a7139ff1abdb1b5e : FAILURE in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c7a4f53b7324694b5a4bbb054421f93 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c8f91948b5d48588b6b85e47f122777 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54d590d724dc4960996a14bdc33efe8d : FAILURE in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f71bd10fef64ad98b39485e9dfba5e0 : SUCCESS in 6m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/57d41e300dc24b8d94cae0655551bb4e : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a14b1ec84534631b886269b6624f7a0 : FAILURE in 3m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c522f700d3824069baf75460e5084751 : FAILURE in 4m 17s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/96e7d66536534d77a15d883d0822a654 : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0b285aefb8194c259a3b4e40f267368d : SUCCESS in 1h 44m 26s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c8c26303358b4f618c3ac4514c7c28c4 : SUCCESS in 21m 08s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f36f958161434b9dacba061291b8ca5e : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"99dbcf7a5190e1bbb62e82ea093d1c0b14ff2dae","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-08 20:37:01.000000000","message":"Patch Set 1:\n\nneed to scrap py\u003c3.6","accounts_in_message":[],"_revision_number":1},{"id":"739e5651936afa7bcd5882af7f1f483e40b9893a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-08 20:37:19.000000000","message":"Patch Set 1:\n\n(aka do not recheck until it\u0027s fixed)","accounts_in_message":[],"_revision_number":1},{"id":"08dfaa5b31af64ccb898d0a2d327a27587a0f60c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-12 15:49:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e4aebfaa1c321d608d13517eb8be8f998bfec8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 16:04:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa24037542974f7ca4a5f720a725f8c1 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061afde8176d4697aa2ad1955fd4ef47 : FAILURE in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c596812e913740abbedba3a4e3a42b3a : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47116c556a414bf494899e1773c60351 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc5425e50bd44ba28d183faf7d8c68b3 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170312e551b04891a7a625e647b19889 : SUCCESS in 7m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0c743a4269e24c09b0adf003f6184282 : FAILURE in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f04b08f896e5475fab1f92e0a602ffae : FAILURE in 6m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fcd6e6220604a748bb7d14a3ede69e4 : FAILURE in 8m 41s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d230fce9843b4a759e47486a388e4131 : FAILURE in 7m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d519b7e144434ec4a6aa850e10b97009 : FAILURE in 6m 28s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/789534f12d6d433b8b679bcb722dc546 : FAILURE in 6m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c40ab21ba83241f9870f3d167ec95fa8 : FAILURE in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e6d11c809c2a6e1a04b37ec9d71eda392941e8c4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 15:32:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ba09bacac3dd8c46ce957fc5c189962f7dfb2fc2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 15:32:55.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"149e9ebe52f9cf28e3e9fd2c8a27a7b0fb0ae0d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:39:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f767004dd342402c9a861ae8328dcc67 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e0ae5f9aabc4d249bf70f53fe1f4d87 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/884bf0ba6f814dc29df31c2fa874adfc : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afcd5e1f9c2e4d01ae74fbc40635d134 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/058602b9f4a14ecf8de5a59891cfb7f4 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f6cefaf7f654b8689dc9cb44005c20e : SUCCESS in 8m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be9eda73416b4fc7be78a2fe67114ded : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9eb2f0761043415b8b7399c716903b20 : SUCCESS in 9m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f9cc6941a4384bec848d1ab8bde40b93 : SUCCESS in 38m 26s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a3b30c9b23c344c1b7500877f6871083 : SUCCESS in 1h 16m 32s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/02586b2397854f82a86e741186b0926a : SUCCESS in 2h 05m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5e1c367ce54643ebb6c5fdc38064ad4f : SUCCESS in 26m 11s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8a8824717e9e4d2780ebcbbb55ddc32c : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":2},{"id":"dcda17e49bdb9b6c3c62b9f1be81e84d6ab96347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:39:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e2587ad05b97b2ea51164b7deafa87b501e4eb0f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 17:40:33.000000000","message":"Patch Set 2:\n\nGot there in the end :)","accounts_in_message":[],"_revision_number":2},{"id":"14592aaa7154890b51b3c0708aba3de0b1eb16d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:27:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2aab31f93edf52516b688036c68c58468a743d83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:27:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c7cde6334a40f39b7979fd8f1e8ff4 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/613ff9f047aa491eb8c2d31651d7cc94 : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6104409997d476290726e435498da79 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38baeb78919146b29d05ac17bf41e4fa : SUCCESS in 11m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/101ae177412a4b9b899f0dcfd197e450 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a605b026ca7b48998c37f05922c424a7 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d67ec240c6ce417a83ecc453f1efcc32 : SUCCESS in 31m 42s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1cfc2d435cac492e8bb49fa8226b3164 : SUCCESS in 1h 16m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a96f12b7556e404e823305897c90f06b : SUCCESS in 1h 43m 16s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/36d626a0f4254d4fb0d3706449167f22 : SUCCESS in 28m 07s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/63e04f974e154c0ea8b60a1870492581 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5e74b6752b653a24159940822235452cb4a2865b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:29:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/253b00dd3ead4de3a796c2eb40507718 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"030ede06e8a99eeb51ae9f0386601849948530b5","revisions":{"7ff5c173a3790d6e45c9eb97bdddf59132df6393":{"kind":"REWORK","_number":1,"created":"2020-01-31 09:37:34.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/80/705180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/705180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/705180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/705180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/705180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/705180/1"}}},"commit":{"parents":[{"commit":"d5a3c74b1628aa77182a57c974d1fabb00b3f1a2","subject":"Update docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5a3c74b1628aa77182a57c974d1fabb00b3f1a2"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-31 09:34:24.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-31 09:36:08.000000000","tz":60},"subject":"Fix error logging of dev/functions","message":"Fix error logging of dev/functions\n\nIn case of failures in kayobe-overcloud-centos, the error message fails\nwith:\nkayobe/dev/functions: line 569: LOGDIR: unbound variable\n\nExample:\nhttps://zuul.opendev.org/t/openstack/build/ce1fadc3ee6d4842a599da57a670cc18\n\nThis can be reproduced with:\nset -eu\n\nif [[ -n ${LOGDIR} ]]; then\n    echo \"LOGDIR set\"\nelse\n    echo \"else\"\nfi\n\nFix the error reporting with assigning an empty string to LOGDIR by\ndefault.\n\nChange-Id: Ieef73950f89e4dfb727ddc59ef2750d9b81f3c58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ff5c173a3790d6e45c9eb97bdddf59132df6393"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ff5c173a3790d6e45c9eb97bdddf59132df6393"}]},"branch":"refs/heads/master"},"030ede06e8a99eeb51ae9f0386601849948530b5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-13 15:32:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/80/705180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/705180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/705180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/705180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/705180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/705180/2"}}},"commit":{"parents":[{"commit":"2fcdff101f888ac310969e4e5e0f2e409288a71d","subject":"Fix sphinx8 script used by pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fcdff101f888ac310969e4e5e0f2e409288a71d"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-01-31 09:34:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 15:32:49.000000000","tz":0},"subject":"Fix error logging of dev/functions","message":"Fix error logging of dev/functions\n\nIn case of failures in kayobe-overcloud-centos, the error message fails\nwith:\nkayobe/dev/functions: line 569: LOGDIR: unbound variable\n\nExample:\nhttps://zuul.opendev.org/t/openstack/build/ce1fadc3ee6d4842a599da57a670cc18\n\nThis can be reproduced with:\nset -eu\n\nif [[ -n ${LOGDIR} ]]; then\n    echo \"LOGDIR set\"\nelse\n    echo \"else\"\nfi\n\nFix the error reporting with assigning an empty string to LOGDIR by\ndefault.\n\nChange-Id: Ieef73950f89e4dfb727ddc59ef2750d9b81f3c58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030ede06e8a99eeb51ae9f0386601849948530b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/030ede06e8a99eeb51ae9f0386601849948530b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
