)]}'
{"id":"openstack%2Fkayobe~706853","triplet_id":"openstack%2Fkayobe~master~I212be3c32f075acbe17a28a973637e1a0474fb11","project":"openstack/kayobe","branch":"master","topic":"grafana_post_configure","hashtags":[],"change_id":"I212be3c32f075acbe17a28a973637e1a0474fb11","subject":"WIP: Apply sane defaults to Monasca configuration","status":"ABANDONED","created":"2020-02-10 14:36:12.000000000","updated":"2020-04-27 08:43:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc47a6bccd8babd053ba290958f776724f7893c1","_number":706853,"virtual_id_number":706853,"owner":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-10 17:09:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"all":[{"value":-1,"date":"2020-02-10 14:36:54.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 17:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28c5167425014ef064ced530b388c53ce5992546","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-02-10 14:36:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"658132fddafd6fa309faddb3e88238670789f19b","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-02-10 14:36:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"55ca4c90bb5a4ef9e187d005a80a1dc31810085b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 17:09:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c14a860c96bd43488d095d55b2a3a631 : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bcd6076a206a49f1be754675b8f63782 : FAILURE in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a725b6b3b16949949531b13c228ae4e9 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b561dfc5d85643cea0dc74df0a12f52b : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/132374ef9a2148d59f65445885819964 : FAILURE in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ab2711f4334750afcc0f8a97495580 : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f8514745cb3479c9d4828a0f32f4afb : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf2e2b7d83154022b7d0eb1fd30088a7 : FAILURE in 2m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/11dff40fd5e14b3887df93c372ba5509 : FAILURE in 4m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1519c2c75df44da18451f7219f0fd11a : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/93d30e495e5d4d20b87ac7559b6df503 : SUCCESS in 2h 25m 20s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/249f041f060f4288a20c2682e67c9d6f : SUCCESS in 27m 46s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d4c471c710e34330825bb220e3292441 : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d912405a63df5918087671e924b2d1a3750d7df3","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-04-27 08:43:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93d4eb40b950f00d86d3229ffc737b5d908c0ecc","revisions":{"93d4eb40b950f00d86d3229ffc737b5d908c0ecc":{"kind":"REWORK","_number":1,"created":"2020-02-10 14:36:12.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/53/706853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/706853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/706853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/706853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/706853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/706853/1"}}},"commit":{"parents":[{"commit":"d5a3c74b1628aa77182a57c974d1fabb00b3f1a2","subject":"Update docs building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5a3c74b1628aa77182a57c974d1fabb00b3f1a2"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-02-10 14:34:01.000000000","tz":0},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-02-10 14:34:01.000000000","tz":0},"subject":"WIP: Apply sane defaults to Monasca configuration","message":"WIP: Apply sane defaults to Monasca configuration\n\nChange-Id: I212be3c32f075acbe17a28a973637e1a0474fb11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93d4eb40b950f00d86d3229ffc737b5d908c0ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93d4eb40b950f00d86d3229ffc737b5d908c0ecc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
