)]}'
{"id":"openstack%2Fkayobe~707137","triplet_id":"openstack%2Fkayobe~master~Ieb1577f859d388a0f326cb6933c953f486dfca15","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ieb1577f859d388a0f326cb6933c953f486dfca15","subject":"Stop gzipping logs in get-logs.sh","status":"MERGED","created":"2020-02-11 13:05:14.000000000","updated":"2020-03-11 13:50:31.000000000","submitted":"2020-02-18 19:31:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707137-1582054314623-bbef5f2b","meta_rev_id":"138a39cc1495ba2d27f19ebb0a7ecee14e3f7404","_number":707137,"virtual_id_number":707137,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 19:31:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-02-18 17:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-11 13:29:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-02-18 17:38:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 13:29:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-18 17:38:12.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-02-18 19:31:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a7da7108618b77ed669a4c088df37c65eedd133","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-11 13:05:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce977d3c83af95a7cc2bc405b9fcb8dc7cf66bad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-11 13:29:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fab7f320233e7ac05ed0013e4b7e6793e687c0c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 15:14:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b20dbf27622644b7ba20bfd1e3cf909c : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96d5f398edb54275b3624ba4c4055c43 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7793e761123d4a8bb93fbeee26773a44 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59928c605df74580ad7d4c4a63ba4d13 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8def8c08af9847a9b161387a08a0c5a6 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6566b77611044589747f5d6e035954a : SUCCESS in 6m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9bc6f0c40d35405b9f4b519faa439783 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa9d17304ed7452284af4bfb02c60ce6 : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/182d6c69cee841788161689bc8d8cc1f : SUCCESS in 36m 58s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/96adc0c2ed8a4acab7922ebeb70eabab : SUCCESS in 1h 13m 57s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e5f6a04bbfbb451faa796515fdab01ee : SUCCESS in 2h 02m 46s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/29f07721f16545f99738ffeb213d7937 : SUCCESS in 24m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a9b74e568ba84349bca0cd3a69bb1749 : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":1},{"id":"483c2836f4a1b11f336b15c0d85e63f36ddb297a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-02-18 17:38:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3cdbef97420345d084cb55961f30b351fed2ef96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:38:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6a31101243c7bc29ba2f4a4e1cdb8dc3623167a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:31:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f29962a19664a14921ed4537b036dc5 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/684979bbca3f4fdc97289b09c48eef35 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd3e80c49fe94b76b9aec68308356114 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1cc8fd20044ffab9cbba1c29f8de3b : SUCCESS in 10m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/985778a5edc14ff3898ca148842d5ec9 : SUCCESS in 7m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e29e331849ee4616863ca77e52f95c4f : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02ed54056e60456c8533f9bb199b6064 : SUCCESS in 42m 10s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/78cdc7f87ba740718738c107a07b865a : SUCCESS in 1h 12m 28s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8bfaa30b97634c70bf846c7dee9c1984 : SUCCESS in 1h 51m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/be862f381fc6457184266a724b26f67b : SUCCESS in 23m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/2eba0b5eb11d44df866c1d3ae256ede2 : SUCCESS in 33m 18s","accounts_in_message":[],"_revision_number":1},{"id":"db666bb14ef555b401deb14ab5f74270d6a9497b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:31:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"52f1c8c3a274c2d501cc2e5dcf36a60a93411dc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 19:35:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345833dd96d742bea743e0121cec97ec : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1},{"id":"dbe68402367378ac2552d8f16dd0b7ed27c66e73","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:50:14.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 1be05c17361c3cd237025ad7e25af4a25f11bd58","accounts_in_message":[],"_revision_number":1},{"id":"2af94fd4caf994f69a74d291a14ed8e50ab99d08","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:50:22.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 8c74a9d0800ebb982820567eb06f125099e0ae89","accounts_in_message":[],"_revision_number":1},{"id":"138a39cc1495ba2d27f19ebb0a7ecee14e3f7404","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:50:31.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit ef232aa14e5a128bb73b5a01fe58fe07234e854c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c657ead4f25148d7f6dd5825ae11c29827ed3201","revisions":{"c657ead4f25148d7f6dd5825ae11c29827ed3201":{"kind":"REWORK","_number":1,"created":"2020-02-11 13:05:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/707137/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/707137/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/707137/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/707137/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/707137/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/707137/1"}}},"commit":{"parents":[{"commit":"c0a1216afa0982098647d8ca87efe821805fadd4","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0a1216afa0982098647d8ca87efe821805fadd4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-11 13:04:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-11 13:04:37.000000000","tz":0},"subject":"Stop gzipping logs in get-logs.sh","message":"Stop gzipping logs in get-logs.sh\n\nAs per [1] we should stop compressing the logs sent to swift in order\nto get them back readable via a browser.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-January/011875.html\n\nChange-Id: Ieb1577f859d388a0f326cb6933c953f486dfca15\nCo-Authored-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c657ead4f25148d7f6dd5825ae11c29827ed3201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c657ead4f25148d7f6dd5825ae11c29827ed3201"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
