)]}'
{"id":"openstack%2Fkayobe~707604","triplet_id":"openstack%2Fkayobe~master~I6316e8959cff987e4e97280889e1038a9537ed32","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I6316e8959cff987e4e97280889e1038a9537ed32","subject":"CI: Test SSH connectivity to deployed instances","status":"MERGED","created":"2020-02-13 10:22:39.000000000","updated":"2020-04-22 00:31:31.000000000","submitted":"2020-04-22 00:28:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707604-1587515324389-16b58591","meta_rev_id":"478dacdd437d553abe922eefd7f298d2cb77e689","_number":707604,"virtual_id_number":707604,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 00:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-04-21 20:01:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-25 19:46:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-16 13:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-04-21 20:01:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-04-15 16:18:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 19:46:05.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-16 13:36:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-21 20:01:56.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-22 00:28:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0430c3557893a693614475cd23d1367918a525b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 10:22:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"786366ac11552909f55f1b26388aabd949f2d8a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:33:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fabe456117484c7e89928220ccb1bd01 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4acf361329b4620ba218ae06bc7be3e : FAILURE in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f8b10e6152e46a190255169e5468fd5 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd7ea5647c0b4d1398d3e2d1d46b591f : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1afeb3e14c2343cf803378e02bebe06f : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5fe2d78aa864308a3b7f4855a4d9736 : SUCCESS in 7m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b208cbb60ebe4ea29dda8ac45c6ef73d : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0672b7b134c4b4388bbb3411f0637c9 : SUCCESS in 12m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93f21b5f92ae4ab1b3c4bddd55984677 : SUCCESS in 38m 40s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d1576699b4c345d6b6e975b84aff67e3 : FAILURE in 1h 20m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/63f31360b7744c39bd06f6da16848aa1 : FAILURE in 2h 01m 54s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/55addc98f7a24fdd871660a78e53dd6c : SUCCESS in 22m 07s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ddfc9e5f40334771a14d916a6bb871fd : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6ebff54c8d0c1704fcd698801ddb2562529aca7e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 12:46:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"925b5ea22abcc0da81812d890b4e694162ef8769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:30:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b89409ab82045edaf783508104168b8 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36cb537b79a5442f8fe2e6518f70f102 : FAILURE in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81b0726f4ad7424abce3fdf5195e85f5 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/828248f0332e4cffa2efeec4d39f053b : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/478df3dfa178456886f651630f663dfa : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d456504db434e7b8e86eec812f9d5e6 : SUCCESS in 7m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4407590ee7c449b9af4c0a71260646e : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd9ca1b972ba49119e2170a0aa938419 : SUCCESS in 9m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f2057f166e54b9ea233b11f64c33468 : SUCCESS in 36m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/64c64256f85b438c9762300d12d72c25 : FAILURE in 1h 05m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6f9b023c4f724efa88712d6c3263f432 : FAILURE in 1h 42m 22s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/43ee8991d4504824861954328bf97dff : SUCCESS in 19m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/afdc66766181460bb1b13369a9ba6974 : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4f0c7687cb47b8e28d3f7b4455943545f985b556","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 14:30:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"010f8ebd280dd92a593d1713f9093bfd2c71a9c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:21:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e1a5221545146c4bbdd5c30ebf6a940 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49bed8de43174e86b2332f176c4225fa : FAILURE in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54a04e8a195a4ae7b16cd4cb755ddd41 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aadf4b028b164d1ea888c55ddaea89d8 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abd881b5b52a44b089cf7d2a93d66e0a : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eda61077c8d413f967cfabfc2a9ec16 : SUCCESS in 8m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7fa92d8dca1b4d2e891777d4fe9967ae : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e45522917b1410087ba18749c26f314 : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/401378a9f4334e8791d8b2cd3eb8571e : SUCCESS in 37m 19s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2a3c5b14e26b4e46b0da32f17f6e450a : FAILURE in 1h 10m 57s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0d0d5f8567894a0d968f2aebb5b6d059 : FAILURE in 1h 47m 26s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b918cbf4ccbe43d4aa30b40430eadbb0 : SUCCESS in 22m 06s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ad28e93628254e1688c2e3021daeb087 : SUCCESS in 36m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8518677bbf0262e0d0ea39cfd793797867ba8a2f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 16:37:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"11d13662b2a49a33177ff8e8ee00b1c0473053d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 18:46:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b36bef4b7a34a069005ca9f2263d248 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a5d1477cdd845bea878e3e8ca018b06 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a372ad82d21455fb60aaf689d839e3c : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/031e238ec9a4473da567babd18e655e6 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/def62c67472d408082e07053ff0754fe : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/742fca2923e34a7aa99947372f4b7ea5 : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21c0e424c7b845f4b6525765895f067b : SUCCESS in 6m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5131dd3d882f4729b2a623b60ad6459c : SUCCESS in 12m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c90eee0a24b4770a027496592d67e4e : SUCCESS in 36m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7246a912f9144c3cb05afa14a728429c : SUCCESS in 1h 19m 40s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/579b0f7984cd4281b335ce31f76d6826 : FAILURE in 2h 02m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1702b46bb1fb45908a3178609093dab0 : SUCCESS in 23m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4225dd63685d4bc7a7c2459dd5f3d0a9 : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ec07cdd3ed7ce9391702d7fe41699ec08f720b9d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 19:48:31.000000000","message":"Patch Set 4:\n\nUpgrade failed because init-runonce was run using previous release which does not set the provision-net network as external. Might need to depend-on a backport of this change.","accounts_in_message":[],"_revision_number":4},{"id":"ea367b712853d25c279273e7107668f168ff0e8b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 16:56:19.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 5e41dfe539800e2eda4c74f3b204b18eab8a0dd1","accounts_in_message":[],"_revision_number":4},{"id":"3ecce17179494ff3616a867333a954b6c496ab41","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 17:01:54.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"eaf99113f9d68c93ff88384a215a465b231c14aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 18:35:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa9b7852ce24415dbc2cc1b3881a6679 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01e76fddf4c945e0a04bfd98263feb55 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/041104bdab264f57b0d6483c951c7d40 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/916894cd9ab6421c84522054867cf358 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00084d6c3d5e4282af2f953d8e80adc5 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303bc98a2d4d46778540f6fe4749ee6d : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b50bff3f351845b398f72c25f36e86e8 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b22915a4b9fd4e1ba55eab19d64ec835 : SUCCESS in 13m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e799d82972054548b2958bf41d04ea40 : SUCCESS in 31m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a9172eca08b24c8bbfd6ed98421b7fc0 : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/761282f9fba844fc8f856df5f0daaf0d : FAILURE in 1h 31m 00s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/932c615d39ef433ca7743d86a016685f : SUCCESS in 26m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b41b6f5438a14ae7bb0e694335155bfa : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":5},{"id":"fca55175697b1e03b9fa7e985eca43c78fccbbcd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 11:19:01.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"3f07a14477188b6e389724a0055608ed4a05df43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 13:10:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa3f5a3378944c0e8b1dfe9cf16ceda0 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097796975426472ca999e986aea8c477 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be653b2f103a44e79fc11227e8929dad : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21374f6cd12a430db521d879be6bad88 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fde644dd7554d518f6cffa74db33d5d : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c855e963434471b679bd4aeefed84e : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/86b2cb8736054b32af484ee7fd4ed09d : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d56efd66342444b9902c337c99375be : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d98bf40f3fc34a20bba019b453c3f8e1 : SUCCESS in 39m 50s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c34ce8afbffe474f9357429e4d2e10fd : SUCCESS in 1h 11m 03s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/683f6221aa4144fda65eca0b7eeff3ff : FAILURE in 1h 46m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/555203b37cbd4be29281d134a97c374e : SUCCESS in 22m 51s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4fbef66eb6fc491c89d7535438988e1d : SUCCESS in 44m 07s","accounts_in_message":[],"_revision_number":6},{"id":"f5fac47a61bd07f5967145d0b3e39a4865f543d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 13:55:56.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"405fd7a3c1f16385758c9a32f4d79671575bfbdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 16:05:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a684665afbe4ff384425dfa7d784e12 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3dce04f09a64115a420f459027cab04 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3952abb98a284d929dec1e49d9cf2120 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c01ec1ffce844e794683a4996284a67 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f04bbd590bc2494390979cbfe7035cdb : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d985a19e529459d907da4a3c35b0d01 : SUCCESS in 8m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/228de346aae847779099a2539cbc4213 : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19a3fc7289f749bc90054bb8aa5b114f : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db9de967ae72494aaa5eeac6c3e64b82 : SUCCESS in 39m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a850556b2a3d49bb9e01e2836c575617 : SUCCESS in 1h 14m 04s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7c402b4faa4345c8a360a57d1952e7b7 : SUCCESS in 2h 04m 50s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1e6ed6c111194c6e893423d36f409ba8 : SUCCESS in 23m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a289aa4d312242e3a9a287db7302574f : FAILURE in 29m 30s","accounts_in_message":[],"_revision_number":7},{"id":"d4103d5d0b242d57098314ab7ad23f297d246fd0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 16:43:14.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c55ce03e628f67dc641835f99d1409590b5520b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 18:53:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96b9d15d2820496c98457ed6d21667e8 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e7751fd7a24993bce9ef4ab0bfbd6c : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c67d0841ecd64c02a757b11913dd5d27 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc8aa1975ee942ac9f896dda2c4e211e : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a15f62db01884d1a9ec9097bb41a189e : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf83a6537fa74ad79111ebcefd83599c : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2b3a464dd7a4c03818a4caa5e9720b9 : SUCCESS in 7m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1972cbee2ee4991b8375309770acd13 : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2db047f614944113a6f9bf622a02bb4c : SUCCESS in 41m 01s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bb8324b0736148c4bf19267cc6c49b37 : SUCCESS in 1h 23m 02s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/58ceec3175c94b638263758dee3c8a95 : SUCCESS in 2h 05m 46s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8e4fb53d77b843d4829b6759626f57c1 : SUCCESS in 27m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5063f57a44a04f0d885845fcde700424 : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":7},{"id":"c5265b11649daf5cb8d6c0181937f5e023627a86","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-23 17:40:23.000000000","message":"Patch Set 7: Code-Review+1\n\nyou might like some of my improvements applied in https://review.opendev.org/709369 (templated attempts, pinging before ssh and verbose ssh - the two latter usually provide insight into where the problem might be)","accounts_in_message":[],"_revision_number":7},{"id":"03259465a69c4b375861d9073246468f3d3956d8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-24 10:31:54.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"c9d576904c4717d9a098013bc3be8d56b5bf1275","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-24 10:32:23.000000000","message":"Patch Set 8:\n\n\u003e you might like some of my improvements applied in https://review.opendev.org/709369\n \u003e (templated attempts, pinging before ssh and verbose ssh - the two\n \u003e latter usually provide insight into where the problem might be)\n\nThanks, adopted.","accounts_in_message":[],"_revision_number":8},{"id":"9327cc128497ccbb500f89eeaab871a7f9333d55","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-24 10:35:17.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"500e1708898bf042dfc4493117daf3fec432abbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 12:46:27.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d5388c667444e92812fcd9a2bd7f262 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f184f0e7c96498fb270ec14e428b7c5 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b6741e885754eceb9729c0a17f4e4fd : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92fac1719610455ab751b473aebd4070 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be8ad5cdb6be480ab39d2ee3278ed299 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe53701372754e6593102b3fd6faf7ba : SUCCESS in 11m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d3563565ee90441fa6d0f849d7737e47 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/244d25e7537d469fa624bb9d1348f337 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ab906c9789342e99cf08e099aa092f8 : SUCCESS in 36m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7a9ea2fe51af4596af317a5f79b53ade : SUCCESS in 1h 54m 07s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3cfa834e16fa476c9613a386622c3deb : SUCCESS in 2h 12m 31s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c75056e9aec747448cae19d1b07ae36b : SUCCESS in 27m 12s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/35099c39c7ad4d67b833cef682a89dc8 : SUCCESS in 36m 53s","accounts_in_message":[],"_revision_number":8},{"id":"51fd26a93326cffa988dbfe5e4f1397ed8d238ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-24 17:50:57.000000000","message":"Patch Set 8:\n\nyou might also want -n for ssh, just got ubu/deb stalled on password prompt","accounts_in_message":[],"_revision_number":8},{"id":"5fcd0e913b0d24f2a70ffb32560db33bc690f01e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-24 20:19:13.000000000","message":"Patch Set 8:\n\nignore that, -n does not help, one needs -o BatchMode\u003dyes","accounts_in_message":[],"_revision_number":8},{"id":"769b631cc341156b5d3d5b9015710cafe795e186","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-25 07:25:06.000000000","message":"Patch Set 8:\n\noh, and I also added -W1 to ping to control timeout","accounts_in_message":[],"_revision_number":8},{"id":"82caf13ff5d4301964d4549348ebba423f17fc98","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 14:12:08.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"a98761f8bb2b85afaa068105cd0aad72cf3d1c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:19:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2b50ab1ae5847c7877721d05b59088c : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae01658c2ae41e5857c85896742a657 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05be1ba5b81444e4993fa26dc473edfb : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/919af1f9a5874c8088af9be8f75d4046 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5be27380d744d6abaef6b294650af3a : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b54cb575530478d9b6d36d83486e221 : SUCCESS in 8m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97c704468cba48c7bee954d247a93bca : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07c15aa404854d9ba50aa9a3a99406e9 : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52165a72e5f244b2b54284942afc7549 : SUCCESS in 32m 07s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/cb62e3359b4c45d9be84b4d77456e346 : SUCCESS in 1h 27m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7e2b2d25c7764adfb42ed123c47f8780 : SUCCESS in 2h 06m 20s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/99a97a6a8c3745ff9639fb4315768983 : SUCCESS in 33m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bdbe1c0c370046ad848873a082d3cd81 : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":9},{"id":"a76d4ba392924dc2d9944c0754cd30b69605518a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-25 19:46:05.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7c1b0d04424f4d5bb29353b3e5659a201370dedc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 16:18:12.000000000","message":"Patch Set 9: Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"3be83f1d2a0e170c876d1ecd4b26a1b25c553c2d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-16 13:36:34.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"cb4c9180279adc71be5c1676bf16660891c600ca","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-04-21 20:01:56.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"6a5fde65d2c936cffedc21f435a1ac23a7e19212","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 20:02:15.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"02ca47ffacca88162b77700358dbdd59fee0f2b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 00:28:44.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0066113266e4bd394a72554e2179a6c : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d32ef52b09d644a1acad5a90df31ed91 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af56bbf40bd34b9a9e9e3ebd0a8794ca : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a325cfe91dcc484386103dd250281038 : SUCCESS in 6m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ffd4486b08e4fafa4fb4b3f9185946d : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f9410d0d86f4d72b62883f3dd1d684c : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ac19a13c0cc475cbfa08a5f77eab57d : SUCCESS in 33m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6aebf4fd899e4823a5850b3fe0f25066 : SUCCESS in 1h 12m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e12fd8a6a3054446bc4cb8b3fade6b44 : SUCCESS in 1h 48m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2945b2f223b34f339999f6d699768cc5 : SUCCESS in 16m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f12012c9807947009be91a204585674f : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":9},{"id":"337a20d8c38a26872c650ca020e13f5b6d45e6e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 00:28:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"478dacdd437d553abe922eefd7f298d2cb77e689","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 00:31:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b0e882741f4cc0bea68d0f8909db08 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"02327b1c546c43b768909a49a4b798e86c2da34c","revisions":{"84a7f575ef2312190e056b99dfd84197baaa3ae2":{"kind":"REWORK","_number":1,"created":"2020-02-13 10:22:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/1"}}},"commit":{"parents":[{"commit":"cc219bc0b5f956aa6592eb59db09310c963ccbe5","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc219bc0b5f956aa6592eb59db09310c963ccbe5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"subject":"WIP: CI: Test SSH connectivity to deployed instances","message":"WIP: CI: Test SSH connectivity to deployed instances\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84a7f575ef2312190e056b99dfd84197baaa3ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84a7f575ef2312190e056b99dfd84197baaa3ae2"}]},"branch":"refs/heads/master"},"cde0fc0df9e10f27742117f8b2901f545e8a01bc":{"kind":"REWORK","_number":2,"created":"2020-02-13 12:46:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/2"}}},"commit":{"parents":[{"commit":"cc219bc0b5f956aa6592eb59db09310c963ccbe5","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc219bc0b5f956aa6592eb59db09310c963ccbe5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 12:46:13.000000000","tz":0},"subject":"WIP: CI: Test SSH connectivity to deployed instances","message":"WIP: CI: Test SSH connectivity to deployed instances\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cde0fc0df9e10f27742117f8b2901f545e8a01bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cde0fc0df9e10f27742117f8b2901f545e8a01bc"}]},"branch":"refs/heads/master"},"ddd9e126107e890f8f593f96c378d69d0a1aa441":{"kind":"REWORK","_number":3,"created":"2020-02-13 14:30:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/3"}}},"commit":{"parents":[{"commit":"cc219bc0b5f956aa6592eb59db09310c963ccbe5","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc219bc0b5f956aa6592eb59db09310c963ccbe5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 14:30:27.000000000","tz":0},"subject":"WIP: CI: Test SSH connectivity to deployed instances","message":"WIP: CI: Test SSH connectivity to deployed instances\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddd9e126107e890f8f593f96c378d69d0a1aa441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ddd9e126107e890f8f593f96c378d69d0a1aa441"}]},"branch":"refs/heads/master"},"f2fbd26357c2600cdf208d19aec973b27edaaff4":{"kind":"REWORK","_number":4,"created":"2020-02-13 16:37:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/4"}}},"commit":{"parents":[{"commit":"cc219bc0b5f956aa6592eb59db09310c963ccbe5","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc219bc0b5f956aa6592eb59db09310c963ccbe5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:37:02.000000000","tz":0},"subject":"WIP: CI: Test SSH connectivity to deployed instances","message":"WIP: CI: Test SSH connectivity to deployed instances\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2fbd26357c2600cdf208d19aec973b27edaaff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2fbd26357c2600cdf208d19aec973b27edaaff4"}]},"branch":"refs/heads/master"},"46d03919f11e84418348bbf18599a317c02baf8c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-02-20 17:01:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/5"}}},"commit":{"parents":[{"commit":"cc219bc0b5f956aa6592eb59db09310c963ccbe5","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc219bc0b5f956aa6592eb59db09310c963ccbe5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 17:01:51.000000000","tz":0},"subject":"CI: Test SSH connectivity to deployed instances","message":"CI: Test SSH connectivity to deployed instances\n\nAdd testing of the dataplane to the overcloud jobs. To support both\nbaremetal provisioning and VMs with VXLAN tenant networks, we use the\nprovision-net as our external network to which floating IPs are\nattached.\n\nWe depend on a backport of parts of this patch to allow testing of SSH\nconnectivity after upgrades.\n\nDepends-On: https://review.opendev.org/708915/\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46d03919f11e84418348bbf18599a317c02baf8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46d03919f11e84418348bbf18599a317c02baf8c"}]},"branch":"refs/heads/master"},"f0a1cb7dc3b67739b51aea44ae0db2659f69b473":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-02-21 11:19:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/6"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 11:18:06.000000000","tz":0},"subject":"CI: Test SSH connectivity to deployed instances","message":"CI: Test SSH connectivity to deployed instances\n\nAdd testing of the dataplane to the overcloud jobs. To support both\nbaremetal provisioning and VMs with VXLAN tenant networks, we use the\nprovision-net as our external network to which floating IPs are\nattached.\n\nWe depend on a backport of parts of this patch to allow testing of SSH\nconnectivity after upgrades.\n\nDepends-On: https://review.opendev.org/708915/\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0a1cb7dc3b67739b51aea44ae0db2659f69b473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0a1cb7dc3b67739b51aea44ae0db2659f69b473"}]},"branch":"refs/heads/master"},"cd5b19fe36511e25f1279e616b52b3d940aeffff":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-02-21 13:55:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/7"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 13:55:53.000000000","tz":0},"subject":"CI: Test SSH connectivity to deployed instances","message":"CI: Test SSH connectivity to deployed instances\n\nAdd testing of the dataplane to the overcloud jobs. To support both\nbaremetal provisioning and VMs with VXLAN tenant networks, we use the\nprovision-net as our external network to which floating IPs are\nattached.\n\nWe depend on a backport of parts of this patch to allow testing of SSH\nconnectivity after upgrades.\n\nDepends-On: https://review.opendev.org/708915/\nDepends-On: https://review.opendev.org/709145/\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd5b19fe36511e25f1279e616b52b3d940aeffff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd5b19fe36511e25f1279e616b52b3d940aeffff"}]},"branch":"refs/heads/master"},"e5c03b572e4065d10a41b96c4076a04a66670aff":{"kind":"REWORK","_number":8,"created":"2020-02-24 10:31:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/8"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-24 10:31:50.000000000","tz":0},"subject":"CI: Test SSH connectivity to deployed instances","message":"CI: Test SSH connectivity to deployed instances\n\nAdd testing of the dataplane to the overcloud jobs. To support both\nbaremetal provisioning and VMs with VXLAN tenant networks, we use the\nprovision-net as our external network to which floating IPs are\nattached.\n\nWe depend on a backport of parts of this patch to allow testing of SSH\nconnectivity after upgrades.\n\nDepends-On: https://review.opendev.org/708915/\nDepends-On: https://review.opendev.org/709145/\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5c03b572e4065d10a41b96c4076a04a66670aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5c03b572e4065d10a41b96c4076a04a66670aff"}]},"branch":"refs/heads/master"},"02327b1c546c43b768909a49a4b798e86c2da34c":{"kind":"REWORK","_number":9,"created":"2020-02-25 14:12:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/707604/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/707604/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/707604/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/707604/9"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 10:22:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 14:12:05.000000000","tz":0},"subject":"CI: Test SSH connectivity to deployed instances","message":"CI: Test SSH connectivity to deployed instances\n\nAdd testing of the dataplane to the overcloud jobs. To support both\nbaremetal provisioning and VMs with VXLAN tenant networks, we use the\nprovision-net as our external network to which floating IPs are\nattached.\n\nWe depend on a backport of parts of this patch to allow testing of SSH\nconnectivity after upgrades.\n\nDepends-On: https://review.opendev.org/708915/\nDepends-On: https://review.opendev.org/709145/\n\nChange-Id: I6316e8959cff987e4e97280889e1038a9537ed32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02327b1c546c43b768909a49a4b798e86c2da34c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02327b1c546c43b768909a49a4b798e86c2da34c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
