)]}'
{"id":"openstack%2Fkayobe~707657","triplet_id":"openstack%2Fkayobe~master~Ia2f25a2e41859159275d418b57ab59bb97f58d6a","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ia2f25a2e41859159275d418b57ab59bb97f58d6a","subject":"CI: Redirect stderr to ansible logs","status":"MERGED","created":"2020-02-13 14:36:40.000000000","updated":"2020-03-19 11:08:42.000000000","submitted":"2020-03-19 11:07:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707657-1584616027346-e3550527","meta_rev_id":"ea56d033c4d96fd5ea5a2baa6925424d99b1a6d2","_number":707657,"virtual_id_number":707657,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 11:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-19 09:56:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-27 05:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-03-19 09:56:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 05:26:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-19 09:56:37.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-19 11:07:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73d9ae00b997636680eb435815fdab184195e35e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 14:36:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d60ade5dc8aa137cabf1450f660a057efe5bc30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:43:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a3139ba70e1428a8f332f6c6b6fde9f : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d6887fd7de9438c87598659c3180fc3 : FAILURE in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a8cd58d40394038b0b61799cd2b503c : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebd76eddba764f369eefabf8f8cd814b : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df870e289b2f440ba1c68244d9a951e3 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b807a8b0dac546059930cc8ad159ab01 : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c6c119239b7144708681783301cb9a84 : SUCCESS in 7m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1de41712675b4ff9970214b58609c201 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/127ee57ba1f14878a47872efa9e12b11 : SUCCESS in 38m 52s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f9c7a03ecdf54499ab7ba95c8c183b96 : SUCCESS in 1h 12m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e5a3539eff3d411596551aa2a3bdc635 : SUCCESS in 1h 56m 51s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ab1abde43c4e40a9861e2e59c11ef5bc : SUCCESS in 20m 58s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/129aedc6bd314d998830e89231e537af : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6ae859de32896e7cca661f31590ea3e5cb5a44a3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 16:57:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3dbda172f23285abc3ed047aeda2d5714d4e5278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 18:54:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f66bab37d5b84758a3f4441e5468f2de : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad836992c49b4b2185cfefcf69ae6d4d : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44d2d2ce949c4d0085e86745fb2bc8c8 : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6c79ae70b554ba78b792d3908116bb1 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a0d345a970d4c6db8e52e3c2df7186b : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8f840a7da047a3a6ed30381a6df066 : SUCCESS in 8m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a152de618544ea9b491f990f8735511 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7965988efb324384a5b34fa6845d4c6d : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db385ab332124492b85bcf5d0515eaa5 : SUCCESS in 37m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7ce48153b1254e8d83f7ca5125453a41 : SUCCESS in 1h 52m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fafd35fbc240484b92af020e513e4f5a : SUCCESS in 1h 45m 02s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ec09d59ab3c9483fb107a89d78492e76 : SUCCESS in 22m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8f2e3dc036224dca8c345b0e81981109 : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0b99b1fe12cae6b0985e316b5db959c158a8ce66","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-27 05:26:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c50778d520c54f52ab4d569bf73790e962a03ced","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-19 09:56:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bc0d9355932f6bde87bd52d6149bb7ed83766fde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 09:56:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7296c3cea9e9dda9cf82f566ee66555ca286c662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:07:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb9ac661bd77456aa88c29bf1f7dfe99 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0b0e05398d04b3096c33d5e5790ed71 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/540714f4068544b087803264f5ba7ee8 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c4d27b35ee4a8e80e6de489fa130c6 : SUCCESS in 7m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/277baa4887374b94bd39932e4f6c31a1 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eacfdcc41c8344e7b63e787b4f87d694 : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5900acfd46b34b94a9811a417eca564d : SUCCESS in 29m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a00d00e96abd486dbdb41c3f30de0dbf : SUCCESS in 1h 03m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d20d0ae3edfa4179bb746e528d42dcaa : SUCCESS in 18m 01s","accounts_in_message":[],"_revision_number":2},{"id":"be9a7aab9b647b7db6ad6f8f6de9b996b23c19a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:07:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ea56d033c4d96fd5ea5a2baa6925424d99b1a6d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:08:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e581a09e959f4f6cbd5ce6a53c5abddf : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ac88bc2dfb9bc2fb802cc6b1e3e2c0589d8d8e9f","revisions":{"6f32f7a41a0a4e6c27b0bf1d4868f82a6d9d8e00":{"kind":"REWORK","_number":1,"created":"2020-02-13 14:36:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/707657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/707657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/707657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/707657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/707657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/707657/1"}}},"commit":{"parents":[{"commit":"cc219bc0b5f956aa6592eb59db09310c963ccbe5","subject":"Merge \"[ussuri][goal] Drop python 2.7 support and testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc219bc0b5f956aa6592eb59db09310c963ccbe5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 14:36:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 14:36:32.000000000","tz":0},"subject":"CI: Redirect stderr to ansible logs","message":"CI: Redirect stderr to ansible logs\n\nCurrently Kayobe stderr is logged to the Zuul playbook output. This\nchange redirects stderr to the ansible log files.\n\nChange-Id: Ia2f25a2e41859159275d418b57ab59bb97f58d6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f32f7a41a0a4e6c27b0bf1d4868f82a6d9d8e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f32f7a41a0a4e6c27b0bf1d4868f82a6d9d8e00"}]},"branch":"refs/heads/master"},"ac88bc2dfb9bc2fb802cc6b1e3e2c0589d8d8e9f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-13 16:57:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/707657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/707657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/707657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/707657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/707657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/707657/2"}}},"commit":{"parents":[{"commit":"2fcdff101f888ac310969e4e5e0f2e409288a71d","subject":"Fix sphinx8 script used by pep8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fcdff101f888ac310969e4e5e0f2e409288a71d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 14:36:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:57:31.000000000","tz":0},"subject":"CI: Redirect stderr to ansible logs","message":"CI: Redirect stderr to ansible logs\n\nCurrently Kayobe stderr is logged to the Zuul playbook output. This\nchange redirects stderr to the ansible log files.\n\nChange-Id: Ia2f25a2e41859159275d418b57ab59bb97f58d6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac88bc2dfb9bc2fb802cc6b1e3e2c0589d8d8e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac88bc2dfb9bc2fb802cc6b1e3e2c0589d8d8e9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
