)]}'
{"id":"openstack%2Fkayobe~707690","triplet_id":"openstack%2Fkayobe~master~Ib8be24d9a2baebcd890a50eda8102ec65256b318","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ib8be24d9a2baebcd890a50eda8102ec65256b318","subject":"WIP: CentOS 8","status":"ABANDONED","created":"2020-02-13 16:54:41.000000000","updated":"2020-02-28 08:00:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4cdee423872be2adef1ada2774842215297e5b09","_number":707690,"virtual_id_number":707690,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-02-27 13:59:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2020-02-26 09:10:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-25 17:25:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 16:54:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-25 17:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-26 09:10:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"6c743fc1a92d062e27cb211f8de61a49f3f75d7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 16:54:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9febe4011591f0614bb9c4953a056b1cfeb5061a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 16:58:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"695ecc35b1056802f09ce4a361ddd80368d045e5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 17:26:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4d43d7271e80c2ad91629bc78825a201ffbb114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:45:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9e4734ddb8b6496fb5756d9806f4989b : FAILURE in 9m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9f4fe221f1bc4405931627f0f13dd21f : RETRY_LIMIT in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"e3e8dc0c9440449d10c4dbe5cc03da0ac5c9c298","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-13 18:04:05.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"541fda96ef164dc4976e3946bdd475a4e41f9d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 18:14:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b7b9936504ff441ab16534cfb6d58b9a : FAILURE in 8m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b80e468675204f97ac6cdf4671f53e22 : RETRY_LIMIT in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"10a84ab3718194f63e4c76ab34ceb3274b023d4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-14 12:38:03.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"df2627232a5ca696bdbb0c4496164b7fe7193d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 14:41:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/46fb7e11634d42b1be967411d9ba6736 : FAILURE in 9m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/95f57a994b024cb88119647ecce488ac : TIMED_OUT in 2h 01m 18s","accounts_in_message":[],"_revision_number":5},{"id":"d86319e570a79c6d510fc71d24780c621bb45c87","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-14 17:54:48.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"d8331b7f8e4af98df025a4133382c2fa06dd7f60","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-14 17:58:51.000000000","message":"Patch Set 6:\n\nThe following seems to kill the bridge:\n\nBRIDGE\u003dbreth1 bash -x ./ifdown-eth ifcfg-p-breth1-ovs","accounts_in_message":[],"_revision_number":6},{"id":"767db025050213b3eaeac8719034e61716f1a553","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-14 18:00:08.000000000","message":"Patch Set 6:\n\n\u003e The following seems to kill the bridge:\n \u003e \n \u003e BRIDGE\u003dbreth1 bash -x ./ifdown-eth ifcfg-p-breth1-ovs\n\nThis is because the bridge has no interfaces left.","accounts_in_message":[],"_revision_number":6},{"id":"3348b3ffed3104d3f3eaf7d33f91544c12ae03bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 18:06:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/fb6199d856344765b79dea7a141d9763 : FAILURE in 9m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/95d27452e6704ecb9b39aa9bffd8b11c : FAILURE in 10m 29s","accounts_in_message":[],"_revision_number":6},{"id":"f4eb83fa7580955f24e7691f1e5fc1d3ef734990","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-15 08:58:44.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"887666cb82ef5c7722641ee3564b50a492066b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 08:59:03.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"dc954524ffc1e56d2ccad4a4c86b9f5633da826b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-15 08:59:22.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"7083565d93ef92d95230dab25208f5214589effd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 08:59:43.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"8452dede7347d4d76b8dfbbf84da2b5ac1dd2f1d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-15 10:48:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8f73813a64bdb9366945e8958e2e9f3310d22d00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 11:04:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e732701ec6f34ac8a54e6bd62720b7c4 : FAILURE in 8m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e0f484b91819436b9984216bfd90a02f : FAILURE in 10m 07s","accounts_in_message":[],"_revision_number":9},{"id":"b6bfa60c9cbb13a43813cd3b9c6f601fa7253b43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-15 16:45:14.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"5834d307755bebcbc2b27bb7103383e5fa247895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 17:50:10.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9f86569601464c8990721747e7b5b3b1 : FAILURE in 7m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0f0027c94ef64a0bbb03509732f3e11a : SUCCESS in 1h 03m 46s","accounts_in_message":[],"_revision_number":10},{"id":"e022c9a263b1cbf7c1c7efc94e32e8286f75e991","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-17 09:22:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"02a399754a5f404b9b40fc3fa1e8e0cbef91ecf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 10:34:17.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a63121cd81bc4d5e8976c97c87105af6 : FAILURE in 8m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7c067d0617fd4e389e5e35f686a25a15 : SUCCESS in 1h 06m 40s","accounts_in_message":[],"_revision_number":11},{"id":"772074a7d0504f01ab6a1d945d9f79471da594b5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 15:47:18.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"bbacbabddc68647b260a984f53a7d94fcb8574ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 17:03:53.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/dcd35033442e492c91f900a460725491 : FAILURE in 9m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0119399ec0d242e89bc6d54cd66259b6 : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":12},{"id":"a7b9fdafb33e842902820087f37812e2bf83ef48","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 17:10:42.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6f0de2cc6b6a06348b33bf7e5dd9cab217db6789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 17:29:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/58d5ed85de6346339a8e5a2baa3fd336 : FAILURE in 12m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/42bc3e55a5fc47bcba6f1a37de471d41 : FAILURE in 12m 38s","accounts_in_message":[],"_revision_number":13},{"id":"1cc9bf9e26ea62d81f04f5f7ef9d140fe553b555","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 17:51:24.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3c2522c41589a628a8896777a2c1d101c66863b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 19:05:47.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/99b74a578d614269a11c7b711f365e16 : FAILURE in 9m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d156086686284e539ef7f5df744a300f : SUCCESS in 1h 11m 54s","accounts_in_message":[],"_revision_number":14},{"id":"18ff09b6d7756f7a3dd010b01d356df0e5be5fe8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 13:47:25.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"be14f6a22eb14e7b6d98f2590ebb91e0ab547ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 14:04:00.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/18c82fc1d62a49339d88e99540d775c2 : FAILURE in 8m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bc9372aff0f94b84bf96687c60ce8ed2 : FAILURE in 8m 40s","accounts_in_message":[],"_revision_number":15},{"id":"1aad1ffc92e2666ba0d5812ce1eeaff3f5935b30","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 15:39:45.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"31a4b21ba3b2ffd954191d3e81fc2327155c728f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 16:53:54.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2ab0323d730a4653a7e97201969da992 : FAILURE in 1h 01m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/981024a61a67420fb5e212ca3e44de83 : FAILURE in 54m 32s","accounts_in_message":[],"_revision_number":16},{"id":"97c3ff52e91f9626e609bb009a1970fdf00c5272","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 17:02:02.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"ecc451d465fe5e6e7e9fd2a43839f220dccd6f29","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 17:09:15.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"ecb8c1c6391292814391e878751e1131ee950e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 18:30:50.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7e5b76cf59ee4ff8a39f6d6f1ed48b23 : SUCCESS in 1h 18m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ad100936bf724a50b7c7569db6709f73 : SUCCESS in 1h 06m 34s","accounts_in_message":[],"_revision_number":18},{"id":"aed9e533e9b31a98d8a0350571c6167a1c33a743","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 19:46:06.000000000","message":"Patch Set 18:\n\nBoth passed!","accounts_in_message":[],"_revision_number":18},{"id":"266e817f1e046365d77f1d7858cb648a7b82acc0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 16:51:39.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"9f69c56feff42bf43abcc8750ee7dfb3e818f6e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 17:25:34.000000000","message":"Patch Set 19: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":19},{"id":"5aafd52612f34f76bee52dc2c940313cfa81c890","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-26 09:10:37.000000000","message":"Patch Set 19:\n\nwhat is veth bounce? :O","accounts_in_message":[],"_revision_number":19},{"id":"27329a4fe2bfe76602219441a6cb63d1a69ca582","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 13:59:07.000000000","message":"Patch Set 19:\n\n\u003e what is veth bounce? :O\n\nwe create veth pairs for OVS networking, and bouncing them takes down the bridge with our IP","accounts_in_message":[],"_revision_number":19},{"id":"0d5a508669503b1ca015159f0891549a5cdcfda8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 13:59:31.000000000","message":"Abandoned\n\nNow split","accounts_in_message":[],"_revision_number":19},{"id":"4cdee423872be2adef1ada2774842215297e5b09","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-28 08:00:40.000000000","message":"Patch Set 19:\n\nah, bouncing as in \"down\"-\"up\"","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"17044ef84c0c91eb558f105cc39a668ffa664907","revisions":{"a20ab54c52d0e40e341221440290c7ae2e505be1":{"kind":"REWORK","_number":1,"created":"2020-02-13 16:54:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/1"}}},"commit":{"parents":[{"commit":"51eac08775942859d33219655d10764759af07af","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51eac08775942859d33219655d10764759af07af"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-01-17 19:34:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:53:46.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a20ab54c52d0e40e341221440290c7ae2e505be1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a20ab54c52d0e40e341221440290c7ae2e505be1"}]},"branch":"refs/heads/master"},"6dfe40e547c4feadc63f820fccd3bb6cf391f616":{"kind":"NO_CHANGE","_number":2,"created":"2020-02-13 16:58:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/2"}}},"commit":{"parents":[{"commit":"51eac08775942859d33219655d10764759af07af","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51eac08775942859d33219655d10764759af07af"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dfe40e547c4feadc63f820fccd3bb6cf391f616"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dfe40e547c4feadc63f820fccd3bb6cf391f616"}]},"branch":"refs/heads/master"},"586a78232cf9a8ea586bf84e6c7f3de1144b2d9d":{"kind":"REWORK","_number":3,"created":"2020-02-13 17:26:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/3"}}},"commit":{"parents":[{"commit":"51eac08775942859d33219655d10764759af07af","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51eac08775942859d33219655d10764759af07af"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 17:25:55.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586a78232cf9a8ea586bf84e6c7f3de1144b2d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586a78232cf9a8ea586bf84e6c7f3de1144b2d9d"}]},"branch":"refs/heads/master"},"2f06b018676606e3ba85afe7531a63c451385ce1":{"kind":"REWORK","_number":4,"created":"2020-02-13 18:04:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/4"}}},"commit":{"parents":[{"commit":"51eac08775942859d33219655d10764759af07af","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51eac08775942859d33219655d10764759af07af"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 18:04:01.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f06b018676606e3ba85afe7531a63c451385ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f06b018676606e3ba85afe7531a63c451385ce1"}]},"branch":"refs/heads/master"},"3601240f7168e8a745dfa8257814395ed17a89b9":{"kind":"REWORK","_number":5,"created":"2020-02-14 12:38:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/5"}}},"commit":{"parents":[{"commit":"51eac08775942859d33219655d10764759af07af","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51eac08775942859d33219655d10764759af07af"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-14 12:37:57.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3601240f7168e8a745dfa8257814395ed17a89b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3601240f7168e8a745dfa8257814395ed17a89b9"}]},"branch":"refs/heads/master"},"abddec8be89bbaabe0d7302bc0e48e472998c148":{"kind":"REWORK","_number":6,"created":"2020-02-14 17:54:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/6"}}},"commit":{"parents":[{"commit":"51eac08775942859d33219655d10764759af07af","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51eac08775942859d33219655d10764759af07af"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-14 17:54:44.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abddec8be89bbaabe0d7302bc0e48e472998c148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abddec8be89bbaabe0d7302bc0e48e472998c148"}]},"branch":"refs/heads/master"},"f08924617bc6276a96a1e9f73698974ec6a41bdd":{"kind":"REWORK","_number":7,"created":"2020-02-15 08:58:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/7"}}},"commit":{"parents":[{"commit":"51eac08775942859d33219655d10764759af07af","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51eac08775942859d33219655d10764759af07af"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-15 08:58:36.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f08924617bc6276a96a1e9f73698974ec6a41bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f08924617bc6276a96a1e9f73698974ec6a41bdd"}]},"branch":"refs/heads/master"},"43678b72f3e262dd02eb3a6ba49af036cb7abd4a":{"kind":"REWORK","_number":8,"created":"2020-02-15 08:59:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/8"}}},"commit":{"parents":[{"commit":"51eac08775942859d33219655d10764759af07af","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51eac08775942859d33219655d10764759af07af"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-15 08:59:16.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43678b72f3e262dd02eb3a6ba49af036cb7abd4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43678b72f3e262dd02eb3a6ba49af036cb7abd4a"}]},"branch":"refs/heads/master"},"9c4b9081d42a2b25942259169c5a04f00e371d9e":{"kind":"REWORK","_number":9,"created":"2020-02-15 10:48:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/9"}}},"commit":{"parents":[{"commit":"1ba5f9bc300597e0cf6b3916ddd90eca8e0e5b02","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ba5f9bc300597e0cf6b3916ddd90eca8e0e5b02"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-15 10:46:39.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c4b9081d42a2b25942259169c5a04f00e371d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c4b9081d42a2b25942259169c5a04f00e371d9e"}]},"branch":"refs/heads/master"},"862abb7d959872aa69028f97d1fecf36da2f6c1c":{"kind":"REWORK","_number":10,"created":"2020-02-15 16:45:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/10"}}},"commit":{"parents":[{"commit":"1ba5f9bc300597e0cf6b3916ddd90eca8e0e5b02","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ba5f9bc300597e0cf6b3916ddd90eca8e0e5b02"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-15 16:45:11.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/862abb7d959872aa69028f97d1fecf36da2f6c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/862abb7d959872aa69028f97d1fecf36da2f6c1c"}]},"branch":"refs/heads/master"},"ec7d221b6557969baba1851892603f84ddc4d41a":{"kind":"REWORK","_number":11,"created":"2020-02-17 09:22:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/11"}}},"commit":{"parents":[{"commit":"1ba5f9bc300597e0cf6b3916ddd90eca8e0e5b02","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ba5f9bc300597e0cf6b3916ddd90eca8e0e5b02"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-17 08:55:49.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec7d221b6557969baba1851892603f84ddc4d41a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec7d221b6557969baba1851892603f84ddc4d41a"}]},"branch":"refs/heads/master"},"10b81a67df78b1032dc729d34d8f4cc176270bc8":{"kind":"REWORK","_number":12,"created":"2020-02-20 15:47:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/12"}}},"commit":{"parents":[{"commit":"2e842ab3f5c7b737b547f4c2c718f96294f582d5","subject":"Remove activate-virtualenv and deactivate-virtualenv roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e842ab3f5c7b737b547f4c2c718f96294f582d5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 15:28:56.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ...\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10b81a67df78b1032dc729d34d8f4cc176270bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/10b81a67df78b1032dc729d34d8f4cc176270bc8"}]},"branch":"refs/heads/master"},"5cbfd229bc30c6286172b8e158e988bb9bbfdd0a":{"kind":"REWORK","_number":13,"created":"2020-02-20 17:10:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/13"}}},"commit":{"parents":[{"commit":"87add8c80095eee5ceb488ff00c727156ad0903a","subject":"Bump Ansible version range to 2.8 - 2.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87add8c80095eee5ceb488ff00c727156ad0903a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 17:10:24.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ntp package not available on CentOS 8\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cbfd229bc30c6286172b8e158e988bb9bbfdd0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cbfd229bc30c6286172b8e158e988bb9bbfdd0a"}]},"branch":"refs/heads/master"},"9a634753698a8f9effc0fda28cf13c3b2a689c14":{"kind":"REWORK","_number":14,"created":"2020-02-20 17:51:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/14"}}},"commit":{"parents":[{"commit":"87add8c80095eee5ceb488ff00c727156ad0903a","subject":"Bump Ansible version range to 2.8 - 2.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87add8c80095eee5ceb488ff00c727156ad0903a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 17:51:16.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ntp package not available on CentOS 8\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a634753698a8f9effc0fda28cf13c3b2a689c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a634753698a8f9effc0fda28cf13c3b2a689c14"}]},"branch":"refs/heads/master"},"c671580bbf7adcb2b8f28fa9bb95117c068c7961":{"kind":"REWORK","_number":15,"created":"2020-02-21 13:47:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/15"}}},"commit":{"parents":[{"commit":"c9934ffe7afb152de17d29dc1fed8bc18f4217ab","subject":"Bump Ansible version range to 2.8 - 2.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9934ffe7afb152de17d29dc1fed8bc18f4217ab"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 13:46:58.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ntp package not available on CentOS 8\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c671580bbf7adcb2b8f28fa9bb95117c068c7961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c671580bbf7adcb2b8f28fa9bb95117c068c7961"}]},"branch":"refs/heads/master"},"2196628db49a578654488f8e63f0621a12630aa0":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-02-21 15:39:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/16"}}},"commit":{"parents":[{"commit":"fc529e38f4a9fd8799b76595c0e4f8f990735366","subject":"Bump Ansible version range to 2.8 - 2.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc529e38f4a9fd8799b76595c0e4f8f990735366"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 15:38:58.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ntp package not available on CentOS 8\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2196628db49a578654488f8e63f0621a12630aa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2196628db49a578654488f8e63f0621a12630aa0"}]},"branch":"refs/heads/master"},"ab6050a313f15d76460db336ee19cfa5a979c50d":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2020-02-21 17:02:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/17"}}},"commit":{"parents":[{"commit":"7ca854690a088a156a03fe71154067ecc06d37a4","subject":"Bump Ansible version range to 2.8 - 2.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ca854690a088a156a03fe71154067ecc06d37a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 17:01:50.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ntp package not available on CentOS 8\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab6050a313f15d76460db336ee19cfa5a979c50d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab6050a313f15d76460db336ee19cfa5a979c50d"}]},"branch":"refs/heads/master"},"5d6ac9fd034756431342f908f25ebfb4b567b83e":{"kind":"REWORK","_number":18,"created":"2020-02-21 17:09:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/18"}}},"commit":{"parents":[{"commit":"7ca854690a088a156a03fe71154067ecc06d37a4","subject":"Bump Ansible version range to 2.8 - 2.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ca854690a088a156a03fe71154067ecc06d37a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 17:09:07.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ntp package not available on CentOS 8\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d6ac9fd034756431342f908f25ebfb4b567b83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d6ac9fd034756431342f908f25ebfb4b567b83e"}]},"branch":"refs/heads/master"},"17044ef84c0c91eb558f105cc39a668ffa664907":{"kind":"REWORK","_number":19,"created":"2020-02-25 16:51:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/707690/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/707690/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/707690/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/707690/19"}}},"commit":{"parents":[{"commit":"f1974d41e220ea23da2696d499a0c274c8dcf282","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1974d41e220ea23da2696d499a0c274c8dcf282"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-13 16:58:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:50:28.000000000","tz":0},"subject":"WIP: CentOS 8","message":"WIP: CentOS 8\n\nNeeds to be split into separate changes.\n\nIssues:\n* veth bounce kills IP on breth1\n* ntp package not available on CentOS 8\n\nChange-Id: Ib8be24d9a2baebcd890a50eda8102ec65256b318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17044ef84c0c91eb558f105cc39a668ffa664907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17044ef84c0c91eb558f105cc39a668ffa664907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
