)]}'
{"id":"openstack%2Fkayobe~708922","triplet_id":"openstack%2Fkayobe~master~I3f8f7f8d7d37e3cb851965a491ac9c43030869d5","project":"openstack/kayobe","branch":"master","topic":"story/2006574","hashtags":[],"change_id":"I3f8f7f8d7d37e3cb851965a491ac9c43030869d5","subject":"Bump Ansible version range to 2.8 - 2.9","status":"MERGED","created":"2020-02-20 17:10:42.000000000","updated":"2020-02-28 21:49:16.000000000","submitted":"2020-02-28 21:47:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708922-1582926420819-bef65be0","meta_rev_id":"d458b6ae08cd42fc25291bc54eafe781ac90924d","_number":708922,"virtual_id_number":708922,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-02-27 15:40:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 21:47:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-02-28 19:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-02-27 15:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-27 15:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-27 15:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-02-28 19:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 11:39:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-27 12:06:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-27 12:27:05.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-28 19:18:53.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-28 21:47:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca4eb422c90669f9454331336157b0779e637dce","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-20 17:10:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adc6f05179a89fecf5f1f438cb8e7301806a4aa4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-20 17:36:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"730a2b2acc0bce23afa720a9e699a8f8affa3ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 18:28:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e4d2f84b2684befb65ae87a1a048de6 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cbd6493e3a74537bb466b5f86ce99d1 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1638d3689d2d45f3bd205d80e0eb0ad8 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a3562b07bdb406c9f1581a177b1923d : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6591c4900ae548859b16149660a49a1f : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383bfc23c03f45bb9c6b42f15088cb23 : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e5d724292aa497baf9c5a305a911471 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eddc09ae55aa4b70b8402dc440e72775 : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce1a504443634f0eb6c95a4860849ebc : SUCCESS in 37m 19s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7c94d2a6edee4d67a891fe78e749bfef : FAILURE in 13m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c9b92c38e9274a20bd3fc2da0ae7df1f : FAILURE in 1h 12m 42s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e9d80145f3e34003af48a8f1571fda9c : FAILURE in 9m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8a578864875b496085a60c914eb0f6cd : FAILURE in 26m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6034af75a9aa8ed992c98ddd72cab3b015c7e967","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 13:47:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"210deb083ed8bf7e172ca64cef50108191e737e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 15:27:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c45aebf5643b4463ae208f0ecc0c6a3f : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eba7858f05a4dd68150346a8f690a4d : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dac672ef0a74bcbb2f55c8f82bedc47 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab4169f6cdf944d392a7c0a04e0d1d8b : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33d2244aa444796977d72b067f41327 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2278cc136a3495490f036dbddd5c5bb : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66c6e1079b6949739ede801fe388ae45 : SUCCESS in 5m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7de430a3d9b84e56a03f87a9710abc26 : SUCCESS in 7m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bea0781580d4f8ea7559beab37a3f45 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03b19de404a6431191003259cdac0c9a : SUCCESS in 31m 44s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1d99df133914438a82c5056f59c428e0 : FAILURE in 9m 40s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dd631f47bf0246aa8ac068f2100e397a : SUCCESS in 1h 38m 24s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f19fdcc5bc7c462296015ab2df51738b : FAILURE in 7m 34s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ce2e4ddb7fd34a7fb0f8c1f5b5d6e535 : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":2},{"id":"9bf99a0059662ca677e743998492e33638fdb80c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 15:39:45.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7ebd6529102ffa7510118f5b290304ce723ba88c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 17:02:02.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b9705ec579018581d138ba005030db3362217acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 18:58:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f13212fbab434be291c09b04d0b2fc91 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19ebb77e7fc240888313f78ae44e65db : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5229471296934a45afc115bee42a6633 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f283977da7cf40769e8d5c61e51bdbc6 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c69ed5cdd09f4643ae7b4d878bb11c89 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696fb1d93209478991071952934cb6a7 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20c0dd1cb6004fa7bf1103442bd1e71f : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/91d47e3d30dd41fbb4031ec3561cdf2b : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/272dae30137d42a89acb269090b06814 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/83ba640b2aac4c89a618abdbb6bb246b : SUCCESS in 31m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/981cd211920d4eb4abe26e5fb2ef1ad5 : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8e0c048f8e424ef79e8da17e4d1483f4 : SUCCESS in 1h 55m 02s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/13c109d8640b481393ea2d564d3023d5 : SUCCESS in 23m 00s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/acb4a63be4354ceaa849c0cb3d031c69 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":4},{"id":"c868052656cea43f578e230edfbd4354b4c5023c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-22 09:57:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"25cd6f8246ab7a8dfe811332c8b7fd3cde99899f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-23 17:41:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"fac9bda52b311c97aa94dd1993709acb02ca0a7b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 15:43:16.000000000","message":"Topic set to story/2006574","accounts_in_message":[],"_revision_number":4},{"id":"690a60d8658517f773859bd2221a1479c0f763f2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-25 16:09:25.000000000","message":"Patch Set 4: Code-Review+2\n\nCI passes, so hopefully none of the existing deprecation warnings have become errors.","accounts_in_message":[],"_revision_number":4},{"id":"4e80506cb9833e40ff598be9ab6a1cf582fdd224","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 16:51:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"98634549db9a6187ffd9c866d7c310453cb09999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 18:54:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49eed317f4594585b5d47f96257b2faf : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54030f4b67b446d6947a9d3c96a5420c : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d7591f880c0442fa8689b12178c873f : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0747f981a2734e30938baa35d1351fa2 : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68d27ad289ca456ebbfc26478b60d612 : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72939dafcdcf4f7b9ed5437c9bef034c : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43042ca182944b638f1365de1ddb0157 : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23820ef72a1242eeb8adfb2d3098036c : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0dc41ed88714901a229e263afc62ced : SUCCESS in 13m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5e61820272e42fb87c01804a261a96d : SUCCESS in 42m 10s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ac39eba1729543ec9c554606daa48129 : SUCCESS in 1h 13m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/25ead055e05649f4b2ecc55d47185610 : SUCCESS in 1h 58m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3f0df0efabb54fbabdf7f1aa6f4552ab : SUCCESS in 28m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/757b16dca86c43e9958ac2b6f2780e5f : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":5},{"id":"97f211e22ea85578a8a98de48a8bf3d48bf294f4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-26 16:39:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e943e96972bbfdd1a3f935601fd07d433203612a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 11:24:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"12fb67213976aeedcabe3829e53bf1dae6d28051","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-27 11:39:18.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"bc16fbda437d4c365ae50816e2863961ee8b1522","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 12:06:38.000000000","message":"Patch Set 6: Code-Review+1\n\nI want to highlight the CVE-2019-14864 [1].\n\nI don\u0027t think it\u0027s an issue in our context (openstack), but maybe for some reasons some of users/packagers use splunk to handle logs so I prefer to speak about instead of ignore it.\n\n[1] https://access.redhat.com/security/cve/CVE-2019-14864","accounts_in_message":[],"_revision_number":6},{"id":"6300d5802daa61d2ab244e6086de1e21817efcb0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-27 12:27:05.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c8a44c29d24342cc3e7f2f21f4db0813d8d63a19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:23:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e071525c7312437daae806bd418b93bc : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbb150986974174b65f2196ae086fed : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e367d528fb3462c97c6fd395220fd95 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a87e847b5f2f4edf810a188ca3c48d99 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bb2519875cc40fe95fd0eb22004818b : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecd37ac4abe24076b6c274446ece3c45 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa204431ef0e4cd6bb98207a80626c1d : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6632bc3bfd99428d8e33f73539785ed1 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/366c9de0999c4a3bbc58b94adce5cf4b : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eaf1455c523748ed9e452462ca3fa58e : SUCCESS in 39m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e2b5e79036344551b3ef475a56cc4537 : FAILURE in 9m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/09f3fdb77e284cc1836b5a299dad829d : RETRY_LIMIT in 25m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/499c0cad0c544b6487691f9e41e0a93b : SUCCESS in 26m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f0e3e2d6c8df48dc9a444845c7fea9fa : SUCCESS in 41m 50s","accounts_in_message":[],"_revision_number":6},{"id":"69fc8529c13c048e2b5b0c6f17f954da6dd118bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 15:34:27.000000000","message":"Patch Set 6:\n\n\u003e I want to highlight the CVE-2019-14864 [1].\n \u003e \n \u003e I don\u0027t think it\u0027s an issue in our context (openstack), but maybe\n \u003e for some reasons some of users/packagers use splunk to handle logs\n \u003e so I prefer to speak about instead of ignore it.\n \u003e \n \u003e [1] https://access.redhat.com/security/cve/CVE-2019-14864\n\nThanks for pointing it out Herve. I don\u0027t think it affects us as we don\u0027t use those plugins but good to be aware.","accounts_in_message":[],"_revision_number":6},{"id":"bbf0c15d6bd7bbbf158b9bb4bc07941068db3206","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 15:39:38.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"802b08379a34925f937e62d408dd4c0a7c97941b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 15:40:35.000000000","message":"Patch Set 7:\n\nLooks like we need to bump MichaelRigart.interfaces role before this bump. I\u0027m not sure why we didn\u0027t see that before.","accounts_in_message":[],"_revision_number":7},{"id":"fd8dc5b650b4b1e3ef9819476b3b2ef19f7b33bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 21:46:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61a8e2dbe494408a8a3657f70c3bfe72 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6710b3dce9d49b9823fcef77d4334d0 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0683b0b9508e48f785dd4f3a989f34b3 : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bccea21c26fa41acb0ca9539bf1cb7c6 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba6a4a030f944acfb8ef5a1cfff80e8c : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb01ad8376a4d2bbb174931c7085055 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0353bcc015ec403f9b4faf4ce3f230ca : SUCCESS in 7m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0ab0616de8f45ee84f70b7a3c1a8cb3 : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2479533823244711a0db6bc38704fb8f : SUCCESS in 13m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a36b1c9c2344a219f4281e6bb0213bb : SUCCESS in 41m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9d03af061af44dd0be26dbcefd123dee : SUCCESS in 1h 13m 57s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/028cac0cd3c94742b1158cbb66a81c5b : SUCCESS in 2h 10m 17s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/cda2b78a3686429bb71d060423f08fe5 : SUCCESS in 28m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5db24e2d14d74f82b8b940b6098e9a5c : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":7},{"id":"a75c96954d76d3a76b1785b1ddbe559ed3672a1b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-28 19:18:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"9b8cd92b98eb9dba4da2c26527a5176543e8d66b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:19:05.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"380d837182167d49d5bd0b551a2a5c7b99ebc0ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:47:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"434aa972e8bb2de2999f19de3063a6bbf62df412","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:47:00.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8677c52ba0d49e1a9ea07b59be0e9ce : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3697529ed48461a893c68e23caffc31 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9e3d2016348497bbd05dfcbe69fee04 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34950112e9814b608a4494bf4c00d1cf : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7a9ada065a345dcaf3cb8d6e4cac5a5 : SUCCESS in 5m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/976e12cb83424a30b48b539bbcbc2f21 : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09d354751dc54e6b9d8827f2653c5d8d : SUCCESS in 14m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2097e36ee2d41d8a501b027d971bd2d : SUCCESS in 31m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a009811fb8684698a10b0e7f35f25551 : SUCCESS in 1h 18m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/05591cac5b454258816164eeb82b0ea8 : SUCCESS in 2h 18m 50s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/45fcec1b9880464fb4f8af3b7dbb3f3e : SUCCESS in 33m 06s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/52fc00c9b1d940beb87ffb89d07d5986 : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":7},{"id":"d458b6ae08cd42fc25291bc54eafe781ac90924d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:49:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fa40691af64b3bb2ba51ab99269415 : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5aa64b301ee449f9b67d08afe492e48 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ebd5aa44777b78be19d04d8bef027399f1074b5d","revisions":{"87add8c80095eee5ceb488ff00c727156ad0903a":{"kind":"REWORK","_number":1,"created":"2020-02-20 17:10:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/708922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/708922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/708922/1"}}},"commit":{"parents":[{"commit":"8a0d2dc66e9c4d07188a2b41fd47efc6574d4ce9","subject":"Make Kayobe code compatible with Python 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a0d2dc66e9c4d07188a2b41fd47efc6574d4ce9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:30:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 17:10:24.000000000","tz":0},"subject":"Bump Ansible version range to 2.8 - 2.9","message":"Bump Ansible version range to 2.8 - 2.9\n\nUpdates the minimum version of Ansible from 2.6 to 2.8, and the maximum\nsupported version from 2.8 to 2.9.\n\nCentOS 8 requires Ansible 2.8.\n\nChange-Id: I3f8f7f8d7d37e3cb851965a491ac9c43030869d5\nStory: 2006574\nTask: 38826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87add8c80095eee5ceb488ff00c727156ad0903a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87add8c80095eee5ceb488ff00c727156ad0903a"}]},"branch":"refs/heads/master"},"c9934ffe7afb152de17d29dc1fed8bc18f4217ab":{"kind":"REWORK","_number":2,"created":"2020-02-21 13:47:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/708922/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/708922/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/708922/2"}}},"commit":{"parents":[{"commit":"41d4768155711c3a95e5b13739e2fc3c1e0c25fa","subject":"Make Kayobe code compatible with Python 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41d4768155711c3a95e5b13739e2fc3c1e0c25fa"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:30:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 13:46:52.000000000","tz":0},"subject":"Bump Ansible version range to 2.8 - 2.9","message":"Bump Ansible version range to 2.8 - 2.9\n\nUpdates the minimum version of Ansible from 2.6 to 2.8, and the maximum\nsupported version from 2.8 to 2.9.\n\nCentOS 8 requires Ansible 2.8.\n\nChange-Id: I3f8f7f8d7d37e3cb851965a491ac9c43030869d5\nStory: 2006574\nTask: 38826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9934ffe7afb152de17d29dc1fed8bc18f4217ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9934ffe7afb152de17d29dc1fed8bc18f4217ab"}]},"branch":"refs/heads/master"},"fc529e38f4a9fd8799b76595c0e4f8f990735366":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-21 15:39:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/708922/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/708922/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/708922/3"}}},"commit":{"parents":[{"commit":"6698f433ec39554c0ddb70105317ea323a209e2e","subject":"Make Kayobe code compatible with Python 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6698f433ec39554c0ddb70105317ea323a209e2e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:30:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 15:38:58.000000000","tz":0},"subject":"Bump Ansible version range to 2.8 - 2.9","message":"Bump Ansible version range to 2.8 - 2.9\n\nUpdates the minimum version of Ansible from 2.6 to 2.8, and the maximum\nsupported version from 2.8 to 2.9.\n\nCentOS 8 requires Ansible 2.8.\n\nChange-Id: I3f8f7f8d7d37e3cb851965a491ac9c43030869d5\nStory: 2006574\nTask: 38826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc529e38f4a9fd8799b76595c0e4f8f990735366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc529e38f4a9fd8799b76595c0e4f8f990735366"}]},"branch":"refs/heads/master"},"7ca854690a088a156a03fe71154067ecc06d37a4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-02-21 17:02:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/708922/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/708922/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/708922/4"}}},"commit":{"parents":[{"commit":"87eeebf26d550c307ef396e5cb139c767d1b6cdb","subject":"Make Kayobe code compatible with Python 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87eeebf26d550c307ef396e5cb139c767d1b6cdb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:30:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 17:01:50.000000000","tz":0},"subject":"Bump Ansible version range to 2.8 - 2.9","message":"Bump Ansible version range to 2.8 - 2.9\n\nUpdates the minimum version of Ansible from 2.6 to 2.8, and the maximum\nsupported version from 2.8 to 2.9.\n\nCentOS 8 requires Ansible 2.8.\n\nChange-Id: I3f8f7f8d7d37e3cb851965a491ac9c43030869d5\nStory: 2006574\nTask: 38826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ca854690a088a156a03fe71154067ecc06d37a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ca854690a088a156a03fe71154067ecc06d37a4"}]},"branch":"refs/heads/master"},"323a918e1c4248377f25e0319271a69f94477399":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-02-25 16:51:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/708922/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/708922/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/708922/5"}}},"commit":{"parents":[{"commit":"78d2fdfdbabd3a4904dbfbf53eaa5b02d084bdb7","subject":"Make Kayobe code compatible with Python 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d2fdfdbabd3a4904dbfbf53eaa5b02d084bdb7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:30:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:15:09.000000000","tz":0},"subject":"Bump Ansible version range to 2.8 - 2.9","message":"Bump Ansible version range to 2.8 - 2.9\n\nUpdates the minimum version of Ansible from 2.6 to 2.8, and the maximum\nsupported version from 2.8 to 2.9.\n\nCentOS 8 requires Ansible 2.8.\n\nChange-Id: I3f8f7f8d7d37e3cb851965a491ac9c43030869d5\nStory: 2006574\nTask: 38826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323a918e1c4248377f25e0319271a69f94477399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323a918e1c4248377f25e0319271a69f94477399"}]},"branch":"refs/heads/master"},"43cf52f17074eaf491fb43edbc97b32fe91083ee":{"kind":"REWORK","_number":6,"created":"2020-02-27 11:24:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/708922/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/708922/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/708922/6"}}},"commit":{"parents":[{"commit":"15109ccb547e603c0aca2a73e43f4db523fa1bcd","subject":"Make Kayobe code compatible with Python 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15109ccb547e603c0aca2a73e43f4db523fa1bcd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:30:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 11:15:27.000000000","tz":0},"subject":"Bump Ansible version range to 2.8 - 2.9","message":"Bump Ansible version range to 2.8 - 2.9\n\nUpdates the minimum version of Ansible from 2.6 to 2.8, and the maximum\nsupported version from 2.8 to 2.9.\n\nCentOS 8 requires Ansible 2.8.\n\nChange-Id: I3f8f7f8d7d37e3cb851965a491ac9c43030869d5\nStory: 2006574\nTask: 38826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43cf52f17074eaf491fb43edbc97b32fe91083ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43cf52f17074eaf491fb43edbc97b32fe91083ee"}]},"branch":"refs/heads/master"},"ebd5aa44777b78be19d04d8bef027399f1074b5d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-02-27 15:39:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/708922/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/708922/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/708922/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/708922/7"}}},"commit":{"parents":[{"commit":"3c9daababe8b8145d948ba066c6391648fdf2e31","subject":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c9daababe8b8145d948ba066c6391648fdf2e31"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:30:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 15:39:27.000000000","tz":0},"subject":"Bump Ansible version range to 2.8 - 2.9","message":"Bump Ansible version range to 2.8 - 2.9\n\nUpdates the minimum version of Ansible from 2.6 to 2.8, and the maximum\nsupported version from 2.8 to 2.9.\n\nCentOS 8 requires Ansible 2.8.\n\nChange-Id: I3f8f7f8d7d37e3cb851965a491ac9c43030869d5\nStory: 2006574\nTask: 38826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5aa44777b78be19d04d8bef027399f1074b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5aa44777b78be19d04d8bef027399f1074b5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
