)]}'
{"id":"openstack%2Fkayobe~709154","triplet_id":"openstack%2Fkayobe~master~Ic49a63d66dfdb8e4235a60c1f01ad70bef6bea43","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ic49a63d66dfdb8e4235a60c1f01ad70bef6bea43","subject":"Update documentation for release tasks","status":"MERGED","created":"2020-02-21 15:03:04.000000000","updated":"2020-04-15 16:26:33.000000000","submitted":"2020-04-15 16:23:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709154-1586967812021-11ac1cc1","meta_rev_id":"2f208387fb782e803b4aa141ac405aa18ce34ff2","_number":709154,"virtual_id_number":709154,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 16:23:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-02 16:24:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-15 14:19:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 16:24:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-15 16:23:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b718f52f23fdfafc2b8ac59398182bee4f2f1fed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 15:03:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c17673e3cc29ded59fa566f54144d54090599cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 15:16:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/077118c15281425d8131b49e5ba1c7fb : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6017672f9f8a41f99ea541d8c665d323 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"473622d9d271be89e93f7bde079fb7600cb58d27","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 16:23:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b9206cfa7edb5600368675182f11647751eff860","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 16:24:03.000000000","message":"Patch Set 2: Code-Review+2\n\nRebased to regenerate preview.","accounts_in_message":[],"_revision_number":2},{"id":"974b44860572fd643daae01520767ab5b201800e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:23:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc8bf320049d4e339a32297019e8a413 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce228afbd2144383bcf49fec5c7c64ee : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":2},{"id":"34ec18bd470ec569a301c8a06e82454e4e187222","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 14:19:28.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cd9bed1d46457676ea4ce7745b48a6eeed53c457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:19:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92beaf32104de4257f0083bf357fa3ec7cf90773","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 16:23:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdcac510291f40d5b5be43a39425dcf3 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0031f7e5f9c64cc8a50ddfa818ceabb3 : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4965bc38557b5d58c92ce695054cf26671b8ddd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 16:23:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2f208387fb782e803b4aa141ac405aa18ce34ff2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 16:26:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b724ee09d80f4fe0b391f087fa344528 : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e1def534c3a50fea3c9b38ab1133a8df34a4bb8","revisions":{"d6381c2d85c68a17be5edae4e3b2c7eb8b628cfb":{"kind":"REWORK","_number":1,"created":"2020-02-21 15:03:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/709154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/54/709154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/709154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/709154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/709154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/54/709154/1"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 15:01:42.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 15:01:42.000000000","tz":0},"subject":"Update documentation for release tasks","message":"Update documentation for release tasks\n\nSince Kayobe is now an official OpenStack deliverable, our releases\ndocumentation page is now outdated. It needs to be updated to document\nthe process using official release tools.\n\nThis change makes the releases page more similar to the Kolla release\nmanagement page:\nhttps://docs.openstack.org/kolla/latest/contributor/release-management.html.\n\nChange-Id: Ic49a63d66dfdb8e4235a60c1f01ad70bef6bea43\nStory: 2007079\nTask: 37990\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6381c2d85c68a17be5edae4e3b2c7eb8b628cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6381c2d85c68a17be5edae4e3b2c7eb8b628cfb"}]},"branch":"refs/heads/master"},"6e1def534c3a50fea3c9b38ab1133a8df34a4bb8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-02 16:23:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/54/709154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/54/709154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/709154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/709154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/54/709154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/54/709154/2"}}},"commit":{"parents":[{"commit":"cdc1d5ada2283a742ee5946b54ac2f87efa5e431","subject":"Fix loading of Monasca Grafana dashboards","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdc1d5ada2283a742ee5946b54ac2f87efa5e431"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-21 15:01:42.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 16:23:43.000000000","tz":0},"subject":"Update documentation for release tasks","message":"Update documentation for release tasks\n\nSince Kayobe is now an official OpenStack deliverable, our releases\ndocumentation page is now outdated. It needs to be updated to document\nthe process using official release tools.\n\nThis change makes the releases page more similar to the Kolla release\nmanagement page:\nhttps://docs.openstack.org/kolla/latest/contributor/release-management.html.\n\nChange-Id: Ic49a63d66dfdb8e4235a60c1f01ad70bef6bea43\nStory: 2007079\nTask: 37990\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e1def534c3a50fea3c9b38ab1133a8df34a4bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e1def534c3a50fea3c9b38ab1133a8df34a4bb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
