)]}'
{"id":"openstack%2Fkayobe~709580","triplet_id":"openstack%2Fkayobe~master~I2547ef2556ca96b614854515069aaca3349cd692","project":"openstack/kayobe","branch":"master","topic":"pip_proxy","hashtags":[],"change_id":"I2547ef2556ca96b614854515069aaca3349cd692","subject":"Allow setting pip_proxy","status":"MERGED","created":"2020-02-24 18:25:28.000000000","updated":"2020-02-28 14:21:22.000000000","submitted":"2020-02-28 14:18:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709580-1582899517389-d353a4dd","meta_rev_id":"79a1d24ec63aead0286b88628630cb642b507a45","_number":709580,"virtual_id_number":709580,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 14:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-02-27 10:26:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-27 14:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-27 14:48:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 10:26:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-27 14:48:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-28 14:18:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7f6a5abe20bc8882e4f1e807da9acc8888354a8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-24 18:25:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f73816b57a5eb92b075285cca1a5310f585b0392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 20:44:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8e51c028e334d18b0c56b80286599cf : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24e7a58affa49d8afbaf79d17b59855 : SUCCESS in 10m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bcebafe52084c7b98b5b37947b5013c : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3a006cfbdca40e79a19cabd6fc1681b : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29f0d8ab937f44e7a9f6ff949518ec43 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c99ddd226e46b38cb63fba80026c99 : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71860858701d48249eb565356e31324d : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd7596c25e2343cca85af02334d6fdf2 : SUCCESS in 13m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/226f81a94d764d5db17f9294871090f8 : SUCCESS in 42m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c22d7622b51e47a6b4f0197578f495aa : FAILURE in 9m 14s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/36c7d0cf1adc432abb387ad4ff0ad441 : SUCCESS in 2h 13m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e7cfae2c001f436eb5c0fbbdcca2a942 : FAILURE in 11m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/15bdaa060bcc48c5b257a9db322fcef8 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e601cb7916eb2a670f65fd2b68dad2b23d96262e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 07:27:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"220edcdda6aadc013dfc498c36047aa00c0ab359","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 07:59:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2cdaf680b9f8043c376f9e2dc6b9d0ef449edf9a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 09:55:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"99096fd0e2ec30d95f5f64ed2393703f81595eae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 09:56:38.000000000","message":"Patch Set 3:\n\nAlso release note and docs please.","accounts_in_message":[],"_revision_number":3},{"id":"435c5f990dcd3e16ef6f6b99d206d021515aa664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:09:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6ae17a7f8fd4ff89a0929ab55b3e247 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd5cef8ee7904e26b3f13c81d75fc796 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd6236db6f9a43ccb7fef9136a056a06 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7c6334d81244249998bdc47b4c01a3b : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce78ec180f9e436b9361d3e76d8c6fab : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3f17d8d9f04f1d82906e8880624cfa : SUCCESS in 8m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99b93d0630cd4973a01a31fc7e49a9b9 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cc898e8a902410e84faba986de3e867 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b35756e262a84f1faf5fe2c38eaa9c21 : SUCCESS in 30m 32s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7ca38f26d7944226a7f31cbf43b76cbb : SUCCESS in 1h 12m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fca51a4967574cb3a142eb2bf878f767 : SUCCESS in 2h 05m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a7915396d9de4d369c62eae648eb0ac7 : SUCCESS in 24m 44s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d42559fbdeab4ba79806cb469e69726f : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":3},{"id":"26a8314bf60940ebba6862a33e794e1f3e2aa87d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 10:38:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e83e63a569a7103a343996bc0bfe40e7805679c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 11:45:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2de6fa38e190766d9bf6275e08a56ea5e5ebf1bd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 11:57:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dee07a6f111aeef36899637ccdf18b47b405809e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 13:57:21.000000000","message":"Patch Set 5:\n\nCould you document in doc/source/configuration/hosts.rst?","accounts_in_message":[],"_revision_number":5},{"id":"7b58a802f756e4474dc216b777da2df4271373f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:25:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e5b1ae0dbea47c3836fb08ecda11dcc : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527c96fa128a4de5a521bdd7f0774a06 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c97e7ef3b24bc4bff6e855523ea665 : SUCCESS in 7m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bc82a2599fa4bc98e1f691c1851ef2f : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc833688ab034b058a3802d22480d30e : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6beae0278e844cfaadf849a6ee76434c : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdb65d72cdbb436785df778894f3dc95 : SUCCESS in 5m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84de147d86b145f3a2378f6848660c04 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a74a5b682ce4fa58a9c82a7a79ea969 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d6e07bf8cd74a418033ecc3372328cc : SUCCESS in 34m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a2ed1e25e457431d9991ec61be5838fc : FAILURE in 8m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ee55ad3fc6a44cf89dc4a0421e657a79 : SUCCESS in 2h 02m 49s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/fff27f4b0e9a4e098cc54a63e8889eb2 : FAILURE in 10m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5bd187c6173b49828f2d75f695c6cb6c : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":5},{"id":"83449ba0299a16c084bd8a4c3fe4cf4b21c584a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 14:43:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"40af41b375c4e00af47743b95d957b762f80468d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:52:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e18480587f643bd955b683e48dea675 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b26e91c41041728da7c0a60909b6df : FAILURE in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e1fb2e565b04e8fa5eacbb6b7db6ef3 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6173cb54fcba47f3bac2a2d4f1b5f7c8 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/becdb13535ec4443a3ad1712fe85d22e : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce60ebf679b5401da327ecb1911d76ed : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6374706f86094f638f0c7d3d15666e50 : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f08d3c0fde74329b7b1a1b82f722cc6 : SUCCESS in 8m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79610fb3a6f44852b531239918f99793 : SUCCESS in 14m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f5ffcccddbe45d98f1d0a4ab22cba10 : SUCCESS in 42m 08s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d89312078ed04ac0a3500a31b44c3773 : SUCCESS in 1h 53m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/83e6055439374356a6525852a9291bca : SUCCESS in 2h 06m 58s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b1a40f91768a46de9789b799fcc26097 : SUCCESS in 30m 46s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b8194920cac641a690c1197ccc17c041 : SUCCESS in 44m 44s","accounts_in_message":[],"_revision_number":6},{"id":"8cf30aa42f890204f08a318f3af8d1eb6e21041f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 17:28:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ef11c3430d3a65024a0c5454c180cbdb65f8a919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 19:49:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a307a33bacd4ea1970d9e0c085ad22a : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2b91f6a9d34d1cb1b6965960139216 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3348ca1814364165b36bd6cc2fe6e35d : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0556643756f241b8a2a193930495e5d6 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f37a94d5eb274e45abdffaa7b9c9bf87 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c08729eea3f4f49ad5c0bf806801a4a : SUCCESS in 14m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dff14dea57f4d9a8e3fc4e1dbd70777 : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec6390a0c1c046f7ad854d7ccfe330d5 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50144bd5cc8645ad94c74d4612efa037 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/582c8f2b91564a6fba20313343ab6a6c : SUCCESS in 39m 28s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/016ff8ec59eb439a8d0a59fe182368a3 : SUCCESS in 1h 17m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f1c01cb88fff4e3a89ae473abaf08c2c : SUCCESS in 1h 57m 28s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c342da6438fb483f9bc6ab4a872d07a0 : SUCCESS in 26m 40s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c80e877cc4d14805899deac8b49444d5 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":7},{"id":"b2dddecbb521b7936d8836f1fdcedffae74b8c49","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-26 10:08:25.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"031b5913fb80c7b8896d73313859e0a9ea65f023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:08:37.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"8d4d5e93b363930fe1009d185a71734aa2af406b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-26 10:10:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4e0ec3e0ec98a25e61f3c0b143798a93305825bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:07:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7520679f83e249fcb7daa416fc565776 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb3c67f44fcd475b85ee95b1817669ce : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93ec4019ef7c4d9eb793afc04842f8cf : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/889319d6637748319c44a0d8f7e3a2cc : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24393ca156004a38869afd9a32ff118a : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1956ff0e204951aa66fbd3cc9c6aa5 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3bfcb4d46284df596263c0c2fd3613d : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79571a36b532451d8f88335d44d94182 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d8afcb18cccc46ceaa59a958c68d9ead : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/392c0896ca2f4bdd90cc1cc07191c714 : SUCCESS in 33m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/215a69d30b1b4898a056ea1f211e440f : SUCCESS in 1h 12m 46s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/09b39adff05a4d93876fa70278766de8 : SUCCESS in 1h 50m 24s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4e7e049b33a148d99ed2f9260cda0d4c : SUCCESS in 23m 41s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/fc397383c71b4672a8cea0d740bd555b : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":9},{"id":"31b63ac075fd458ccce99a5a772628eb96faa90b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 10:26:10.000000000","message":"Patch Set 9: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":9},{"id":"6daa25f96bec4c9af90360df82df9383a19a2f0a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-27 14:48:57.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8e553c5214be786de821d0f24eed2b81ab0c377b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:49:15.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d1ff020aace613d1afc6a69b496837959881d1a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 18:14:39.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e11667695f45dd86e043ab2f5520c3 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63d44dc395274ef4bc239f3f654b4aa7 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e77e13b12fc445ebda51f07faa0f1cf : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ffd7295b9c4106811c13b37c830a86 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8947f77089a3404abaf292e16dd050f7 : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d03bcc75f844078909c675063fc89e9 : SUCCESS in 8m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7de6f7bfee45433580c3eebf7255c72b : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8fe27e41eb848179460193e46dfe633 : RETRY_LIMIT in 39m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5636b07e9daa489eb7086f37ed576cbc : RETRY_LIMIT in 39m 25s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9ee6592b4f064cfc9f2ab921cf31114c : RETRY_LIMIT in 38m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/908a21600b164e99a8753742faed97e1 : SUCCESS in 24m 27s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/cadc8b4b94ff4614845f5d2d55385abd : RETRY_LIMIT in 40m 16s","accounts_in_message":[],"_revision_number":9},{"id":"c80af31e9bc85fd4b387218f8f7bfe9cebd0699a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 09:44:16.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"146455b4083a5a43c93ef55d6e10dd5af558733a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 12:18:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f94b36ec58e4037b4a524693d37717d : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d607ba0c7ac3437bb4de53dc235667ac : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a8eb8e68fb74890ad465ed756c0e4b3 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6d5bf8f80ba465db4f02b9272f7e74f : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c2a0e7e76a64c1d899c36c9409fe7ad : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6066c48a0b74c1aba01447a586a8b02 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae68c8cec8da4616b87412763856e79e : SUCCESS in 6m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd11dccba64a43b8aefb7d93cfbe3a54 : SUCCESS in 8m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/88b01d6196fe46a49ecda242cc7a62ef : SUCCESS in 13m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da9d3cbe99f64cd7b9efb4fbf695e45c : SUCCESS in 40m 37s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6f672ced2d544df596c9299a25850ec8 : SUCCESS in 1h 22m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dcbff8d1f7c34365856eef7ff848d702 : SUCCESS in 2h 30m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/bc06abea6453458a878f68e85c3030eb : SUCCESS in 26m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/6057d8bc52b04816a73c829d288f4904 : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":9},{"id":"8d9bd11ce3d37e498f783efaa6086d0fcf25277c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 12:19:06.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"99613eadb83737ee5391b67d9947f4c31d925b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 14:18:36.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a482e5a57f6a4582a04cc5327e64b051 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c9cdeb80c414cc1a71fb23579a99345 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76ec17f973c544339c2fa481d8513bdd : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fe66e3297e4c139c3dafe7bee6fb33 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/267c7a4d9e204454bb9ce06248e050b5 : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aedcdef34a48453ebdcb7cb9d1b66cc8 : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c55222288c92490fb9d6a2b0f045c375 : SUCCESS in 10m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89c642225b2849d68c4eb351bdb26c19 : SUCCESS in 38m 28s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ab47ca2523ea443ebbb4ab0e63dd47bc : SUCCESS in 1h 26m 33s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e912e345d82240e89d23f4770a008813 : SUCCESS in 1h 58m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e16a80339a7243f4bd7edd00d8861436 : SUCCESS in 24m 44s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/14ec22edf3f64d5c95b7577fb118a60c : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":9},{"id":"cc8606291059cf809213a8013f9513b976393391","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 14:18:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"1a401431a74711bf69165e994d72622592cac84e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 14:21:22.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b80d287752c4f5c8dee1b3968c7d3ab : SUCCESS in 2m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ac5c1c382fa46b7ac8af26eae8f1ef8 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"a0ab81d7527585937a4b3035ace78d3be0e7f4d6","revisions":{"e5a16c89d20cd7f4309ef13045c0de38b6fd8915":{"kind":"REWORK","_number":1,"created":"2020-02-24 18:25:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/1"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:25:25.000000000","tz":60},"subject":"WIP: Allow setting pip_proxy","message":"WIP: Allow setting pip_proxy\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5a16c89d20cd7f4309ef13045c0de38b6fd8915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5a16c89d20cd7f4309ef13045c0de38b6fd8915"}]},"branch":"refs/heads/master"},"5ac9c582a34759bfa14e83f6c430fdb1aafe9114":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-25 07:27:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/2"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 07:27:10.000000000","tz":60},"subject":"WIP: Allow setting pip_proxy","message":"WIP: Allow setting pip_proxy\n\nDepends-On: https://review.opendev.org/709665\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ac9c582a34759bfa14e83f6c430fdb1aafe9114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ac9c582a34759bfa14e83f6c430fdb1aafe9114"}]},"branch":"refs/heads/master"},"7f2930cc5a1cbe57c27edb0e2af0ea3f60927352":{"kind":"REWORK","_number":3,"created":"2020-02-25 07:59:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/3"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 07:58:35.000000000","tz":60},"subject":"Allow setting pip_proxy","message":"Allow setting pip_proxy\n\nDepends-On: https://review.opendev.org/709665\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f2930cc5a1cbe57c27edb0e2af0ea3f60927352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f2930cc5a1cbe57c27edb0e2af0ea3f60927352"}]},"branch":"refs/heads/master"},"c4467ab62d77413dc4fbe008d50e47e834d0b50c":{"kind":"REWORK","_number":4,"created":"2020-02-25 11:45:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/4"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 11:44:07.000000000","tz":60},"subject":"Allow setting pip_proxy","message":"Allow setting pip_proxy\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4467ab62d77413dc4fbe008d50e47e834d0b50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4467ab62d77413dc4fbe008d50e47e834d0b50c"}]},"branch":"refs/heads/master"},"f46e39c147b703508f5c659c4c4a4a9edf9c1b96":{"kind":"REWORK","_number":5,"created":"2020-02-25 11:57:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/5"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 11:57:06.000000000","tz":60},"subject":"Allow setting pip_proxy","message":"Allow setting pip_proxy\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f46e39c147b703508f5c659c4c4a4a9edf9c1b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f46e39c147b703508f5c659c4c4a4a9edf9c1b96"}]},"branch":"refs/heads/master"},"90bbd1701a063092faef34f0c9fb4b7268bb5b29":{"kind":"REWORK","_number":6,"created":"2020-02-25 14:43:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/6"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 14:43:42.000000000","tz":60},"subject":"Allow setting pip_proxy","message":"Allow setting pip_proxy\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90bbd1701a063092faef34f0c9fb4b7268bb5b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90bbd1701a063092faef34f0c9fb4b7268bb5b29"}]},"branch":"refs/heads/master"},"170d182d9e609bd5f005e3cc3156623d4d57397b":{"kind":"REWORK","_number":7,"created":"2020-02-25 17:28:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/7"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 17:27:54.000000000","tz":60},"subject":"Allow setting pip_proxy","message":"Allow setting pip_proxy\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/170d182d9e609bd5f005e3cc3156623d4d57397b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/170d182d9e609bd5f005e3cc3156623d4d57397b"}]},"branch":"refs/heads/master"},"11e66333b9a10a33c52e6d37c39a21a57aa07c4e":{"kind":"NO_CHANGE","_number":8,"created":"2020-02-26 10:08:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/8"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-26 10:08:14.000000000","tz":60},"subject":"Allow setting pip_proxy","message":"Allow setting pip_proxy\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11e66333b9a10a33c52e6d37c39a21a57aa07c4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11e66333b9a10a33c52e6d37c39a21a57aa07c4e"}]},"branch":"refs/heads/master"},"a0ab81d7527585937a4b3035ace78d3be0e7f4d6":{"kind":"REWORK","_number":9,"created":"2020-02-26 10:10:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/709580/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/709580/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/709580/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/709580/9"}}},"commit":{"parents":[{"commit":"2e842ab3f5c7b737b547f4c2c718f96294f582d5","subject":"Remove activate-virtualenv and deactivate-virtualenv roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e842ab3f5c7b737b547f4c2c718f96294f582d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-24 18:24:50.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-26 10:10:01.000000000","tz":60},"subject":"Allow setting pip_proxy","message":"Allow setting pip_proxy\n\nChange-Id: I2547ef2556ca96b614854515069aaca3349cd692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0ab81d7527585937a4b3035ace78d3be0e7f4d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a0ab81d7527585937a4b3035ace78d3be0e7f4d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
