)]}'
{"id":"openstack%2Fkayobe~709689","triplet_id":"openstack%2Fkayobe~master~I6f065073feee42b2a0921a4aea98f7004fe5f7bf","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I6f065073feee42b2a0921a4aea98f7004fe5f7bf","subject":"Add framework to deploy user-defined containers on seed","status":"MERGED","created":"2020-02-25 10:31:42.000000000","updated":"2020-06-11 10:55:16.000000000","submitted":"2020-06-11 10:52:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":60,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709689-1591872739590-1e1c92d6","meta_rev_id":"4a924972c6d7d83f958989a81370f9a8272edace","_number":709689,"virtual_id_number":709689,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 10:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-10 18:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-06-11 08:26:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-06-11 08:26:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-06-10 18:36:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 18:42:54.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-11 08:26:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-11 10:52:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ecf3e9ddf95a4b5ac5af7b5368b30fdbad7244d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 10:31:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43eba83c1ca207a559f0c950c488e0d330ea1955","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 10:42:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"246f391e9a8ae0275f88c805123bc254369f972e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 12:22:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b247158b844b7e934cf4fd05dcb4fa258ad7c3f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:19:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e674a9aad0ca447c986293163933b9cf : FAILURE in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32e6a81b93d446196755cac94e56164 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f62ff005d2f4c339c763140bfd1aa34 : FAILURE in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5fa6e3b6ea64c538267ca02bbfe3c7f : FAILURE in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/231954329008422b8c0679ffe17bb9bb : FAILURE in 9m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14225bbc30794026a7aff8038ad8a294 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cab5dc2e90c473e8585e10453df89fa : SUCCESS in 5m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a697bbef4d674a30b7b3bfdff0e7506e : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c002e04172e447309a741032f8630697 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d2e2cce2cd8244989bde52e5165cf82e : SUCCESS in 46m 42s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ad80869ea0914b72a8d1bb8b75a9f130 : SUCCESS in 1h 17m 40s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9abc816d20ac49349f2479142a6c2589 : SUCCESS in 1h 50m 02s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8fee219a95f44f889ee372477d399de9 : SUCCESS in 28m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/382e6b43be8945638e12f890d594ca5a : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":3},{"id":"5b0df1cb0be55093c9ce95520ff1ccb633075c48","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 15:10:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"381a260c59ce47ea2eeb3dc7baeb8849bac4a462","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 17:09:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52bffaca7d984bed8997fb837e09e659 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3dc7c38cc74c6ca6f4f00e054a0603 : FAILURE in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71d20c0ac79647b1a847a86384a7093a : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be54d22c13204bb4bf23a67c86f14a85 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81045d23ed2c49249e174fd4bc05b3cc : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a4336e275c8448b91df721e24e9d634 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/150476988f3147c4bfdc98edccd42bc4 : SUCCESS in 5m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c2d1f7c1dbe4c9b96f7ce8c9f73806f : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2a6c152328d3453fbf3091dc0a4156da : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ff83f050c784da2b40978874fb8eb22 : SUCCESS in 42m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8df3625ec2ba4ae0bf2cc623f3ca0802 : SUCCESS in 1h 22m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/711323d91b6a4d3cb6d5f5610894bf32 : SUCCESS in 1h 49m 24s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c957b48e13f243d6ac86ac428df9b1a7 : SUCCESS in 28m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e3842136d404405382528239802b323d : SUCCESS in 48m 20s","accounts_in_message":[],"_revision_number":4},{"id":"28159bb2126be53b9c78f7d66de4b0e332bc635b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-25 17:40:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5a2cc1c896c74fe30bf886685a8a358793b26408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 19:47:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd33a40443d34e4cb7f519c64f89b9a2 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3f181445fe4dfda3c894fb8803bda9 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c866ef83d4f4c43bacb49e6ff777661 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/984ba76add3e4323893e98ba8688ad03 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51f141fad7e943cdb62f75901a222f4a : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b56f179a814a2c93b94eed8409f173 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdaec8d342434ef2b7f0606f1536c640 : SUCCESS in 5m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa0a01ab83c44eb8ad423f332cebcf94 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea7aeb7e60ee443db15591820384b56d : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4955ce7f115b40d78bb54a0f368e701a : SUCCESS in 34m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/098383a104894dcc9d743123fd9a0217 : SUCCESS in 1h 14m 25s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/26ebdc0993784dfaa8e77481b906b1ed : FAILURE in 1h 50m 26s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9e9983e7470e4b479edb0f7f0d78dfda : SUCCESS in 30m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b3d16280100d4b409dd1032802e2967f : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":5},{"id":"2676b88a000628d065721e3cf20c56eb360be3c4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-02-25 20:37:44.000000000","message":"Patch Set 5: Code-Review-1\n\nThere is a bootstrapping issue as the seed requires internet access to set up docker in the first place. Doing a `kayobe seed host configure` without a gateway:\n\nTASK [Ensure the python-virtualenv package is installed] ********************************************************************************************************************\nfatal: [seed-0]: FAILED! \u003d\u003e {\"changed\": false, \"msg\": \"Failure talking to yum: Cannot find a valid baseurl for repo: base/7/x86_64\"}\n\nI believe the intention was to remove the dependency of having a gateway on the provisioning interface. The default network interfaces for the seed are the overcloud out of band and the provisioning interface.\n\nYou should be able to run through the commands in the deployment section without having to resort to a temporary workaround of setting a gateway on the provisioning interface, only to remove it later.","accounts_in_message":[],"_revision_number":5},{"id":"71dc1600fdd81c07a193dc2661488f8dd65064da","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-02-25 20:42:48.000000000","message":"Patch Set 5: -Code-Review\n\nOK, technically this change isn\u0027t explicitly about running a squid proxy and it will work for other use cases.","accounts_in_message":[],"_revision_number":5},{"id":"0b9926b38d617e0c2091822f4b18d2973d9049b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-26 08:44:56.000000000","message":"Patch Set 5:\n\nrecheck (500 from Docker API on upgrade job)","accounts_in_message":[],"_revision_number":5},{"id":"c43f47bc09d570e6112b2dd069b92bf7bbfd58c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:55:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d275ae9a3344a5097e7f731cd6e5b9a : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0313e68e910a45d9bf8b988544df6872 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e394a7f7f8245e3b836b057d1607620 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de190a8a14394982beb7a4b75f2eb761 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/393ac2eb7c9b4132a1478497c6acecd0 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e795b92aea44817a60492d37ce54990 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3283a9674c684b6db7878028ba052750 : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c97cb4be3aab4b99adc4dadbc0a2e302 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57ca4a1cfbcc4d0e947d1d6f601eb23a : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bad75530f6d145088c5cad86cd6e9587 : SUCCESS in 39m 58s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/cdb53b52a8184cd897c0c4b6a5a279ab : SUCCESS in 1h 20m 28s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fd17073abbce4b6dad4475a9818e3a49 : SUCCESS in 2h 06m 04s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/958c32bbc9534ea5a673c766b2786bb3 : SUCCESS in 31m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/85827117f96946d4901dd87b1e9d5b98 : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":5},{"id":"78a23969d194e88e5832238c78aae37d3e68b164","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-02-26 12:51:55.000000000","message":"Patch Set 5: Code-Review-1\n\nTASK [deploy-containers : Deploy containers (loop)] *************************************************************************************************************************fatal: [seed-0]: FAILED! \u003d\u003e {\"reason\": \"Syntax Error while loading YAML.\\n  expected \u003cblock end\u003e, but found \u0027\u003cscalar\u003e\u0027\\n\\nThe error appears to be in \u0027/home/stackhpc/kayobe-env/src/kayobe/ansible/roles/deploy-containers/tasks/deploy.yml\u0027: line 2, column 31, but may\\nbe elsewhere in the file depending on the exact syntax problem.\\n\\nThe offending line appears to be:\\n\\n---\\n- name: [{{ container_key }}] Ensure we have latest image\\n                              ^ here\\nWe could be wrong, but this one looks like it might be an issue with\\nmissing quotes. Always quote template expression brackets when they\\nstart a value. For instance:\\n\\n    with_items:\\n      - {{ foo }}\\n\\nShould be written as:\\n\\n    with_items:\\n      - \\\"{{ foo }}\\\"\\n\"}\n\nrequired this patch:\nhttp://paste.openstack.org/show/790026/","accounts_in_message":[],"_revision_number":5},{"id":"9d1e73091365bb0036e9447b3f952bbe34f5fda9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-26 13:01:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e7636f705ae8bdd4a315c9eb594c93ad3c381daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 15:37:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97e47ae6b15c47d08634c70396633998 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2018713ee55341bc97acfe2966db5920 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae9c0badb0974b11b2ea6a35bdb39570 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/683400e4499e4956a1b803ae88369f14 : SUCCESS in 16m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/490bab9201784091b873c40a5b0aba37 : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b66af9c5c9946ce837abfc456fb4a26 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bb6f4e8a6844a27b611c6428f337328 : SUCCESS in 7m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17697b78e85341b790256aedb8b689f9 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20827af6f45a4e2d8880028cd28f5bd4 : SUCCESS in 19m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73fe48f95cb64a7987a97357bd77b309 : SUCCESS in 30m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7c930ef5eda840dbab9ccecc73091b9c : SUCCESS in 1h 16m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/61f1dd79f4244d828f8de5bb6f13afbd : TIMED_OUT in 2h 32m 09s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/543cdcd9261d400e957b3db9c16f940b : SUCCESS in 24m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b4a86871be3a48c69633fadaeb258903 : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":6},{"id":"fb5a678894b5fd65f13ae0313f9a7eee2d29ebe4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-27 10:45:30.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"196de73c52cffc3691095f1621944a1a0cfee0ed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-27 12:57:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b241d7b3b37b6e232373d0304fd3dde8f09ebb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 18:14:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/427c81668047410e8926e1bf0e0fb9a5 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9e0b5c409b408cba938a77a82ec7a7 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3ba8f815ce0430c92c0278b1cf74db8 : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4b9385c34894b6592f380d2a44ec194 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93aeebb4f3fa4d5f9aed0d2064f3f408 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72215e7f2ec241a899ea0051360c6357 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c534ee1e97a4c0e969fe39c4c1d25cf : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3a0787f7d894e32ba2a6a975c03bee1 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b9bc4470c62494fab6444c467e88872 : SUCCESS in 12m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f702f99ae004f7dab2747b6e729c823 : SUCCESS in 40m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/681cf1c67e0d4644a32d540371f57783 : RETRY_LIMIT in 17m 59s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/44eaffe4646a46dfac3c71b5629485e5 : RETRY_LIMIT in 14m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c3e7ac93e15849ab91e55cf318cf7ec4 : SUCCESS in 25m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/846917e476d843c3ad7fa3884eeb762a : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":8},{"id":"d15d179622559c1059f877914dd09acddf976f15","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-02-28 11:58:13.000000000","message":"Patch Set 8: Code-Review-1\n\nTASK [deploy-containers : [ntp] Start container] ****************************************************************************************************************************\nfatal: [seed-0]: FAILED! \u003d\u003e \n  msg: \u0027template error while templating string: unexpected \u0027\u0027}\u0027\u0027, expected \u0027\u0027)\u0027\u0027. String: {{ container_config.capabilities | default(container_config_defaults.capabilities }}\u0027\n\nmissing a bracket","accounts_in_message":[],"_revision_number":8},{"id":"e760cf16428a7c545144d7b18794d457123c0c34","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 18:15:52.000000000","message":"Patch Set 8: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":8},{"id":"dc1f040f17f5bdda66ee5aadf9e735f60ff4f1e1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-17 13:07:36.000000000","message":"Patch Set 8:\n\n(16 comments)","accounts_in_message":[],"_revision_number":8},{"id":"86922daf4cb7bad4eadc8105cde1b34033248e8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-17 13:07:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"cd9cce993825e95daa07bb3a0b6ddad37d7c610e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 14:31:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94f5efdd391b4ff99b5015632c1bc359 : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588868a12c07499f9e4e942227fb4bd4 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4648cb10e3c14ffbacba81b0869945de : FAILURE in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c3aa788bfaf4cf5951311f7f1707ee3 : FAILURE in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d71f7f295a464b74879bc4b322be0916 : FAILURE in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b142475b942f460eb2660932a371c6a5 : FAILURE in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1427b3a76e7a45a5bd80b5a378fe96fc : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31310a6316b241eba2701bb4d287d113 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93a434907b7949ddb845da66262c3247 : SUCCESS in 13m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f3d1abfd5b1432b8197eb4914bb288c : SUCCESS in 27m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/37bc6bd27a614ec99643e46832b7bd98 : SUCCESS in 1h 13m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0d3c9a1c880f47f68384704813d52e12 : SUCCESS in 18m 09s","accounts_in_message":[],"_revision_number":9},{"id":"248e6624c28ac2d895f502aef25cf671ed9cbf00","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-18 07:14:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"92d81a08503e81f5257e26fded9c4b77b1df9fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 09:23:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76ee18e2d925437fbbc0b1ef3cc78494 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3a5b0566ddf47a1806d7d528b289a7f : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28f933d980774892b69d55607af0b8a1 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f154d0c84ba64db696ea88b3487f57ef : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78c88c2d6fc24050954488de7e09b534 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86aca9080c74ea18237db26866c8f25 : FAILURE in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e304a69f95a9448a8c324cc75df9f085 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/69802914954b47b7bc777d6540cb97f3 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3a807961a6ca491696267a2b2d8ab0f6 : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ea6f2fa8cc44315a5773c9e7ee3ef8a : SUCCESS in 33m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/42151c19eae74dbda0020b6aba1413a5 : TIMED_OUT in 2h 02m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/525cf125831949c1a649bc8cbc67786d : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":10},{"id":"969d5968731ebe243594c1049e381064d3eff8eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-18 09:52:51.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e7fbc5a883049880a4354982e1d125068aa054d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 11:07:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c76ef338581b49848ed9f3262e4b4d91 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8549ac524f1244ba95a4970fa28095b5 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12b03d31f1fb4317983fd4402b2db769 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6be0bbfd7f146bcbcaa8da80b64edcf : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8257891b18344b5c8864def6c386e29e : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da548ea16ee4d6f8fca7cc262b96eec : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38e77cb8af024cbbae769be3d6bc1a37 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30e2eb2608014ce8bc96b1bf865fdf0e : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37d521a8089841d3a8f65ac6cd9449ed : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b09dcde26bd849bbb82c204e4b2eedbd : SUCCESS in 34m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1d0023faaff94dac9b4a16687500c1e4 : SUCCESS in 1h 08m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5835e66097674f349fb9c083baedd270 : SUCCESS in 17m 09s","accounts_in_message":[],"_revision_number":11},{"id":"7f33d53f5679787c56991e61df3763f1fb6c5cd8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 18:14:01.000000000","message":"Patch Set 11:\n\n(7 comments)","accounts_in_message":[],"_revision_number":11},{"id":"17605006a84c897d4986f669d7540eeb6cae93ae","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-20 06:47:16.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"208c077fb0b5d2f9ac5514e67eb41ee086db416a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 08:57:33.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b73fc6eddb7423993f88601e8805790 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83fd3945b4bd42ee86bc812a0d77802a : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f05fe43405624f93bb0b2bc38946c037 : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83d4029b10c54d848d99a328d68bbcfe : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3d26a18cb1f4d469f64ba9cf6baece3 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa50244e9ff4dde885a06f894448653 : FAILURE in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6197c14ae274a5a9f847746cc512373 : SUCCESS in 5m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9df908b6ce64715852f9fa1c67f6fec : SUCCESS in 8m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8641a76068754740a5967d7b138f5f7d : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d04f7745c8ad46ee9c0c38eb4b23c6ea : SUCCESS in 31m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ff54c556ecb49d5998bd3533f3453b4 : SUCCESS in 1h 14m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5b0b49cbffef4f28b7ee9e71704416a2 : SUCCESS in 21m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fdb56d40755245a2a9c5247cf380f9fa : SUCCESS in 2h 00m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/34c3c414211f47239bf0a0e515e65d6a : SUCCESS in 21m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f9b6dd99a6484bb983b1742b53266a93 : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":12},{"id":"12ac1783eeb0e5956de848151c7eb3ad87b3fb48","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-20 09:22:43.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d0ec0ce6df324c5420aa862de69a1811d1dbe01f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-20 09:25:14.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"8d4390041e2ece1fdf3c6da288020e7f2642868d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-20 10:11:53.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"4df887ea3a074e0e1897e6392f72425284158ae2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-20 10:12:56.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a6fdbcc55ba51c3ced3310689c9a15f35fd6cf76","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-20 12:21:54.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"2526a0ab9d83b63be49f337667f41595322d07ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 14:21:12.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3831fcb97874c589185d73d861149b4 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36335692f37462391178be13df447a5 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfa64c8d99e2499fb529a543ba0bf4ca : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f283b9f3ea54b2f9a6e7d596395d31c : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6563212033214af0a5248ad583cce76e : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8659422312480699821e404428b2ee : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/222a5f6901f244cfba41555bd3adfb0a : SUCCESS in 4m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d8d43a370450424d964f5f7548273a50 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7342e7d1eb3745269222232ea0c80278 : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2464f10fb2af4b42af3eb11bd146ad2c : SUCCESS in 35m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/315aeac430264e39b54a343f27ecfee1 : SUCCESS in 1h 19m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/80d1b2ad24774ab0b4d508c02a835e80 : SUCCESS in 23m 18s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/03a1c12fe7c442f59234fcf2651d0c87 : SUCCESS in 1h 58m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/25afc9946fcc428f93871d3e0d30ad79 : SUCCESS in 21m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ba50e841372a494f93233c8751fd6ba6 : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":16},{"id":"c673d5f0c311d965b80f7e7817a639c199ab4cdf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 12:55:00.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"52f5e406b66bda63a7b410affb7963ed226f9bfb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 12:55:05.000000000","message":"Patch Set 16: Code-Review-1","accounts_in_message":[],"_revision_number":16},{"id":"79c4fd754b3448bb8bd075cc62e23dd4cc47d940","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-04 07:20:17.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"2ad6c50b4271891600b68331b38b68e8868b386e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-04 07:24:16.000000000","message":"Patch Set 16:\n\n(9 comments)","accounts_in_message":[],"_revision_number":16},{"id":"99e7158166f6940ce7550282ceacba18bdce3010","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 08:13:07.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e3b034c1ee2d75e5d2d142c52bd8221550d60232","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 08:15:03.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"130572e3a9ceec2258a05cd1299d9c47a55af756","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-04 08:45:58.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"e2ca2cc1858ac9b5d7a2c54b53d71462f79e6d62","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-04 08:50:17.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a032d6dc70c9e216fa93224881ef8c17cdd8af2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:26:51.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8ebae57b93d4c33bc9acb208c9b8627 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006a7678e5a446c796a1aeb7a6f26341 : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2429a6387d424b609d82335cdd0cadc5 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16d67d0009764906958779adc48ee338 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f01002007d3349bfaafc52e3c200f5b5 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da26e3a2d8248c39019390ef10819b8 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/984782435b5c4305b13cb5b014b19dc2 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aca665a4e34e4207a00dc2b481483f4f : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9713c183f8f74f15947c32d2a3a70ff1 : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ce7a6d589cd4168b6f0e75b542f2341 : SUCCESS in 47m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/460fccc380ad4b12986775f7e4bdbaa8 : SUCCESS in 1h 12m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8b1735a75314400baaa2585dee04ccd6 : SUCCESS in 21m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b664ab8181604f019c55a95c9fd1c725 : SUCCESS in 2h 02m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ee2bd2517e864ecdad3d2955c7fa15c8 : SUCCESS in 25m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1ac30f27028e4ccf922ec78ee35455ae : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":17},{"id":"074fc2c2f4755120d3328af5f07ef9f397a898db","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-04 09:36:46.000000000","message":"Patch Set 17: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":17},{"id":"0bf36bb1458b75ca9495e67344faa0d2573dd400","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 10:43:00.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"ef4c954b412da145a830296e9c2e1f33641e7bc1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-10 15:34:23.000000000","message":"Patch Set 17: -Code-Review Review-Priority+1","accounts_in_message":[],"_revision_number":17},{"id":"5e9bbb1871303be640990e0c518bdb369bd7f5e7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-10 18:36:21.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"0d37594abc9128c2b2eb21023ffb39f63dbbe14a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-10 18:42:54.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"9fdb5b9bebcd9af647e350f67e2d270a3df15732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 20:35:50.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d44fea2542d4fa6a892d1fb7f185e5f : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44fd544337ef4b19b6d144a9b2a513fb : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8fda1e8cf44c58bbcd16b46ed20a1b : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50be897614084086b0601255482ac14b : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139a3fee8ef244de81e6f03f09eae3fe : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a1d4554e5c747cf805f33c43801dd11 : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e68fcbe46c634a0daa7caff5b02b8ef8 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/249b12eeceab4e53a168bc5d65a5ab13 : SUCCESS in 9m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f9c2cc290104fd8b3bbaaf368cc9952 : SUCCESS in 35m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8e87b77399754276a9c36ce04fcdc839 : SUCCESS in 1h 14m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5c036e2213504baf8ada42d913661a6e : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/94067a42cd51490583f1c4d4d8e67e3c : SUCCESS in 1h 53m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/87be2ed8b11f419ab7d48ccc6367f268 : SUCCESS in 19m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/47c430809ac74126b4bfda3b34d086ab : SUCCESS in 28m 51s","accounts_in_message":[],"_revision_number":18},{"id":"ce9e78d002923d3a4808ae76296fba962b82ca15","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 08:26:02.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"c621b8b7ceeae2778852be32796bfde57a3095f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 08:56:22.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"2e3ded7ec02ff5402d04ae405e6a3cb4a282c587","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 10:52:19.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347b347a20184fc785735d5ce3f4b98d : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3018099df0f4d91abf92a8f63d9de57 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/303c50e7893941ac84dd33842a32ffdb : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b8e5f7d59d400682ffc5cc2fd0b133 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b64dfcd3e3b24746babd73287bd6c53e : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/686797fd73e94773b4aa5618ed747736 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b4d7ceb3a6740e8a19441ae3b3a5d79 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9154485dcb64459f9a7be199d163fb78 : SUCCESS in 34m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fc44430646e944f7a68db4dbf2ac9502 : SUCCESS in 1h 06m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9b49b03acff14f95868110393f1f5d16 : SUCCESS in 18m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5e5f97ef7a9e49128664284361d7cea6 : SUCCESS in 1h 51m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e072f59c9ad844059fc1c029cebd353e : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db80c8a3be934177af099df48717d29a : SUCCESS in 27m 10s","accounts_in_message":[],"_revision_number":18},{"id":"3d11f894cf1e000a5a1d880a8fb1b18d7c37b3f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 10:52:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"ebebc2aa3287fa558bb3a7d9525e7df391916b7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 10:53:54.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df194affb86e4844b78bc850a9cc585d : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/049cab6b760c4fac93a2a4ad3c1cf22c : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":18},{"id":"4a924972c6d7d83f958989a81370f9a8272edace","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 10:55:16.000000000","message":"Patch Set 18: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit af8fcdb9f9a5a132f5d786d979c6758dbe28cf8e","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"aefd6e5a56ba9b172a8111f09c621afd09041a47","revisions":{"572c93810acca501e435b11b85a3dccca9012afd":{"kind":"REWORK","_number":1,"created":"2020-02-25 10:31:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/1"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:38.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/572c93810acca501e435b11b85a3dccca9012afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/572c93810acca501e435b11b85a3dccca9012afd"}]},"branch":"refs/heads/master"},"a28b4d5eaa6f6ad6d84d93bbbd327700ea15e341":{"kind":"REWORK","_number":2,"created":"2020-02-25 10:42:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/2"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:42:39.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a28b4d5eaa6f6ad6d84d93bbbd327700ea15e341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a28b4d5eaa6f6ad6d84d93bbbd327700ea15e341"}]},"branch":"refs/heads/master"},"25f0b1d7095c7cc114df4a06f340a7feea70bf3a":{"kind":"REWORK","_number":3,"created":"2020-02-25 12:22:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/3"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 12:22:32.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25f0b1d7095c7cc114df4a06f340a7feea70bf3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25f0b1d7095c7cc114df4a06f340a7feea70bf3a"}]},"branch":"refs/heads/master"},"8eaab5e7dade78844dbc2e95e05b2ccfb1822093":{"kind":"REWORK","_number":4,"created":"2020-02-25 15:10:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/4"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 15:10:14.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8eaab5e7dade78844dbc2e95e05b2ccfb1822093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8eaab5e7dade78844dbc2e95e05b2ccfb1822093"}]},"branch":"refs/heads/master"},"7c8572b24d6952294225438415b2a88ad0b26703":{"kind":"REWORK","_number":5,"created":"2020-02-25 17:40:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/5"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 17:39:56.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c8572b24d6952294225438415b2a88ad0b26703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c8572b24d6952294225438415b2a88ad0b26703"}]},"branch":"refs/heads/master"},"0225acde39bcf4a16529d14099c4b0afacfeb280":{"kind":"REWORK","_number":6,"created":"2020-02-26 13:01:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/6"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-26 13:01:50.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0225acde39bcf4a16529d14099c4b0afacfeb280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0225acde39bcf4a16529d14099c4b0afacfeb280"}]},"branch":"refs/heads/master"},"3bdeaecbb407105465cd23b4cea08acfa12197c1":{"kind":"REWORK","_number":7,"created":"2020-02-27 10:45:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/7"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-27 10:45:23.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bdeaecbb407105465cd23b4cea08acfa12197c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bdeaecbb407105465cd23b4cea08acfa12197c1"}]},"branch":"refs/heads/master"},"32f238cd5485d093c559526e6434028529577c1e":{"kind":"REWORK","_number":8,"created":"2020-02-27 12:57:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/8"}}},"commit":{"parents":[{"commit":"3de40b57ba69d3362fd391db808d9a37e103df07","subject":"Merge \"Use latest release of StackHPC iDRAC role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3de40b57ba69d3362fd391db808d9a37e103df07"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-27 12:57:37.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32f238cd5485d093c559526e6434028529577c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32f238cd5485d093c559526e6434028529577c1e"}]},"branch":"refs/heads/master"},"15c92d4a32adb66eb45dcd1853ca2b93b9320540":{"kind":"REWORK","_number":9,"created":"2020-03-17 13:07:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/9"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-03-17 13:07:06.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15c92d4a32adb66eb45dcd1853ca2b93b9320540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15c92d4a32adb66eb45dcd1853ca2b93b9320540"}]},"branch":"refs/heads/master"},"02510eae9daf1374e6cd58c507689c1ded1dd7a6":{"kind":"REWORK","_number":10,"created":"2020-03-18 07:14:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/10"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-03-18 07:14:12.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02510eae9daf1374e6cd58c507689c1ded1dd7a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02510eae9daf1374e6cd58c507689c1ded1dd7a6"}]},"branch":"refs/heads/master"},"beee85c7748706ca388a79d002557964f5847e87":{"kind":"REWORK","_number":11,"created":"2020-03-18 09:52:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/11"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-03-18 09:52:42.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/beee85c7748706ca388a79d002557964f5847e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/beee85c7748706ca388a79d002557964f5847e87"}]},"branch":"refs/heads/master"},"990805602333e8c16616c7c205ce0807cf4347cb":{"kind":"REWORK","_number":12,"created":"2020-05-20 06:47:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/12"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-20 06:47:14.000000000","tz":120},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/990805602333e8c16616c7c205ce0807cf4347cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/990805602333e8c16616c7c205ce0807cf4347cb"}]},"branch":"refs/heads/master"},"3c03dbc0fcdc6129b6664f1232a151417536b9a7":{"kind":"REWORK","_number":13,"created":"2020-05-20 09:22:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/13"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-20 09:22:31.000000000","tz":120},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c03dbc0fcdc6129b6664f1232a151417536b9a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c03dbc0fcdc6129b6664f1232a151417536b9a7"}]},"branch":"refs/heads/master"},"5166eac2308209f7b91efd70ed670c9566c3c1df":{"kind":"REWORK","_number":14,"created":"2020-05-20 10:11:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/14"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-20 10:11:45.000000000","tz":120},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5166eac2308209f7b91efd70ed670c9566c3c1df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5166eac2308209f7b91efd70ed670c9566c3c1df"}]},"branch":"refs/heads/master"},"38893a5a2dfa60e1fcf40dcb3f750df9e200e4c9":{"kind":"REWORK","_number":15,"created":"2020-05-20 10:12:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/15"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-20 10:12:48.000000000","tz":120},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38893a5a2dfa60e1fcf40dcb3f750df9e200e4c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38893a5a2dfa60e1fcf40dcb3f750df9e200e4c9"}]},"branch":"refs/heads/master"},"6f0ef14d411c5d663bb7961275af5a26432f26fd":{"kind":"REWORK","_number":16,"created":"2020-05-20 12:21:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/16"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-20 12:21:44.000000000","tz":120},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f0ef14d411c5d663bb7961275af5a26432f26fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f0ef14d411c5d663bb7961275af5a26432f26fd"}]},"branch":"refs/heads/master"},"e31ec161bf555afbe38a46694b5373968d4f6c2f":{"kind":"REWORK","_number":17,"created":"2020-06-04 07:20:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/89/709689/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/17"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-04 07:20:15.000000000","tz":120},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e31ec161bf555afbe38a46694b5373968d4f6c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e31ec161bf555afbe38a46694b5373968d4f6c2f"}]},"branch":"refs/heads/master"},"aefd6e5a56ba9b172a8111f09c621afd09041a47":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2020-06-10 18:36:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/89/709689/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/709689/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/709689/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/709689/18"}}},"commit":{"parents":[{"commit":"c96803d8d1974c87e0dca004b4e360d48ee0fe8f","subject":"Disable broken linting of literalinclude sphinx directive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c96803d8d1974c87e0dca004b4e360d48ee0fe8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-10 18:34:04.000000000","tz":60},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aefd6e5a56ba9b172a8111f09c621afd09041a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aefd6e5a56ba9b172a8111f09c621afd09041a47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
