)]}'
{"id":"openstack%2Fkayobe~709786","triplet_id":"openstack%2Fkayobe~master~Ie597e1bee46e1f7b7a7129d37e70d88c08cade74","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ie597e1bee46e1f7b7a7129d37e70d88c08cade74","subject":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0","status":"MERGED","created":"2020-02-25 16:51:39.000000000","updated":"2020-02-28 21:23:45.000000000","submitted":"2020-02-28 21:22:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709786-1582924928565-97979107","meta_rev_id":"24359f92541e07c9a485d58ae2d3821a12fcf788","_number":709786,"virtual_id_number":709786,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-02-28 09:40:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 21:22:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-02-27 15:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-27 15:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-02-28 19:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 16:42:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-28 19:17:28.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-28 21:22:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"165cd1be925a26560f9e6306f6c19d17389a8a07","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 16:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c93c4fe6ae0a47f568ff590bbf62266318ebb21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 19:36:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b08da64c8f64c508ddfae0515756965 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba856a71829f477580b964bb05c61b6c : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74fec3c6673348a9998269c26c9364a9 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a875e4a6ea14de28072c4a9b9c4af22 : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a6c8ee8caaa4386b09caa2c882d45b1 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1112e4eb4546b6bb5848a4a60ff417 : SUCCESS in 11m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a58b7dabb53e4708a31a7815e821f398 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8887fbef52434a2481a0ac12529aae60 : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4e8da3d429b24c12a7278406a57c3397 : SUCCESS in 41m 26s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/933766329663402b9d2f74f4785e7ad2 : SUCCESS in 1h 29m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8d1524cbd8d7460d954881b0247d0082 : TIMED_OUT in 2h 31m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/35b07f23da3f4d65a3288f37ccdfc4aa : SUCCESS in 31m 49s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0dca9e6e1fa44d8ab2b66c159604f4f8 : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b0a4989c2fb1258118b38909089ca42f416dc986","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-26 09:40:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fbe722bc76fc50887c580aa02c83b8c93298e1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 11:59:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d67d0e6d4b54e0ab17d677b94899eb1 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc252c73d0f042798cb4b5261240ffba : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/438ec707464a4dbbae4d904a24dd4246 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e41306370f74cd285ecd16bdc8bcc4f : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cb2b12bbb2c490196044dbaa60839e0 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13639e078c8a47e9ab946605208cddae : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c5c2be4a5444f73820f0dfaf54c30f6 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/83c2743c27c749168062d9df0b807aef : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e37349d94df74ac8b9ff8aaf1ec1d8c2 : SUCCESS in 43m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9259837a0e804ece95c98c853569cc44 : SUCCESS in 1h 14m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2a726f6f97b2472886684f856b2819e0 : SUCCESS in 2h 14m 26s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/433702e6beac41149b8a43284671d6c4 : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/99633f0586094814b6b1bd8fff6b2a97 : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":1},{"id":"33b920060b520ff704f8047f0dba135b7996abeb","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-26 16:27:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19b5077e8f5fa38665cfc334bcc66c977655abb6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-26 16:42:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5bd8c98ed9712eb35dc6a17b10bc65458b12b06","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 11:24:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"87c3928523051abd2fa061b9f1e192b4f9135126","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 15:39:38.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2b7f413836f6c74443703868a841b6f4ba00be5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 22:09:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4625b30fefa841fa8196ac5d2d159b98 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5e3a1298f54f0e804b0ff5130b96c3 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbcb404e900a4c2e900bfd4c241f0830 : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c70343755119404aa882c0833835033b : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f196138bb654ac3b5c5c21402287ded : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be4b8708084b41cd96df4d4fd14b5262 : SUCCESS in 9m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/893863a2a4114de8a63fea667ea14a93 : SUCCESS in 8m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8296b5b654e74aaea71604dba8aefab6 : SUCCESS in 14m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c8f87f3355f749ed87e2f486ebe40f81 : SUCCESS in 41m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5cb9562ecaf0429ebc26ac8b22ab7a63 : SUCCESS in 1h 42m 41s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b352f8c6ac8b461586e38bd6a4dcefc6 : TIMED_OUT in 2h 32m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/aa28659d794e4734b2ddfef2830d25a0 : SUCCESS in 37m 45s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/df96a70be7e94f1bb217163d56274fb8 : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":3},{"id":"d91bb53e60a6a0f410700b65509834f84f2c22ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 09:40:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c43ceefcf7659003f67d6f11b503b84e2bd592d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 12:18:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65249d2672614e459b4a08010b4cffc5 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c2b0995c184458382e8e8b4e6afd621 : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04b975eec0274fadb02c39aa0991436d : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d11c901636e34892ad9e4eff7767ad7e : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1a2e44aeb784b698425ce2a530d047b : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d283c5778e48d98e6ad7b1bd0829b3 : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7dd3d74109f846e1ace0b98af886b88d : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/08e087264dac41e786fe227cca4b3fbe : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca465eca9bff44a6afac2f49939aaf9e : SUCCESS in 39m 12s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/220456148129404d8813d5145d10892b : SUCCESS in 1h 18m 21s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1a505ae7555f4ed7ba60b1f3268eb7eb : SUCCESS in 2h 27m 29s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e90333312a7f42c7b0e503e8b2701abf : SUCCESS in 26m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/385f55ef84c44782b22d7907eba85f68 : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":3},{"id":"221a0cf61533df075246eb13cd9b089bd1a66bb8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-28 19:17:28.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e8a431caa404567334e147d0bb86307a367d849d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:17:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a624708589c4ca01d2dc3eabcf85fe942fe9b7dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:22:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0497a2922cd105df366e6991b758b7517e605acd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:22:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e508c645ae410686e7b7bbe2eecb15 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6d3b3bde2a44d4294f4e4bd97bfe305 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd22d599d37c42a5abd758adcab73d4e : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f83c872d1314384a7e2f300edbcd211 : SUCCESS in 7m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b7769259f91470ba8117b756d8f614d : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9790808dde6a4872b8bc1148216323de : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03b7633a988e4143956a35fb32f21312 : SUCCESS in 39m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d2680412f00443db9f53ed6785af3ec1 : SUCCESS in 1h 16m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/310d95f03cbf46d2a5d7876b058c75b7 : SUCCESS in 2h 02m 51s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d5b5687d36184296b02cb60f4bbfcfac : SUCCESS in 28m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c0af597563324ce78f6ea04c4fcae8f8 : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":3},{"id":"24359f92541e07c9a485d58ae2d3821a12fcf788","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:23:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f02fa6a42014adbb3e9b986b8bb9856 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3c9daababe8b8145d948ba066c6391648fdf2e31","revisions":{"5a20289adec969d0132ab4157a373adc7bfe7b12":{"kind":"REWORK","_number":1,"created":"2020-02-25 16:51:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/709786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/709786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/709786/1"}}},"commit":{"parents":[{"commit":"a9fe6bf36b97ac69aac6ef992cfc154a89517ee1","subject":"CentOS 8: Disable ntpd, enable chrony container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9fe6bf36b97ac69aac6ef992cfc154a89517ee1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:19:23.000000000","tz":0},"subject":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0","message":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0\n\nThis release provides support for CentOS 8.\n\nChange-Id: Ie597e1bee46e1f7b7a7129d37e70d88c08cade74\nStory: 2006574\nTask: 38869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a20289adec969d0132ab4157a373adc7bfe7b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a20289adec969d0132ab4157a373adc7bfe7b12"}]},"branch":"refs/heads/master"},"52dfafa244501a1a098da97ba284941ec50a1edf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-27 11:24:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/709786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/709786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/709786/2"}}},"commit":{"parents":[{"commit":"d301d0325833e56d2965d6be9936b6fd197a1520","subject":"CentOS 8: Disable ntpd, enable chrony container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d301d0325833e56d2965d6be9936b6fd197a1520"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 11:18:12.000000000","tz":0},"subject":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0","message":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0\n\nThis release provides support for CentOS 8.\n\nChange-Id: Ie597e1bee46e1f7b7a7129d37e70d88c08cade74\nStory: 2006574\nTask: 38869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52dfafa244501a1a098da97ba284941ec50a1edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52dfafa244501a1a098da97ba284941ec50a1edf"}]},"branch":"refs/heads/master"},"3c9daababe8b8145d948ba066c6391648fdf2e31":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-27 15:39:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/709786/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/709786/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/709786/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/709786/3"}}},"commit":{"parents":[{"commit":"15109ccb547e603c0aca2a73e43f4db523fa1bcd","subject":"Make Kayobe code compatible with Python 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15109ccb547e603c0aca2a73e43f4db523fa1bcd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:17:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 15:39:27.000000000","tz":0},"subject":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0","message":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0\n\nThis release provides support for CentOS 8.\n\nChange-Id: Ie597e1bee46e1f7b7a7129d37e70d88c08cade74\nStory: 2006574\nTask: 38869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c9daababe8b8145d948ba066c6391648fdf2e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c9daababe8b8145d948ba066c6391648fdf2e31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
