)]}'
{"id":"openstack%2Fkayobe~710277","triplet_id":"openstack%2Fkayobe~master~I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94","project":"openstack/kayobe","branch":"master","topic":"dnf","hashtags":[],"change_id":"I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94","subject":"CI: Add overcloud host configure job","status":"MERGED","created":"2020-02-27 14:26:04.000000000","updated":"2020-04-24 00:07:56.000000000","submitted":"2020-04-24 00:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710277-1587686604165-16b58591","meta_rev_id":"4904673ee59b0a5f10fde97b01ec0bf3b84d8fba","_number":710277,"virtual_id_number":710277,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-04-23 19:57:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 00:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-04-23 12:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-16 15:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-04-23 12:26:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-04-24 00:07:56.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-16 15:44:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 14:04:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-23 12:26:23.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-24 00:03:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a1f81310cf93ed3490a99b413f1c7d0ccc03dbd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 14:26:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8debe85106edf8e1650fc291f8e8fd930cc1d49d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 14:58:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/a265de9f472049608b6f1581c9854d89 : SUCCESS in 14m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5f8e7cba06094d02bda38ce47a0162b1 : SUCCESS in 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"68a7ba86aa4eb37d9ed3073fb36f73a3b146e45a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 15:39:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd3ab2a162c525278e5720407baa7ae8cf922e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:51:07.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"a50a007957417b2303f0bc659d6dce79167eb759","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 15:56:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"06f236fb8092cd5cb120413f3d0bcd5cd1300ff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:05:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/f5d10534965a47e499e54ed602b1f547 : FAILURE in 5m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cde823384cbb45d790318f74d6fced78 : FAILURE in 9m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b4459be5179c400aac2cb2619a6133f035901504","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 17:16:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab0e0dbe182fd6e60c61abe74330b01e33ea31ad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 17:23:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"de8c56e8356855f6eddc3266f79e36f14938f7de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:58:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/f44fd16babc64fce98bc2546af257336 : FAILURE in 18m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2c4950b5bc6a4047b51b8f7c8ac094c8 : FAILURE in 10m 47s","accounts_in_message":[],"_revision_number":4},{"id":"71565b587d45984c13b5637bb93a94e08993d265","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 18:07:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0eeee7efff6f7de34e3bf231c0ecaecb1eebf3ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 20:46:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/45de246e335b4db9ac515afb3fc98991 : FAILURE in 15m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dbd07d8463cd4c4495866deb4c2d9006 : FAILURE in 10m 33s","accounts_in_message":[],"_revision_number":5},{"id":"80670826b21c058bad5ffefec8439e06dd311445","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 10:16:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4b120438b61b95b6a02142f99e086099ebb02a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 10:40:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/806527e642d9490dbf1598db868b57f3 : FAILURE in 17m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/54b152554dce41b883ad54dbe6c5a5e8 : FAILURE in 9m 33s","accounts_in_message":[],"_revision_number":6},{"id":"7d9aa856522bb74c25073ca3f379f18061ec9e5b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 11:00:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4008c84c03eb2f827292edc0c65204af8eb99d28","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 11:08:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2b231e1447acabc1229bcdee1268fbd3afcc7a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:34:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/a3446a28eaa84a75b6368e0974b4df61 : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/13028b59ec59451cabd48e66aabb74e6 : FAILURE in 9m 18s","accounts_in_message":[],"_revision_number":8},{"id":"1c20633994701c2b7b7a3b980999f0c295f48d38","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 14:31:59.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"60d6128a5fb23ece26a2b3769e4af751ef0ab909","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 14:49:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cf2483b8b3eea489484361b8f7d9b427051a975a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 15:40:05.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/a16e5ea4a1f742e4adfa67309c4b903b : POST_FAILURE in 43m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4711364c46124df988dfb4e33353f1c0 : FAILURE in 18m 41s","accounts_in_message":[],"_revision_number":10},{"id":"f07b0ceb557550b6111cef70e8d5f343ec7866e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 15:48:46.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9c443e51b3d18d519b3e9291b2b7acc7d2f606b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 16:17:27.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/419ed295577b45358d45a9a3a49ee6a0 : FAILURE in 21m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0d7319ac1c204ac88303864335a6f29b : FAILURE in 23m 05s","accounts_in_message":[],"_revision_number":11},{"id":"f72b97bbe4c4920366658f9a85e149f2943f21e4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 16:26:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f34650d7a881d971bb67807b9acdf960a5feefcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 18:07:23.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/b129d27d1fac4a179b56857846f4929c : FAILURE in 18m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/966fdd8f139f4aaab994346002578491 : MERGER_FAILURE in 9s","accounts_in_message":[],"_revision_number":12},{"id":"e71c0e5ce7863a76a7fd2602304142f3b38bbecb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-02 12:07:01.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"bd6fb4f927f6fb87cbefe93a506b4bba76d80f5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:24:43.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/c2b1134d4f664ad882fb8c8fe4a84505 : SUCCESS in 16m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5d12056bad5544df89b716d12a2cc8b0 : FAILURE in 15m 39s","accounts_in_message":[],"_revision_number":13},{"id":"7db3b3b26162b0bf666283fafe361a5a6c85dd3c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-02 12:43:17.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"36390bad6b9b41e1b6da651215753c7cf42fc571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 13:07:50.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/6a91e66758ad48b89a66e277662760b3 : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a4eebb1e304e48efa6eddbb2464fff63 : FAILURE in 18m 03s","accounts_in_message":[],"_revision_number":14},{"id":"ba98f95123abc312946dd69853292164b77be491","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 12:15:35.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"04db534a1255dd8428d11047284f4217846c1959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 13:31:24.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/a6695db3de6341b2b98e016f3519c7ab : RETRY_LIMIT in 8m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cdcd63e15dc94fcc8a7c344199b84629 : RETRY_LIMIT in 9m 31s","accounts_in_message":[],"_revision_number":15},{"id":"94f5684b5415b0de49b7ec3f0cc3f4dc2a633851","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 14:07:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3522b20c66935a1a302018c59f2e2a335c4f602d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 15:19:03.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/64606c32756d48c9961af96313109b96 : RETRY_LIMIT in 7m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d036ee2286544b85911caef6357cbbd4 : RETRY_LIMIT in 5m 57s","accounts_in_message":[],"_revision_number":16},{"id":"c06f2ccd876d3819bd714c6a47022c0eec78a8dd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 16:00:13.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"5c00cc7b8b8361d555e88d755142fb18077cf095","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:33:58.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"b1f69921899ecef18f9843854bd458947aebdc71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 20:18:32.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/ee9cb564f9724fa6bfbdc858c334ab63 : RETRY_LIMIT in 7m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a37e2591f40e47d5942f2a89df014385 : RETRY_LIMIT in 7m 43s","accounts_in_message":[],"_revision_number":18},{"id":"7483cbad5e7998cabcb31acfcff2e4a1d1fc3016","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 10:13:26.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"b6775659e1904d60b01de5feceecdee7578da05b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:11:43.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/f474f3f3e4dc4321946815d2a965d12d : RETRY_LIMIT in 6m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5b22006c9e4f4970a88125c76a30bb16 : RETRY_LIMIT in 5m 53s","accounts_in_message":[],"_revision_number":19},{"id":"d5b228f023cd42c2c0f92a842a297361b7bdc843","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 11:58:02.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"15024fbcd05eb68d42ab9861ac3b6ad871b24a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:55:21.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/2f4ada9c82d04828917139bfab185758 : RETRY_LIMIT in 2m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/16c2bec60c9541c8b0ce129b3e8137b2 : RETRY_LIMIT in 4m 54s","accounts_in_message":[],"_revision_number":20},{"id":"a0c7efc29a3cb43109542d97738e899c4cb49dfc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 13:24:11.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"2bebe03d848627ec0f102e8c5afa81293202f4ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 13:43:48.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/ac20960aa30a47a0871f2d321b3a6e56 : FAILURE in 16m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/be9949323e3e4eb9b81a00c06d209475 : FAILURE in 8m 02s","accounts_in_message":[],"_revision_number":21},{"id":"0eb3cf3edcb6d664485bb2a5b31560973d3c5495","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 13:48:18.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"610420f3ad090cd0b589951b5753f1de6bf4c553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 14:30:14.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/e21c00a6e8dc4d93b26a0db4664d552f : RETRY_LIMIT in 4m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fdf99d5ace064a679cf068db11bc35e9 : RETRY_LIMIT in 8m 41s","accounts_in_message":[],"_revision_number":22},{"id":"a0334ec32f56421b5641d60da08427c18dc5367c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 14:50:31.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"2de27e73b219fa824462842c42f77955b2983ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 15:25:55.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/203e66c8d96245d5b3c9528f53351fb1 : FAILURE in 17m 15s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c3e6b40e45984a57ac06fdef9912a403 : FAILURE in 9m 31s","accounts_in_message":[],"_revision_number":23},{"id":"23af9b6a022ad267e848b71b48452fdffc419515","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 16:31:32.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"86d9597a647a3d3194e30712a091c465ba4b1944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 16:54:34.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/f679f384124b4f7e999f8270c9cb7058 : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/32b8fce8842348248cfe316fa82d9f30 : FAILURE in 9m 00s","accounts_in_message":[],"_revision_number":24},{"id":"f77217b4f2a7728ca4469395b66129f9a1efda12","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 18:03:39.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"b33d02039118789b09051cac47f60217ffe1f6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 18:51:09.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/5f218df3d48d4af9a941d107065a5631 : SUCCESS in 18m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e3fec28f976d4033afc5e8d1088eaf07 : SUCCESS in 15m 55s","accounts_in_message":[],"_revision_number":25},{"id":"50715cb18255f71ef5b79969e72a67ba3ef46f98","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 12:02:46.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"40f8b7efc2d1f8f14f58f18ec6c88a0de72a1580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 13:34:34.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0d2acbc1616486cbf820c7cd843208d : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3709596ab915471cb1e17a6e06db47ea : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e2ec4d89ac54a4cb147392cca0b67e7 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c6f816bb0d54c939352c6c023f6c6f6 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdc99eac52b049f58e88aa7e761b8ce0 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14680eb9946243fca07ebf832e574db8 : SUCCESS in 7m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6bde3e74b5c844c3b93057c42ebb89b1 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33672c31fb07404b9c862a2b0ba6245c : SUCCESS in 12m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71df090fa9484507b8959c401218211c : SUCCESS in 28m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7704e3a9a11142f2928a02f35097c856 : SUCCESS in 1h 15m 25s\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/295d66416e194dc0b808e06a27b85846 : SUCCESS in 22m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9743b22587d74d0281df4ba105845b7f : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/76541d21417d4fd885ed128479074d1b : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/eba642a8dda641c58837879ac7c7ee55 : FAILURE in 1h 02m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e9f9cc4640844905990a42e7cac97b21 : SUCCESS in 25m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/faa56e13fec440069a574cf3ab0378ef : FAILURE in 45m 12s","accounts_in_message":[],"_revision_number":26},{"id":"828a8bb681ab5a39b291f76793eeff08e2ebe873","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 09:36:40.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"4fb91cff1b9f747f494044ec7512be12211acdf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 11:40:54.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b5d06944ca34e9ebf49df5cb2e4ffb9 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d07d6d8ef44fb1a8d6801d8be2c76c : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2402871a8ca49feaac25709c38b5aee : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a5680de25ef4857b5a8e536a6532687 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a38170badb74ef098cccd1215e6b7c7 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e83dfd21f04203bfafc5b117c4943d : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2081c964a77142bf9b6b23c1581726c7 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50e06d7c50014092bd7ed2fd895f0697 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81f9b0620f4b42b8b9aa08b229b8fabd : SUCCESS in 34m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6667426c6bd84e7b94b1ca6d1178cc2c : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/48f14f0a8cc4468da6712f125f8911b5 : SUCCESS in 15m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f2e7a580a6ba4e6aadb437e16452b749 : SUCCESS in 1h 00m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/adb4b95965fb431bb5063a9873711222 : SUCCESS in 15m 17s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ec2f2f3872ba48d09c3047fb99160022 : SUCCESS in 1h 59m 17s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/831a0c35fbc9412183cb907a2abd850d : SUCCESS in 19m 45s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/1d7b75071fd245c5ba203a1e8fc7d845 : SUCCESS in 35m 43s","accounts_in_message":[],"_revision_number":26},{"id":"c52d07509fd6b5b5475bf7b05a60604628f2e10d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 16:44:15.000000000","message":"Patch Set 27: Patch Set 26 was rebased","accounts_in_message":[],"_revision_number":27},{"id":"d6c4819219b0ddc144da0c26fbffc2021d30fa0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 16:44:27.000000000","message":"Patch Set 27: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":27},{"id":"702b0bcf47edaffaaee504f1e835ed9c9791ca99","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 16:49:52.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"9415036c1acd2c1af710efbe2190f49a2a8ccba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 18:09:46.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/334d2d44635d4aff86d7ddc91fd7ddfc : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed41e4f25274b71ae89b3da187ced03 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f26599519165444b8ab6b39c003d581a : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03f7a28759ff493a878e88344535a79c : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a981b45cd624608bff386a64609d182 : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32da18505af14287863769d7694054d5 : SUCCESS in 10m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d927dbf937b446259276690bf86fed27 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/86fa55e04697412ab23317168ff98aac : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b12f9b1093844dbbbb96e746495b4821 : SUCCESS in 37m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c58ef6c6a15247c89b7faa077cfc2c9b : SUCCESS in 1h 08m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/25d3f5efd44b491fa14f81ba88d3b808 : SUCCESS in 16m 49s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e1124466b94c4057a9b70f2019b7b94b : SUCCESS in 19m 48s","accounts_in_message":[],"_revision_number":28},{"id":"31815127f846045b264b72d120de2703e3ac47af","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 10:24:38.000000000","message":"Patch Set 28: Review-Priority+1","accounts_in_message":[],"_revision_number":28},{"id":"2c0d325fa4497664ee0d0ec52f74b5e11f90921b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 11:23:53.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"76095eaa202eb3483761396446672124eb9c4a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:37:15.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc516514f9fe4cd1a302d2898fb47731 : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c54622387f4e009bc7735577e09a0a : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e013f5b33c34f9a93d3799efe7c3d92 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c4903db9f8ae4d89960fda148d38297c : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5667569a20f04fbd9bff6c582f3dbb64 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6831db1efe4222be67649061e032a0 : SUCCESS in 8m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8df5b7bc3284dd0bca07d2fc8ecc355 : SUCCESS in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a2c3898c330049d3b0dc121b29e3a251 : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f464fdbe71d0466195e568d13d19979c : SUCCESS in 36m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/25ebbfd1c9964ee5a88754eaf313dff6 : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ebf6aedb1fc842c8a391b0760b645dcd : SUCCESS in 16m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/00dea11e1bf64fc5b1f4380ad79821b9 : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":29},{"id":"95fd5f3c843255414682ede4487d20f6c9c9912d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-23 14:33:09.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"a2798d05c7a41db08326ea4cdbe9eb3440b61c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 16:36:14.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d256a226d33f4512b5b8988b788b494a : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/628f27cc6eb54eb5a918615619f67f3c : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db77c49927974b5f9eb6b4d65919f208 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4dce9b85104442ccbe0afe9bf096254e : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dd28317ae674094a04025c699a86d04 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590567cb928e4acc888becd850ecd55f : SUCCESS in 9m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e8ab2736d4e4831aa3bd52ee90fd554 : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bc23093f2fd4ed58d911dad5553de62 : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98a7840630364c018a45d49f27de86d2 : SUCCESS in 34m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8848aee9c57949ffaae4b657a5b327a5 : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/24eb1ebdfe2943f99e17b59f2178f8f3 : SUCCESS in 19m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8c85175bd784498e864bc6a5e983041a : SUCCESS in 1h 53m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/85b44ba023274724a4a485e0e4086f9b : SUCCESS in 18m 37s","accounts_in_message":[],"_revision_number":30},{"id":"77287ab2ed10a42681bbdca476cdbcaa7ff087ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 16:18:42.000000000","message":"Patch Set 30: Review-Priority+1","accounts_in_message":[],"_revision_number":30},{"id":"a24312d86464f4b90d03eac0624fd6ff27b47419","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-16 14:03:30.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"0c8bc738f487d4eb8843b2840289075040c5a704","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-16 14:03:50.000000000","message":"Patch Set 32: Patch Set 31 was rebased","accounts_in_message":[],"_revision_number":32},{"id":"101c7eb99da6972557524d50fa3d4e671ff462a7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-16 14:04:05.000000000","message":"Patch Set 32: Code-Review+2 Review-Priority+1\n\nExcellent! Just fixes a small typo.","accounts_in_message":[],"_revision_number":32},{"id":"77eb3ee2f8a9c02b9174053e55b7ad8f0a4df638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:33:10.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3322c856d174e6db13607807cf36f97 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9aeb64d71a841deb3dc728653477172 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a9782e099174319a9f07d915bbf2191 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f18607ba59f94ab1826b7ade73fc3ea6 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1c6f478102b412a8a1788e3ed9d84be : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dad4157937d4d31a88ec3e94f9bd702 : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90556bde40734012a961b2d67d0644d8 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6176b7f1587240cd9cd45d270f7d1a11 : SUCCESS in 13m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4d46b3934534f50940d3468b8dcb286 : SUCCESS in 38m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d8c5b7e729624869a8af104d09291d9f : FAILURE in 42m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ca48916355a5444e9150b6c738f67ff9 : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a63a071db4654e53af06f222f4c7c366 : FAILURE in 1h 21m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7dc470ab42744c13b6c6e7cb74d4338d : SUCCESS in 28m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0c867375a6cb4143bb276a5e592f5ae1 : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":32},{"id":"0c4a16dfcb8565d1b30e06f028cf1a7d0b7dffdb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-16 15:44:49.000000000","message":"Patch Set 33: Patch Set 32 was rebased","accounts_in_message":[],"_revision_number":33},{"id":"0638ce85302234f9230a060f7dd76c4f5e116c8b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-16 15:45:16.000000000","message":"Patch Set 33:\n\nRebased onto https://review.opendev.org/720521 which updates inventory.","accounts_in_message":[],"_revision_number":33},{"id":"443b9a6b32313fc12d067644cfb8e0c16cb570fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:44:48.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af56a8378a19421c85b4196954971e94 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cfea341aec34ecd925177097f663014 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/956275bc5b364aacb981e6264fcc9cd4 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f44ed84d8cc74fdf88ab319d728b58d4 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c8e076c9c554ed4b44ec02495f73df0 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f97c4e4229e4736906431ec3b168155 : SUCCESS in 7m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0ee6df0268c14b8a803e5db45139f119 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8769e4da4b3246388ca4bd6634599229 : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8f4afa2e48041c5bee58dc2e66491de : SUCCESS in 29m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/39a6d17ce8214b83a30b4dbf58a75402 : SUCCESS in 1h 09m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5aecab02d58e47e3ae5336723df77df0 : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9bb15bdfc5d44d9ebc73c6f487844d9f : SUCCESS in 1h 52m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fe80b70f6ecb49ca93d4b09f7f5a26bc : SUCCESS in 19m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/da7165384515431485022dff28c69d03 : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":33},{"id":"e5be7f5a8bf7619106bb1d62b013d9fd589484c8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-04-23 12:26:23.000000000","message":"Patch Set 33: Code-Review+2 Workflow+1\n\nNice work!","accounts_in_message":[],"_revision_number":33},{"id":"8f37633be2d24c608bf83ae77c022405cabdea32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 12:27:40.000000000","message":"Patch Set 33: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":33},{"id":"c6bc1aea496eae4c9b4b05ae6f632aa9ae3e4279","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:43:31.000000000","message":"Patch Set 33: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb1225faf4cc45c289ab898786f7a795 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb69eb43012f4f14bc2e114699a0d2f4 : SUCCESS in 7m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93bd160653434cbb8e84ab7b12499456 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd8ccd2d935c43a588f97b3a8077f055 : SUCCESS in 8m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17ca042575ef4b8eb8014514d465c703 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/979b0cf27c3f4db389e1b05bc0b532c0 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e2ee283d8d294fac9f31b0980fe14932 : SUCCESS in 37m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/27887dc3cdca49f8a14bd125616ef0d4 : FAILURE in 58m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3ea37321489043aba7c7b842d3efc656 : SUCCESS in 16m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7035e91e0de249c798bc861b15974a79 : FAILURE in 55m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d38214d60b3346e7a5e703ebea042a0d : SUCCESS in 18m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c0f20308b7d046bcafda157e61cc898e : SUCCESS in 28m 18s","accounts_in_message":[],"_revision_number":33},{"id":"a9a771058dbae6c0969b71e786424249d25b3590","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 19:57:56.000000000","message":"Patch Set 33:\n\nrecheck","accounts_in_message":[],"_revision_number":33},{"id":"809a26b0929c52b9ca86ea90b30e9b930e5e1432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:02:13.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58f867846c0c498ab9b1fc69eb84f6dd : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d1a380a6d44349b6b72d9e2f8ee2b9 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5218bd609cc4fe89f5973a3403afd25 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e0a18244d384c829a2d1f8f75dc37ea : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2b423c5712f47549d327efd6fbfa416 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb79d222d87d426089a789efe891a2a5 : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/362c224b03d74e7a8b2700eee5cc2e4d : SUCCESS in 7m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da9230f268b84cd991046446d069bdac : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7889c0417bec4100abd01d13efb3e0e5 : SUCCESS in 35m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/948392415ee04365be162ddb31edb7e3 : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d976f366dcb14a769bd5d20bafdef884 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d04c4492ce414c00bc4b7681e5a66c53 : SUCCESS in 1h 57m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c873d11659e746d4b06fb3ed536b034a : SUCCESS in 18m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/910f240640fe495d8cb082bf0665244f : SUCCESS in 28m 10s","accounts_in_message":[],"_revision_number":33},{"id":"50d7aa4fcdeed951a513b80e9803d4123b33e509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:02:35.000000000","message":"Patch Set 33: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":33},{"id":"8d1120e482b4c56bf68afd144b433265d6543421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:03:23.000000000","message":"Patch Set 33: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9590e1a294b4787917375e94d137894 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb3e3d28330c405aa4170e9904f4bba6 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b687da93e7a74dedab1f786a4d532204 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a52113082314f3abf336b35389fabe2 : SUCCESS in 7m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa97bab70bf0494aa9eb5ee105bdfad7 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/67391dd7c26f4a97b08b0f2ebb8ec1e6 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6892de8c4f14104b0557f9d6423f291 : SUCCESS in 33m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4761186ffa194b3a9870734703887041 : SUCCESS in 1h 07m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3574a8952f0443ae9a6a58b4b0f36a64 : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/68a6433162c04da1888eea9dbfca6c6a : SUCCESS in 1h 59m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/771fbf66c8474e92a7a51fdb225b8ce2 : SUCCESS in 16m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e5001d6d2d18404f92acda4119dadcfb : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":33},{"id":"46e8c4e279edd388f183d0d43fb561651e84ebdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:03:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":33},{"id":"3ba43d2070d8a200e42b4c1ffb200d6d49716e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:07:56.000000000","message":"Patch Set 33:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2fc2cbabe284a469c99bcae912d1474 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"92a437f63ce99241af5930493a90bf8f11049afb","revisions":{"0cbcaa44b7e6bcda1361249183db139af004959e":{"kind":"REWORK","_number":1,"created":"2020-02-27 14:26:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/1"}}},"commit":{"parents":[{"commit":"86d6557c974d950f4a62c2ef826fc76d9aaccc38","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/86d6557c974d950f4a62c2ef826fc76d9aaccc38"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cbcaa44b7e6bcda1361249183db139af004959e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cbcaa44b7e6bcda1361249183db139af004959e"}]},"branch":"refs/heads/master"},"04d1b28cb02a26b116279072a16ce12e198a13e8":{"kind":"REWORK","_number":2,"created":"2020-02-27 15:39:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/2"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 15:39:27.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04d1b28cb02a26b116279072a16ce12e198a13e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04d1b28cb02a26b116279072a16ce12e198a13e8"}]},"branch":"refs/heads/master"},"ab308aadd871fb1172ffc8735c82a9acf1c3d8a9":{"kind":"REWORK","_number":3,"created":"2020-02-27 17:16:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/3"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 17:16:39.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab308aadd871fb1172ffc8735c82a9acf1c3d8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab308aadd871fb1172ffc8735c82a9acf1c3d8a9"}]},"branch":"refs/heads/master"},"ef79debbec86557b5004b9c38c524f16e3106255":{"kind":"REWORK","_number":4,"created":"2020-02-27 17:23:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/4"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 17:22:56.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef79debbec86557b5004b9c38c524f16e3106255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef79debbec86557b5004b9c38c524f16e3106255"}]},"branch":"refs/heads/master"},"b33f37485ace13a26a4f4c3dc100b4cb6dad5dcf":{"kind":"REWORK","_number":5,"created":"2020-02-27 18:07:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/5"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 18:07:48.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b33f37485ace13a26a4f4c3dc100b4cb6dad5dcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b33f37485ace13a26a4f4c3dc100b4cb6dad5dcf"}]},"branch":"refs/heads/master"},"d42faf105d88e7a75c1f635b36ec2b5b30a848a0":{"kind":"REWORK","_number":6,"created":"2020-02-28 10:16:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/6"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 10:16:02.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d42faf105d88e7a75c1f635b36ec2b5b30a848a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d42faf105d88e7a75c1f635b36ec2b5b30a848a0"}]},"branch":"refs/heads/master"},"93445137be15d87e9727a3f17bdb5fb9339d8451":{"kind":"REWORK","_number":7,"created":"2020-02-28 11:00:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/7"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 10:59:50.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93445137be15d87e9727a3f17bdb5fb9339d8451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93445137be15d87e9727a3f17bdb5fb9339d8451"}]},"branch":"refs/heads/master"},"56b6e0f360394136d6a0cd0348bce2aa3b731cc8":{"kind":"REWORK","_number":8,"created":"2020-02-28 11:08:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/8"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 11:08:36.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56b6e0f360394136d6a0cd0348bce2aa3b731cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56b6e0f360394136d6a0cd0348bce2aa3b731cc8"}]},"branch":"refs/heads/master"},"5eac924d8d1df6ab61464ea45c258823e661619f":{"kind":"REWORK","_number":9,"created":"2020-02-28 14:31:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/9"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 14:31:50.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5eac924d8d1df6ab61464ea45c258823e661619f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5eac924d8d1df6ab61464ea45c258823e661619f"}]},"branch":"refs/heads/master"},"db09b83bd9ea2f02830a2c4fe944c634a5248de3":{"kind":"REWORK","_number":10,"created":"2020-02-28 14:49:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/10"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 14:48:52.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db09b83bd9ea2f02830a2c4fe944c634a5248de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db09b83bd9ea2f02830a2c4fe944c634a5248de3"}]},"branch":"refs/heads/master"},"91ed1058f978f0a56f85290df6f0c982cee93f7c":{"kind":"REWORK","_number":11,"created":"2020-02-28 15:48:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/11"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 15:48:38.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91ed1058f978f0a56f85290df6f0c982cee93f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91ed1058f978f0a56f85290df6f0c982cee93f7c"}]},"branch":"refs/heads/master"},"9914bcbd4ff2ab3f2c7afa4fbbd7d7d664ae824f":{"kind":"REWORK","_number":12,"created":"2020-02-28 16:26:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/12"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 16:26:30.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9914bcbd4ff2ab3f2c7afa4fbbd7d7d664ae824f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9914bcbd4ff2ab3f2c7afa4fbbd7d7d664ae824f"}]},"branch":"refs/heads/master"},"ff3840afb14df7dd5729cb47f7936164c74e9fb9":{"kind":"REWORK","_number":13,"created":"2020-03-02 12:07:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/13"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-02 12:06:50.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff3840afb14df7dd5729cb47f7936164c74e9fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff3840afb14df7dd5729cb47f7936164c74e9fb9"}]},"branch":"refs/heads/master"},"083fa784ed89c9d66246263d84d20003e0321847":{"kind":"REWORK","_number":14,"created":"2020-03-02 12:43:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/14"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-02 12:43:06.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/083fa784ed89c9d66246263d84d20003e0321847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/083fa784ed89c9d66246263d84d20003e0321847"}]},"branch":"refs/heads/master"},"68f4098d41c0f10750746d3a1e7fb4e4690e076b":{"kind":"REWORK","_number":15,"created":"2020-03-03 12:15:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/15"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 12:15:23.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68f4098d41c0f10750746d3a1e7fb4e4690e076b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68f4098d41c0f10750746d3a1e7fb4e4690e076b"}]},"branch":"refs/heads/master"},"e18fedee8ac8785f6e22df27cbf5a2903b0902e3":{"kind":"REWORK","_number":16,"created":"2020-03-03 14:07:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/16"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 14:07:04.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e18fedee8ac8785f6e22df27cbf5a2903b0902e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e18fedee8ac8785f6e22df27cbf5a2903b0902e3"}]},"branch":"refs/heads/master"},"5b4db686fdbddd4d233417e28af0d49d6fcf19d8":{"kind":"REWORK","_number":17,"created":"2020-03-03 16:00:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/17"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:44:34.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b4db686fdbddd4d233417e28af0d49d6fcf19d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b4db686fdbddd4d233417e28af0d49d6fcf19d8"}]},"branch":"refs/heads/master"},"0ec182d2440bb00c12c82a79c36a342d27fbcc38":{"kind":"REWORK","_number":18,"created":"2020-03-03 17:33:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/18"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 16:45:32.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ec182d2440bb00c12c82a79c36a342d27fbcc38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ec182d2440bb00c12c82a79c36a342d27fbcc38"}]},"branch":"refs/heads/master"},"9b0bc3bb97d773ebe4fcb4601e36fbc3cdfd711d":{"kind":"REWORK","_number":19,"created":"2020-03-04 10:13:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/19"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 10:13:13.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b0bc3bb97d773ebe4fcb4601e36fbc3cdfd711d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b0bc3bb97d773ebe4fcb4601e36fbc3cdfd711d"}]},"branch":"refs/heads/master"},"b7f2c16e51464d8167ec830365bd2854abd1ede6":{"kind":"REWORK","_number":20,"created":"2020-03-04 11:58:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/20"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 11:57:47.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7f2c16e51464d8167ec830365bd2854abd1ede6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7f2c16e51464d8167ec830365bd2854abd1ede6"}]},"branch":"refs/heads/master"},"c717cffbd0c9b94b43e172c0ed56efd8618e46d5":{"kind":"REWORK","_number":21,"created":"2020-03-04 13:24:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/21"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:23:55.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c717cffbd0c9b94b43e172c0ed56efd8618e46d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c717cffbd0c9b94b43e172c0ed56efd8618e46d5"}]},"branch":"refs/heads/master"},"e3339f344796547023314cedcd21fd324e8e6742":{"kind":"REWORK","_number":22,"created":"2020-03-04 13:48:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/22"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:47:56.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3339f344796547023314cedcd21fd324e8e6742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3339f344796547023314cedcd21fd324e8e6742"}]},"branch":"refs/heads/master"},"454a91f9eca2d389c8487e4952ad7f306f848558":{"kind":"REWORK","_number":23,"created":"2020-03-04 14:50:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/23"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 14:50:19.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/454a91f9eca2d389c8487e4952ad7f306f848558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/454a91f9eca2d389c8487e4952ad7f306f848558"}]},"branch":"refs/heads/master"},"bd8b35aee1dfac3aac5967bd76147429160da500":{"kind":"REWORK","_number":24,"created":"2020-03-04 16:31:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/24"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 16:30:49.000000000","tz":0},"subject":"WIP: CI: Add overcloud host configure job","message":"WIP: CI: Add overcloud host configure job\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd8b35aee1dfac3aac5967bd76147429160da500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd8b35aee1dfac3aac5967bd76147429160da500"}]},"branch":"refs/heads/master"},"badd945514da5b4b4b67ecefd438098baca381be":{"kind":"REWORK","_number":25,"created":"2020-03-04 18:03:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/25"}}},"commit":{"parents":[{"commit":"7fa3d5bb9556d8d945036d7401908038e20f2fa6","subject":"CentOS 8: Support DNF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fa3d5bb9556d8d945036d7401908038e20f2fa6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 18:02:45.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/badd945514da5b4b4b67ecefd438098baca381be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/badd945514da5b4b4b67ecefd438098baca381be"}]},"branch":"refs/heads/master"},"afe5e0087aa6dad5c129cf1a8389b29413cf05c3":{"kind":"REWORK","_number":26,"created":"2020-03-05 12:02:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/26"}}},"commit":{"parents":[{"commit":"5cff6d6c5a2d393b7d18ec7731cb94e10aac19a7","subject":"CentOS 8: Support DNF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cff6d6c5a2d393b7d18ec7731cb94e10aac19a7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 12:02:34.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afe5e0087aa6dad5c129cf1a8389b29413cf05c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afe5e0087aa6dad5c129cf1a8389b29413cf05c3"}]},"branch":"refs/heads/master"},"22f2eb96f725b5278634a84f3df89536d34efa9b":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2020-03-12 16:44:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/27"}}},"commit":{"parents":[{"commit":"ae48fbb0edbfd89f4f11dfd2590c16e8f9ede60d","subject":"CentOS 8: Support DNF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae48fbb0edbfd89f4f11dfd2590c16e8f9ede60d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 16:44:15.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22f2eb96f725b5278634a84f3df89536d34efa9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22f2eb96f725b5278634a84f3df89536d34efa9b"}]},"branch":"refs/heads/master"},"5f671498038933f16ba3adceecbbbe109493c19a":{"kind":"REWORK","_number":28,"created":"2020-03-12 16:49:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/28"}}},"commit":{"parents":[{"commit":"8adca0598ccbeaab5aa052f0727e2666a9c7c0c2","subject":"CentOS 8: Support DNF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8adca0598ccbeaab5aa052f0727e2666a9c7c0c2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 16:49:34.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f671498038933f16ba3adceecbbbe109493c19a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f671498038933f16ba3adceecbbbe109493c19a"}]},"branch":"refs/heads/master"},"c23f411777e19586e2de8f92e13fddcb2d248fa5":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2020-03-19 11:23:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/29"}}},"commit":{"parents":[{"commit":"dc32b52f08da551f23070d0915d78da0627aec3c","subject":"CentOS 8: Support DNF","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc32b52f08da551f23070d0915d78da0627aec3c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 11:23:45.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c23f411777e19586e2de8f92e13fddcb2d248fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c23f411777e19586e2de8f92e13fddcb2d248fa5"}]},"branch":"refs/heads/master"},"e6ae0f0ac24efbac2e9895951dccb1769ea67539":{"kind":"REWORK","_number":30,"created":"2020-03-23 14:33:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/30"}}},"commit":{"parents":[{"commit":"15e2dce049f7e76939d522ee9e7a44463fd0efca","subject":"Merge \"CentOS 8: Enable overcloud upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15e2dce049f7e76939d522ee9e7a44463fd0efca"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-23 14:32:49.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6ae0f0ac24efbac2e9895951dccb1769ea67539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6ae0f0ac24efbac2e9895951dccb1769ea67539"}]},"branch":"refs/heads/master"},"666958d6401f1ab0ab0313ea52e8f081a6673ced":{"kind":"REWORK","_number":31,"created":"2020-04-16 14:03:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/710277/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/31"}}},"commit":{"parents":[{"commit":"15e2dce049f7e76939d522ee9e7a44463fd0efca","subject":"Merge \"CentOS 8: Enable overcloud upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15e2dce049f7e76939d522ee9e7a44463fd0efca"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-16 14:03:26.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/666958d6401f1ab0ab0313ea52e8f081a6673ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/666958d6401f1ab0ab0313ea52e8f081a6673ced"}]},"branch":"refs/heads/master"},"2561b23e4f1098fde274135ccff4e7e1986b2ba2":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2020-04-16 14:03:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/710277/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/32"}}},"commit":{"parents":[{"commit":"c31870591deb6a46af211ca31825d7e24d7e9a1c","subject":"Merge \"Prevent openrc files from using wrong OS_CACERT value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c31870591deb6a46af211ca31825d7e24d7e9a1c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-16 14:03:50.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2561b23e4f1098fde274135ccff4e7e1986b2ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2561b23e4f1098fde274135ccff4e7e1986b2ba2"}]},"branch":"refs/heads/master"},"92a437f63ce99241af5930493a90bf8f11049afb":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2020-04-16 15:44:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/77/710277/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/710277/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/710277/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/710277/33"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 14:25:56.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-16 15:44:49.000000000","tz":0},"subject":"CI: Add overcloud host configure job","message":"CI: Add overcloud host configure job\n\nTests various non-default configuration:\n\n* Custom users\n* Network interfaces, VLANs, bridges, bonds\n* Software RAID\n* LVM \u0026 docker devicemapper\n* timezone\n* Package mirrors\n* yum-cron / DNF automatic\n\nThis improved test coverage allows us to be more confident about these\nfeatures working on CentOS 8.\n\nChange-Id: I36148e4356deb7d5ec00d8d3ebeb2d3932ff4f94\nStory: 2006574\nTask: 38938\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92a437f63ce99241af5930493a90bf8f11049afb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92a437f63ce99241af5930493a90bf8f11049afb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
