)]}'
{"id":"openstack%2Fkayobe~711044","triplet_id":"openstack%2Fkayobe~master~I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643","project":"openstack/kayobe","branch":"master","topic":"dnf","hashtags":[],"change_id":"I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643","subject":"CentOS 8: Support DNF","status":"MERGED","created":"2020-03-03 16:00:13.000000000","updated":"2020-03-20 16:38:12.000000000","submitted":"2020-03-19 18:34:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711044-1584642850574-c496cb8f","meta_rev_id":"c77a7bb5e7942173677f1d16d7b40d009bd3a1d0","_number":711044,"virtual_id_number":711044,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 18:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-19 11:45:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-19 17:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-19 17:14:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-19 13:14:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 11:45:15.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-19 17:14:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-19 18:34:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96b873153fdae588ee35c5ac153c41f25cb0c366","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 16:00:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"889c559fb02bde1312ddfdc113fd8ff9f405006e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 16:08:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4e7b814f25244da77c1f98b2e0486de62821400d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 16:15:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"28528c65f620088cc55717ccc47e5eb1d0e3d4b5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:33:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"de3685a15920beb95cd27cfb0561f396b805b31b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:46:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"43f0cccd5fce5178eb1464f2e4b56c3be97d8fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 20:34:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/9e9055c78d154c39aced7807d0b1583c : RETRY_LIMIT in 6m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c16d990dcbe041268b5b32fb852be84f : RETRY_LIMIT in 9m 22s","accounts_in_message":[],"_revision_number":5},{"id":"39bf892f1de484bffb122868ca902ebcfe8698cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 10:13:26.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"df700c45a154de9a5e36b371664c1172bbaf1be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:02:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/dc5174daf3ca48779c4c0e1c36f451cb : RETRY_LIMIT in 6m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8e431e7464ab466383aa13de03a90bf7 : RETRY_LIMIT in 6m 48s","accounts_in_message":[],"_revision_number":6},{"id":"aaad76e3a6eab45e7692d137e772a3dfd42ca037","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 11:58:02.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"6a2728841d109ec297e9ab9b3a673c824ecae2e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:40:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/49315ba3a29b49789261e430edca6b4e : RETRY_LIMIT in 5m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/74d4a6f27ae4442b81f0736f2d316c0f : RETRY_LIMIT in 5m 44s","accounts_in_message":[],"_revision_number":7},{"id":"3415c110e4d80d5647fa8acfaff8231ec724cb87","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 13:24:11.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"ea5c3c81f8f94821a441aa682764ec94e5bf02d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 13:48:18.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"82585d89297455e8c2a89950b69b2050f3df7470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 13:48:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/7acb05e0072742f98d8fd81b6ed37cc8 : FAILURE in 18m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a5ad23f466e64655bc0d274cc6e9f3a0 : FAILURE in 8m 34s","accounts_in_message":[],"_revision_number":8},{"id":"2f30ec493264940f5d709861bb011686dee3aa0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 14:41:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/b00147b81f474beb879b8bfe33b9ef7b : RETRY_LIMIT in 5m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2994551f6bfe4c9f907fc520a31391ba : RETRY_LIMIT in 8m 10s","accounts_in_message":[],"_revision_number":9},{"id":"5fca9f9aab6c5f4200f392690381093473c994b6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 14:50:31.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"98ee5c7d854ed1190e59bf15d4bc8d79272db81b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 15:26:53.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/416c9b2ccd5e4119be1bf60d499b8e55 : FAILURE in 19m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0c83a5cd631043219ea75e1c6fcb5fc6 : FAILURE in 9m 43s","accounts_in_message":[],"_revision_number":10},{"id":"82b426544aef3c7aa6062b91ca8c1225cd243d55","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 16:31:32.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c0b32e8f8bc076b7fe69dc3da507a0e9a2244810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 16:58:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-host-configure-centos https://zuul.opendev.org/t/openstack/build/71473aab8d8b463999ca919d9636036a : SUCCESS in 19m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d9cb8ec1b726448ca6b63ff4f8acb389 : SUCCESS in 16m 19s","accounts_in_message":[],"_revision_number":11},{"id":"3b2872b24d3519b4159d757d0e5e6b9aab060cfa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 18:03:39.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"83b85060d0442bff5eabff2d27780daf104f72d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 20:25:14.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b231319c59a44ff985b3349acace94d4 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8be6068ba2145f88833eb60a358465b : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9dabefa3a5de4c60b5b134f9577144da : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a074067912a740a9baa19c0023394e25 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56024eee314a48989f61249207d37ad4 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee7d6091e9c4454aa94c8ad6a2bb8d8a : FAILURE in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f43df39054354e42ab0a06d895fc3234 : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ebdde30a268427386071c8dbe827ab8 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ba6ce7289ad4a8595e56c246a0e7f3e : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6532d5cee8364d12a504f03ff8b1a6d5 : SUCCESS in 36m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/90aad85c4a13414c9aa10fc72bc6ad6b : SUCCESS in 1h 12m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/21f4ee9c6ec14f87aa8204015421f0c9 : FAILURE in 9m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2295ef9905c7425e8b739e4dad7e35e3 : SUCCESS in 2h 15m 09s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/febc2c9dbc4645dea83be698389b2772 : SUCCESS in 28m 33s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/221026cfe8244880adc02d26ffb53285 : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":12},{"id":"e7802e2c46f90d11d14c629cc61947577045f2d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 10:02:27.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"11d99a327b1ca48e0483232587a3a8d67ae95154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 11:28:21.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82221484be1d40bdae2c173819218d8b : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baed7dea158b4759ae440139772895f5 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb3e04f05fe944f28da44917987b325d : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e95d438d34b4c0c85f5d09931095c88 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e24bff538b548a4b975ab3291c9f826 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a0ebd5f314249c6b2e972968726fe50 : FAILURE in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08c1bce21347442aaf8dfe7d9283d959 : SUCCESS in 4m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85cee02ea5ed4253872c6f997a6de3a5 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b0ecc3fe5954c0ca34cf86f4a858eef : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32072cc280574e82a39f7d0c75b07faa : SUCCESS in 26m 53s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c13a559eb7944e3da8e37465332b17ed : SUCCESS in 1h 23m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/99132998a82f4125bf1221b8b888816a : FAILURE in 9m 09s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c24a9f58174048ecb9e37f78876d94af : FAILURE in 1h 06m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d9ac1a42b6ee4efa8ccc482b06127369 : SUCCESS in 21m 31s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/066fcc3b19784c73a446965bb0112b7f : FAILURE in 38m 24s","accounts_in_message":[],"_revision_number":12},{"id":"6a5ad5a297eb5c0aad8b44ee3b86358825872b04","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 12:02:46.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"96112493d9c6be4f47b48ccf65bfc345bc2322c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 14:06:57.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f08048b8b75b49c2a2f8d23b8da99678 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201c44cb36b84914907f12eb52e1e2e2 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/649b69e88fdb402885d01ecfb76a70ca : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be4096c31850445eaf2834d5377f232b : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb23f704bf284f4fb149d514bcde7825 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afee297878b043579a6937f707da26b8 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89c42ef6f54c4231941d270a7bc301ee : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e1e7d6eb2cf4d95b9184aedde7a96e5 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3dfc0b3ef99a436d968bc8d6c5eb84ca : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9fe78934d684838b8fd3ac1c4a80f54 : SUCCESS in 35m 36s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/246d1cb8fb9b49b08bf657c82109eefd : SUCCESS in 1h 16m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0ddd7955c6354f3eb3ac8ab04c87fecd : SUCCESS in 1h 33m 17s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/642e1273cc85410fac709592d7a2f1fb : FAILURE in 59m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e2f7d18de13d4c55a189180eec52e5a2 : SUCCESS in 23m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/32ef13429ddf41a3887fcab6614bcd5e : FAILURE in 38m 49s","accounts_in_message":[],"_revision_number":13},{"id":"c5d4e9e8008098ef81d7e86fbcfa891284eb9a74","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 09:36:27.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"ae89c2832edf805d89f81096670de1a27bb5a769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 11:41:32.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35702fff3cc94bab9e6979770f477eb4 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae0d3024088402f8fc367529e51ccf2 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a72786b9bf254653b40317b38208444c : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf8ce57cf5f8425380339345e3c7b6f3 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6992c0f2245144caacb1b397bb9dbb2c : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e011ab476ef6410fb987ec18072951d3 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22ace29d88ed4723a2aa6fd4459986bb : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f57e11c073d4d8c84addba7b405cce9 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73b61424926b4a79a156a041a02c30df : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6c050c0ea694d27975b415d2e17c785 : SUCCESS in 29m 07s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/032e317b8b684adc92b281b1fd9f99c4 : SUCCESS in 1h 10m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e9008fcfa9fd4f99a63b05ce591df22b : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e9ff0bb927d94ac0a1236e1fe3cf269f : SUCCESS in 2h 00m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2987a0ebc1714b3caf3ef8363d989a75 : SUCCESS in 23m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/be9f638688154a8a825e064959ffe001 : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":13},{"id":"553b42681ff706b9bbdb1c262033477495dfc915","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 12:28:24.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"150596a6775bb1ba80e373926884b47454620292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 14:39:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1a4904b906241e19c12648b3f8c5037 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ab7d29b9324113b3d5fa6c9a050007 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3c1e081012741109fb48bf30cdecf09 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf7aa478d7654e2d8369a4fc4e1baf2a : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90571c30645143e19e06c5345cab497a : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f500bc341c84424a15c429f296e8e24 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38a1bc3c842e404c9db9b62fcf51b45c : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e0f316fc119344138bfc7af25934c860 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a89753e883e417da81f55bfa13a47f8 : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90c44fd3d2a84cd28dbc2d83668818a8 : SUCCESS in 34m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/75f6cce6632e414ca35fb09dcf2678c8 : SUCCESS in 1h 27m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ce3c05cf5da24e0fb2479e5d26d407df : FAILURE in 1h 15m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f98e697f2e054c0aae6eb397f935b002 : SUCCESS in 2h 06m 31s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/75308b1adb3746368529d47ea19bb3e1 : SUCCESS in 24m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0103b1a126b64499af2f4518ce0c8196 : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":14},{"id":"3db27f3e4b8bcefb4565d875e590ded8209ac259","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 14:52:51.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"9342017c5219853393749d7edff8a918012558d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 17:28:33.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a6d095133f04ae78b8a5c26afc0c43b : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6f82b75924443c955e782806e1baef : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e8eabe69e36411984bbfacc9318bd2a : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec2c1242a35d4fcb91cb2e3b71bde0d0 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/746753fdeded4b6d858e839761f737d2 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45553f39432848879023409fd5fbfa0e : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f6b451030604d30be4d6936b824c9c0 : SUCCESS in 4m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46db6fa3536e468c9396690c9a0c59fc : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7770860d1b104ec3b0de9673444d4b3b : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f32fc1cda36f41298d3fa10d0c3fd48a : SUCCESS in 31m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b318f8bf09894eb480c49c1376df3fc7 : SUCCESS in 1h 18m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3e1f76edbaf14e92afecb12caf6b7e7d : SUCCESS in 1h 19m 06s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d1e65666711d475ea95040c784e7a478 : TIMED_OUT in 2h 33m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/25ad045c4f804553a58672f46a3ecad2 : SUCCESS in 28m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d946656eca4c4cff88605d18da254c1e : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":14},{"id":"b1fc324d1488bbf28f68e1487936bbf5ce4fa521","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 17:32:36.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"f75bc703b2ad8292d5f650b45ef4c9c63644aec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 19:43:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a452f1016a04b1fa1114a109924fb53 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535a963425bc40c79c0e745e7bec6003 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24b61710a29f41839a8124a61311855b : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e2845d28db54711bd1ede880d0db622 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13b078dd0498436baebd03d313771a40 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28147b49eccd41429adfed4b062cbf37 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00e7023f44be4fbca06724e327159ba4 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/69287ab961d0400783f564c115237c50 : SUCCESS in 8m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2df095cb23d4412fb579929a9218645b : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8a1fe9cbf2fa4cff91e6ae011bdb56b5 : SUCCESS in 31m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9d07cf73ec2e4ff4b58b925cf5b18f02 : SUCCESS in 1h 32m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/73dc4a610d864108854481f5860399e8 : SUCCESS in 1h 10m 12s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/713ea4d8330b42908f61b6e19b0ace0a : SUCCESS in 2h 05m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4dc0727753564e8aa6d5f13ed3716fa9 : SUCCESS in 21m 25s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9b97f777d6994720a6a19c2eebc5dd61 : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":14},{"id":"758009d52e88b8c2983face96580edf7cfe13fed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 16:49:52.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"3d0eaf4217b8417429b5a1d877537da05fc4a398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 18:09:41.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6305cc48b6e469da1f79ee683296258 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1295e456a7494496b4a718fbd11c0db6 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d06370f23914305890d68147345c604 : SUCCESS in 7m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9609e526741c4aaab8d1bff384fccd18 : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a24f7e6a4e6d446c9003f4081d428d9f : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c126606ce9844a4b75371f403e63bcb : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee04878f867948338e6c95681721f12c : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7067e38f62824647af505503937b2504 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/736cc5c8dc6b467aa2e26e6a4b147d4c : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/24728677f762454299b90536e06a355c : SUCCESS in 37m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/85a8a1d63fcc43798127702470816d1f : SUCCESS in 1h 09m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4be0b16559034c2fbc74b03b9d8b49b2 : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":15},{"id":"4b36495ab964621613d89f8fc621fe9972b4a427","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 10:24:53.000000000","message":"Patch Set 15: Review-Priority+1","accounts_in_message":[],"_revision_number":15},{"id":"45df4e51209de09659f3c177ef6fd180c2e787da","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-18 17:25:21.000000000","message":"Patch Set 15: Code-Review+2\n\n(3 comments)\n\nNice job.","accounts_in_message":[],"_revision_number":15},{"id":"f13ea123997255671743b9926ab2f7240a3bfc06","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 11:23:53.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"b63d4d66ad83bd97bd333c145f585388c71fcd52","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 11:24:17.000000000","message":"Patch Set 15:\n\nThanks for comments, updated.","accounts_in_message":[],"_revision_number":15},{"id":"c08bbfddd850eebbf90c92f0a339cdb76a6d7e6e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-19 11:45:15.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"8bf519a9a2713e5f664f5961ea66ad5d7abbdbaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:41:54.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7ccd6883b8c4d4eb899c3d555921af1 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3eed76f7a1425cb1f1838f8bc538a1 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4399aa7bc1248489d58357534b6b6dd : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e947351ce5cb4f328a18a12b01153b9e : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c815bd0f281c49359cd50680d653ce1f : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9da5aa54b634b279fa30a67d2041250 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0924945baa2450086bddac5d708ae52 : SUCCESS in 7m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54db09f9698142159ebfd9a00a362da2 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/118bff394ca54273b83fe8f4250db444 : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d526f03bb7b440a84a6a00c9b3b5082 : SUCCESS in 35m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/35bce86ca5e849f59fe526cb9effcdfe : SUCCESS in 1h 10m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d5025206b78f4296bf7a9e24d7a04f9f : SUCCESS in 17m 51s","accounts_in_message":[],"_revision_number":16},{"id":"3c2c0f3607418a82a9f72adfca63014aed78ceff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 13:14:45.000000000","message":"Patch Set 16: Review-Priority+1","accounts_in_message":[],"_revision_number":16},{"id":"5ce10ad0ab0d2d4af805807122c672bb7e1c8122","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-19 14:12:21.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)\n\nSorry for the long comment, but I feel this is important.","accounts_in_message":[],"_revision_number":16},{"id":"755e0887912d8d0324b72bbc4d0ba529e15b9fbe","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-19 14:39:38.000000000","message":"Patch Set 16:\n\nGood point about the missing metadata - I\u0027d forgotten about that. Since this commit is just providing equivalent functionality with DNF, I am happy if it is fixed under a different change.","accounts_in_message":[],"_revision_number":16},{"id":"c968cec4fd03ea2403ac5267ec078678c1ad861e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-19 17:14:15.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\nOK, let\u0027s fix it with the other reference in another change.","accounts_in_message":[],"_revision_number":16},{"id":"c02bf498a6214e89c85a75c7ea50f447ef95da04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 17:14:27.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"bc77f96d6328a2daab722ec7729f70427cb0e0e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:34:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"303f29e9dc85c89acd309a0c14cc0799546815be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:34:10.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03043aa101fa4d81b248f30a9c3fa65e : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8445780f928a44939f3ce4fdca4eb0fd : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/112a30b6820f4f6e8f8738e11a5fc490 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b970d9dba2044a9871315376b478bf3 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c850f04bf8648c18c716e7e4c714e32 : SUCCESS in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f28ecb67333437685839ab6d263a3b5 : SUCCESS in 7m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/580aa61c47b24b7c8a8c2a61bc7c47db : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/120795b7cc72418d9fdd8178181f3fd6 : SUCCESS in 37m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c67bf1e27fda4731b5e1e5b1682134e2 : SUCCESS in 1h 12m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c6f0233a63374ffb8f7e0c87d36d9e5c : SUCCESS in 23m 47s","accounts_in_message":[],"_revision_number":16},{"id":"5c0b23ebac1ea7cba61190dd4c77b0ffb592e0b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:36:46.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9acb0eb52fd745f285a36b86e011db28 : SUCCESS in 1m 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50186727317144f7a98d3b78c721345b : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":16},{"id":"849d7d16126ef3ae410b7c6547a692aca4e3fc73","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-20 16:38:12.000000000","message":"Patch Set 16: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 02c9835ceb7c45f7679df37e94bc4feaf2e37b36","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"dc32b52f08da551f23070d0915d78da0627aec3c","revisions":{"bd1025e41c111b1c2ea69eb9d8035394d85d8009":{"kind":"REWORK","_number":1,"created":"2020-03-03 16:00:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/1"}}},"commit":{"parents":[{"commit":"5b4db686fdbddd4d233417e28af0d49d6fcf19d8","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b4db686fdbddd4d233417e28af0d49d6fcf19d8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 16:00:05.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd1025e41c111b1c2ea69eb9d8035394d85d8009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd1025e41c111b1c2ea69eb9d8035394d85d8009"}]},"branch":"refs/heads/master"},"464c3643de1ffd26079bdbc7d72ea1dc5c8b39da":{"kind":"REWORK","_number":2,"created":"2020-03-03 16:08:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/2"}}},"commit":{"parents":[{"commit":"5b4db686fdbddd4d233417e28af0d49d6fcf19d8","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b4db686fdbddd4d233417e28af0d49d6fcf19d8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 16:08:11.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/464c3643de1ffd26079bdbc7d72ea1dc5c8b39da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/464c3643de1ffd26079bdbc7d72ea1dc5c8b39da"}]},"branch":"refs/heads/master"},"15351d3a109ce215d3153af68ab22c50fc93068b":{"kind":"REWORK","_number":3,"created":"2020-03-03 16:15:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/3"}}},"commit":{"parents":[{"commit":"5b4db686fdbddd4d233417e28af0d49d6fcf19d8","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b4db686fdbddd4d233417e28af0d49d6fcf19d8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 16:15:27.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15351d3a109ce215d3153af68ab22c50fc93068b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15351d3a109ce215d3153af68ab22c50fc93068b"}]},"branch":"refs/heads/master"},"1282e35db1e70b507fe035931ad9575cdb7d365d":{"kind":"REWORK","_number":4,"created":"2020-03-03 17:33:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/4"}}},"commit":{"parents":[{"commit":"0ec182d2440bb00c12c82a79c36a342d27fbcc38","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ec182d2440bb00c12c82a79c36a342d27fbcc38"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 16:45:32.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1282e35db1e70b507fe035931ad9575cdb7d365d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1282e35db1e70b507fe035931ad9575cdb7d365d"}]},"branch":"refs/heads/master"},"2044f5144c1faa6f2742d6befbe9462c76f8c271":{"kind":"REWORK","_number":5,"created":"2020-03-03 17:46:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/5"}}},"commit":{"parents":[{"commit":"0ec182d2440bb00c12c82a79c36a342d27fbcc38","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ec182d2440bb00c12c82a79c36a342d27fbcc38"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 17:46:42.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2044f5144c1faa6f2742d6befbe9462c76f8c271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2044f5144c1faa6f2742d6befbe9462c76f8c271"}]},"branch":"refs/heads/master"},"97b7466a3016f399c4b9e28d640d0a0ec5cd19ae":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-03-04 10:13:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/6"}}},"commit":{"parents":[{"commit":"9b0bc3bb97d773ebe4fcb4601e36fbc3cdfd711d","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b0bc3bb97d773ebe4fcb4601e36fbc3cdfd711d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 10:13:17.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97b7466a3016f399c4b9e28d640d0a0ec5cd19ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97b7466a3016f399c4b9e28d640d0a0ec5cd19ae"}]},"branch":"refs/heads/master"},"f9ff92adbf773d7e292b3e918b10a95e003369ec":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-03-04 11:58:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/7"}}},"commit":{"parents":[{"commit":"b7f2c16e51464d8167ec830365bd2854abd1ede6","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7f2c16e51464d8167ec830365bd2854abd1ede6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 11:57:54.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9ff92adbf773d7e292b3e918b10a95e003369ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9ff92adbf773d7e292b3e918b10a95e003369ec"}]},"branch":"refs/heads/master"},"38af14573602c1c88c97faa28950f0145fe6ebde":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-03-04 13:24:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/8"}}},"commit":{"parents":[{"commit":"c717cffbd0c9b94b43e172c0ed56efd8618e46d5","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c717cffbd0c9b94b43e172c0ed56efd8618e46d5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:23:58.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38af14573602c1c88c97faa28950f0145fe6ebde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38af14573602c1c88c97faa28950f0145fe6ebde"}]},"branch":"refs/heads/master"},"5ad1740cef0dbebf575e9bbc5992622c83dd498d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-03-04 13:48:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/9"}}},"commit":{"parents":[{"commit":"e3339f344796547023314cedcd21fd324e8e6742","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3339f344796547023314cedcd21fd324e8e6742"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:47:59.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ad1740cef0dbebf575e9bbc5992622c83dd498d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ad1740cef0dbebf575e9bbc5992622c83dd498d"}]},"branch":"refs/heads/master"},"c0184f2784a25e25913296b75805723403ee5b68":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-03-04 14:50:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/10"}}},"commit":{"parents":[{"commit":"454a91f9eca2d389c8487e4952ad7f306f848558","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/454a91f9eca2d389c8487e4952ad7f306f848558"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 14:50:22.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0184f2784a25e25913296b75805723403ee5b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0184f2784a25e25913296b75805723403ee5b68"}]},"branch":"refs/heads/master"},"eba73c50d94d168168addab39290d5afcb399c09":{"kind":"REWORK","_number":11,"created":"2020-03-04 16:31:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/11"}}},"commit":{"parents":[{"commit":"bd8b35aee1dfac3aac5967bd76147429160da500","subject":"WIP: CI: Add overcloud host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd8b35aee1dfac3aac5967bd76147429160da500"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 16:31:22.000000000","tz":0},"subject":"WIP: CentOS 8: Support DNF","message":"WIP: CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eba73c50d94d168168addab39290d5afcb399c09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eba73c50d94d168168addab39290d5afcb399c09"}]},"branch":"refs/heads/master"},"7fa3d5bb9556d8d945036d7401908038e20f2fa6":{"kind":"REWORK","_number":12,"created":"2020-03-04 18:03:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/12"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 18:00:08.000000000","tz":0},"subject":"CentOS 8: Support DNF","message":"CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fa3d5bb9556d8d945036d7401908038e20f2fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fa3d5bb9556d8d945036d7401908038e20f2fa6"}]},"branch":"refs/heads/master"},"5cff6d6c5a2d393b7d18ec7731cb94e10aac19a7":{"kind":"REWORK","_number":13,"created":"2020-03-05 12:02:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/13"}}},"commit":{"parents":[{"commit":"4c390bb74355ca8d200b9854373d6ebafeece99b","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c390bb74355ca8d200b9854373d6ebafeece99b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 12:01:39.000000000","tz":0},"subject":"CentOS 8: Support DNF","message":"CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cff6d6c5a2d393b7d18ec7731cb94e10aac19a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cff6d6c5a2d393b7d18ec7731cb94e10aac19a7"}]},"branch":"refs/heads/master"},"ae48fbb0edbfd89f4f11dfd2590c16e8f9ede60d":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2020-03-06 12:28:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/14"}}},"commit":{"parents":[{"commit":"a8b6527e275dab7b6802e76f0ae44dda967d6613","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8b6527e275dab7b6802e76f0ae44dda967d6613"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 12:28:24.000000000","tz":0},"subject":"CentOS 8: Support DNF","message":"CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae48fbb0edbfd89f4f11dfd2590c16e8f9ede60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae48fbb0edbfd89f4f11dfd2590c16e8f9ede60d"}]},"branch":"refs/heads/master"},"8adca0598ccbeaab5aa052f0727e2666a9c7c0c2":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-03-12 16:49:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/15"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 16:47:26.000000000","tz":0},"subject":"CentOS 8: Support DNF","message":"CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-crom.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8adca0598ccbeaab5aa052f0727e2666a9c7c0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8adca0598ccbeaab5aa052f0727e2666a9c7c0c2"}]},"branch":"refs/heads/master"},"dc32b52f08da551f23070d0915d78da0627aec3c":{"kind":"REWORK","_number":16,"created":"2020-03-19 11:23:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/711044/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/711044/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/711044/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/711044/16"}}},"commit":{"parents":[{"commit":"a5f1f2bc49a23f188228039b096228d18b357f51","subject":"Merge \"Junos switch: update ncclient to 0.6.7+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5f1f2bc49a23f188228039b096228d18b357f51"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 15:07:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 11:23:39.000000000","tz":0},"subject":"CentOS 8: Support DNF","message":"CentOS 8: Support DNF\n\nAdds support for configuration of DNF repo mirrors for CentOS and EPEL\nrepositories, as well as custom repositories.\n\nAdds support for DNF automatic, which is a replacement for yum-cron.\n\nConfiguration is backwards compatible, falling back to the equivalent\nyum variables when DNF variables have not been overridden.\n\nChange-Id: I8bef5e9c8e1c77c25d6077ff690da8f2cde6a643\nStory: 2006574\nTask: 38922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc32b52f08da551f23070d0915d78da0627aec3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc32b52f08da551f23070d0915d78da0627aec3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
