)]}'
{"id":"openstack%2Fkayobe~711060","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ida5b60b723c8208bb7305c3d669eafdab6dbbe01","project":"openstack/kayobe","branch":"stable/train","topic":"story/2004959","hashtags":[],"change_id":"Ida5b60b723c8208bb7305c3d669eafdab6dbbe01","subject":"Make local kolla-ansible Python executable configurable","status":"MERGED","created":"2020-03-03 17:19:20.000000000","updated":"2020-03-13 01:30:25.000000000","submitted":"2020-03-13 01:29:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711060-1584062941675-99049c42","meta_rev_id":"972a7db39861bbdc5308e788635dbef201cea68e","_number":711060,"virtual_id_number":711060,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 01:29:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-12 12:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-12 20:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-12 12:28:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 17:25:24.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-13 01:29:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"697da6689933bdaef71d06ae7c7875ac5078e8f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00ed4436dbdacf5265cfd71dbfbce24a55609d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 19:52:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/501453949b154e559c6440925d6395f3 : SUCCESS in 9m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88e9c3c199974c468684b830b5a80499 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f48976b0ca724791a05a7782a44e8b66 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93f5a572dae24ab88035950e3392c224 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/525583d299064ba491529a7e518a810d : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b05cc0d6dc47e492695207958ac388 : SUCCESS in 9m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8015fcff12124ad682e937a4651c0ff9 : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33d8793cb956453c8d6756a271de4aac : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b35654f934a8482fa21db480b8a5d1aa : SUCCESS in 32m 33s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/27ba443a3dc94f608eaba82288eb1eda : SUCCESS in 1h 17m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/65bc55e2dc2e42a29f055d5221746ff3 : SUCCESS in 2h 06m 24s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9078ba147174441f9de870d48367ec44 : SUCCESS in 26m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/61d02573132649fcaac231a1a4755de5 : SUCCESS in 41m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ef30a66eb15ac6f962fa056f288d35a6347086e4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 10:35:35.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"58dbcd637aeb07c22b2d736742777da1d6ec5540","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 10:35:47.000000000","message":"Topic set to story/2004959","accounts_in_message":[],"_revision_number":1},{"id":"4b50798b23ceea5aa4a63ab41ce5f4e67f551dc8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-06 17:25:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"623b195d50cf5c341b2b0254e1a472c2d5f23548","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 10:23:42.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a7cc3b726b8eeeead254bd7b0ccef7e66037827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 10:23:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3be8de771b26f66afbb6a14969c22689d9c5c1e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 11:29:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/28bc939f6eee41ab8c0c8daf114b34f2 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b9a9fb3c9374f3782e00f07e1a7ef22 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac972688029744b794aaf930231158fe : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e32805c24d640729739c367708b0bf1 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48832df1249a44d4b6ba76a6d8d5182e : SUCCESS in 5m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/583cd6c0feba41988f60212dcb3f0c6f : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5542c925c3aa49e2bde31eb61349e5ba : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5cc3d4fa73fc4e2a9a7b3a030b480e30 : SUCCESS in 35m 37s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/fdb266201421498d927be431036640bd : FAILURE in 11m 30s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/cd79bebecf6f41659417fb19320382f4 : FAILURE in 9m 25s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/84a063f4936944b4806110b7c0be9fa4 : FAILURE in 6m 08s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/79688d5ce0e84135bc9c796fc4fb167a : FAILURE in 9m 55s","accounts_in_message":[],"_revision_number":1},{"id":"673508a9b3499bc97efe9570afb7fd37cdc09afa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 12:28:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"13eef56e418ea83f2d32e79d117cd9a570b66780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 13:00:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ed0a737d25b4aca86594decad3e2a8b : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dfcdab877079428c970cb8875c10cf30 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab32715cbf5c4e71a86a20422ba29fa3 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1c701022d547f694548d481e7434ab : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc8c651c8cc0417d83b4dadd67cf1b05 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9c00ec158f4606b00b4ce8f5c3d4de : SUCCESS in 8m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ddb751f2c8604159b87a3dab21a9c4ce : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f741b222f4a477e8a47742abb4486dd : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7b66772eac04a939fd652d451e3abc2 : SUCCESS in 28m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7b9db11f21e94937a660a398bbd3cdbe : FAILURE in 8m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/243adeb4ec4e4d8f988d89d8c7336328 : FAILURE in 7m 37s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/391fe1a558bd40b9bea8330cf47657c7 : FAILURE in 6m 43s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/435db45f061a4c71869223bafba74f66 : FAILURE in 9m 25s","accounts_in_message":[],"_revision_number":2},{"id":"19c21453e1883e78bff8e7cba73621d200739ea0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 15:33:38.000000000","message":"Patch Set 2:\n\nrecheck\n\nShould be fixed by https://review.opendev.org/712107.","accounts_in_message":[],"_revision_number":2},{"id":"a3970c3d614ca92ddc18c5c3bf17726b14f17ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 17:47:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/252a9d6d67d94b3abbed085d3f7706c1 : SUCCESS in 6m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f8ebcc01b4cc49ec91fdffb50921c69f : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea2bfb0daab34bd9b24a11738b5d0c99 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc3ff46d64f41a888488be7d97abd57 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3c6de5fbf9143aa8405629e717cd532 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf7f2ef8d884e438ba43a3a8637eab8 : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41afa8b5227d4619b58992e2fa824ef4 : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e9a085440b24be6b725c0d86d979069 : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e162732a1ec54258b9b3cc4b883f5cdc : SUCCESS in 29m 17s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7232c2d824614b17a418179a1bfd3c95 : SUCCESS in 1h 19m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/06ea33f62037487aad18710dfd219331 : SUCCESS in 1h 59m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/20b7780c149944f59b522aa20c296626 : SUCCESS in 25m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/67009285e77744cfa1baa069604c9e1e : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":2},{"id":"34da649915e7042213909b40b77b14316a3db4f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 20:02:14.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"15ca83541b74fc5048239a2793383179d10df142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 20:02:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16ba0ecd8a1fca88303cf1f1dd6902e590f2697a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 21:48:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8f18361058db429cbb07ffddc5875042 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5f5387a6aed4a98877f68201cfdbad3 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ff7e29e91b4d19bee567aa71134e7a : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f337c23941c143bd8ccfcad4c11870c9 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1c099b46fa4533bee10a65c078f0d3 : SUCCESS in 9m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ade903c63bc45a7bb8e0fb90293f45e : POST_FAILURE in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66b8f49dbcc34025a2093e34889b983e : SUCCESS in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/68841668c5484b53b9a225311c883feb : SUCCESS in 28m 40s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b246018e300a4edfa019fb7ec8d53363 : SUCCESS in 1h 07m 39s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/83aced38b7b74aa7b3794bbae04261db : SUCCESS in 1h 45m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8691b76193454f37bc1de30036b996fa : SUCCESS in 20m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d4f6f2e27f0642f287c439d0da892b85 : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3afdb35200c0354d5fd5285d963bcd3cd87c69e5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 21:53:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d8da9348165397318bbf25fd363ad2d062933585","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 23:32:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22cb101344a542a18390680303a6f6db : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e1d4c615af27449e93abe890cacdf5af : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60b71df36e7e48c1beba23ff74e94a91 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3644b6c5b8324f18a635367b268bfee6 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/842b7214f89b464287144b2214dc7ba2 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb526b4d2cc247b498fe7f79688d5c50 : SUCCESS in 6m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b238058b5da4065a6e75d937bfe0900 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a8c1ccca687842d18b121433a8671012 : SUCCESS in 7m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5cadcf680fd04c678dde9bfab2feefc7 : SUCCESS in 33m 40s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/23ee7bd74f2e477e89347acdcf7fdf6d : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/67dab60c58884e39be4159a9b07d7b74 : SUCCESS in 1h 37m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/fb88cf0ee6444da080fa9eea8c57e2f0 : SUCCESS in 20m 25s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4234f102d6b74d47b32cdaa80ef4d20f : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":2},{"id":"c02a17ec89511b5a9ac978b1b67bc873c5df9d3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 23:32:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2eb72bfe579e8fdae751698b3358b57f1f18cf04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 01:29:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e9c604b99bfebf4499c37ee399b5206a57a038f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 01:29:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d56d783b9e9f4878836e83e3af60ef8c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/785ea1e459c4414aa999bdd6d6551e30 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/669e3b3b3e944458a12ed78a011ea9d5 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a02b212b55e4caeb8e880038e772259 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1482f052e5cf42e59e9a9545449ee41c : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d13407eda244b0e801c4cfc39cb668d : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fdecb5e4917c4dac9be16b2d29a90913 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f21b26e4742d46239601baad14be4fed : SUCCESS in 34m 01s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0c9a7a10ce42467cb59d4ec437f1bb6f : SUCCESS in 1h 05m 14s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/beaa73e6a4da4c2a801dc05d472a5b95 : SUCCESS in 1h 52m 13s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/312fe712f321443387596d55b7c670e8 : SUCCESS in 23m 04s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/96e9785cb0d7427988cb7c582649c981 : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":2},{"id":"972a7db39861bbdc5308e788635dbef201cea68e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 01:30:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13979778615144c6a2bf0686d8cda87f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f6e5344863efa64ef171a3223e5a29e5eabc993","revisions":{"544a04cf79968ac83f83f3e5287685eb769fef50":{"kind":"REWORK","_number":1,"created":"2020-03-03 17:19:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/711060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/711060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/711060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/711060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/711060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/711060/1"}}},"commit":{"parents":[{"commit":"98d70aac676fe50585e9962fd2ebce1d0b2b4160","subject":"Merge \"Fix IPA builds without packages specified\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98d70aac676fe50585e9962fd2ebce1d0b2b4160"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 12:41:26.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 17:03:21.000000000","tz":0},"subject":"Make local kolla-ansible Python executable configurable","message":"Make local kolla-ansible Python executable configurable\n\nThe default is still Python 2. This is a necessary prerequisite for using\nthe master branch of kolla-ansible, which requires Python 3.\n\nChange-Id: Ida5b60b723c8208bb7305c3d669eafdab6dbbe01\nStory: 2004959\nTask: 38767\n(cherry picked from commit 6b47717c43999fe93d6a612db4fc34b391256468)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/544a04cf79968ac83f83f3e5287685eb769fef50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/544a04cf79968ac83f83f3e5287685eb769fef50"}]},"branch":"refs/heads/stable/train"},"2f6e5344863efa64ef171a3223e5a29e5eabc993":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-12 12:28:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/711060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/711060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/711060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/711060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/711060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/711060/2"}}},"commit":{"parents":[{"commit":"61690b7e750dfe139cea62cb3760c20b85569dac","subject":"Blacklist Ansible 2.8.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61690b7e750dfe139cea62cb3760c20b85569dac"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 12:41:26.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 11:52:32.000000000","tz":0},"subject":"Make local kolla-ansible Python executable configurable","message":"Make local kolla-ansible Python executable configurable\n\nThe default is still Python 2. This is a necessary prerequisite for using\nthe master branch of kolla-ansible, which requires Python 3.\n\nChange-Id: Ida5b60b723c8208bb7305c3d669eafdab6dbbe01\nStory: 2004959\nTask: 38767\n(cherry picked from commit 6b47717c43999fe93d6a612db4fc34b391256468)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f6e5344863efa64ef171a3223e5a29e5eabc993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f6e5344863efa64ef171a3223e5a29e5eabc993"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
