)]}'
{"id":"openstack%2Fkayobe~711062","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I4121d01dc83ac028350d4d98d3e1158e15fdfd63","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I4121d01dc83ac028350d4d98d3e1158e15fdfd63","subject":"Remove activate-virtualenv and deactivate-virtualenv roles","status":"MERGED","created":"2020-03-03 17:19:20.000000000","updated":"2020-03-13 01:51:04.000000000","submitted":"2020-03-13 01:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711062-1584064106853-46c24852","meta_rev_id":"b30fa57a96ba5285eb8d9385c4a6ea6a41685767","_number":711062,"virtual_id_number":711062,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 01:48:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-12 17:18:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-12 19:22:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-12 17:18:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 17:29:45.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-13 01:48:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68a078d2d42f9ae6e78926197add538f2fea2265","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16bf471b6deb5a571639069e787f8cc4eae4a4ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 20:17:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cfc05e1fb8f48108acd846a5ae17323 : SUCCESS in 6m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dc695ead02814d59a5cd411082246dcb : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53d19a5c6bf1498d9dbb156b42da1224 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a6e9bcb723948e985519cb1988885e5 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8aa8b5c63d6c4d09b714eabe57cfc13d : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c0d6bd1eaf4809982a8c84a979b7df : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/03e763ab112f485bb20d2b526d8c1d4b : SUCCESS in 8m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f26f81daa934ab79d33d7256f769453 : FAILURE in 7m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e958141199f74c2eb0ef97b4f26ea121 : SUCCESS in 40m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ed923e11dc3d4055b0feb47dbfc69bfa : SUCCESS in 1h 22m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4fd70e4ea14b4447bc88c426b3e2eb67 : SUCCESS in 2h 22m 28s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/19be336220b7431ab35f09395ef28e53 : SUCCESS in 30m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/17455089f33f44b5ba0e530156c010a2 : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":1},{"id":"850a60648838fc3ef25643eeeb68fea298f03731","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 10:35:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26079ae4c86a442026d56009fa9a9dac20dc4f1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:44:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/300f02ca6f4648118963ed89a40ce6c4 : SUCCESS in 5m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/64d01dd4d65b494d82b0af74a8080326 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9031c0311e694932a683ca3b29a65453 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74129605c4ca42bfbb51c4e96b405f21 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bdb1b6cfc83248c6bd687ce3f872df7c : FAILURE in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fcd42d740f41d6a17e169815beb165 : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e214c9ad77414874863e2017e4258b33 : FAILURE in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/99e64b590e6b45c7a1d7ce02283faba6 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a33173908416449db36f819d65fdae10 : SUCCESS in 36m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/990d2e1859ee47c4ae73021af4efad92 : SUCCESS in 1h 11m 01s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ca51f459e7ca476c96fa6e3ef4c7ad39 : SUCCESS in 2h 05m 58s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/45ae69b3071646b6bf45e4c31c5d8e97 : SUCCESS in 30m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4cfab4b660a74dd2b5f3f3ed19c15826 : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3b44a9affa0dd2c974de13e188481811a08eef02","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 13:36:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f27b7912ee5fd61501b21f212eab03651fe5952c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 15:38:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b384626d89504077920a648ac197c3eb : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b49949cdc4184315834dd5be653aca57 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e7cea2d795436f93ba92fc855328db : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9ce3e075fc14f41b9d3334e6e0e1751 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85baa6eceebc45bb835b395a31cd0d8d : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4513162b3a8c4d1685da5b5adf6ea940 : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a1bdbbfac0f40f1b28b965c6ec6ab5a : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20d3d66e22f34bde850b6dba039a9b1c : SUCCESS in 9m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f8b5b01b32b4c8c92d5ad300f3bfaa7 : SUCCESS in 28m 52s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/636d686862474b3b9c94d2240e639676 : SUCCESS in 1h 10m 37s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a17e2afe6f94484d84f649b765456c60 : SUCCESS in 1h 59m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/70a37e768ffa4ea5803702868927d915 : SUCCESS in 25m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/37a9461f57954d38842b828297fa7dc9 : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":1},{"id":"413763cbf7f2244c0ad6038b81d56c32fc540f03","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 09:33:24.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"8ed657ff00bb111374df828e5068d11132403125","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-06 17:29:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c436938c40536d5739c373d2822cad7d761959f5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 10:23:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17ee849f90ad7cab21eebc62c6e8cb6f28ec80f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 10:23:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9bae71d6dcb6cf7170ef110eb907b1111b7405d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 11:29:52.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"3229870bc75f6c06ffc8a36af06a1de0141b4feb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 17:18:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a591ecdf3e49d12f051a079c8483479fabc2a149","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 19:22:25.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"462f453cec6df0cb36597704445ba390bcc5f61f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:38:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fba8a847c4643819d135dd683241fe4 : SUCCESS in 5m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ca3ab601b05549b8b742c454d97cbd00 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7541d62884e349f19dafd513a48d3e60 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2be1a0dbfe94825b88f216e080dc32c : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f62c1e8e3157459a929f1d06230a6c6c : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f60f54ac8404a598040ce36de856b5b : SUCCESS in 7m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/449f3a72cf4a438998d36e3945c0945c : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5e1a409bfa149f3aba7ee13d3ad88e8 : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d8255c68e26d4332942f3834f4661d09 : SUCCESS in 38m 53s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4a9e355e16f04a44a9db08825cbf4738 : SUCCESS in 1h 24m 59s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f0d3da5c65114ebc914c257ddb622c57 : SUCCESS in 1h 57m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/707653691ec54d95a40162ec3942fe79 : SUCCESS in 21m 17s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b2c5aaeaedd24d79b2323f08e8e134bb : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":2},{"id":"05e0910bb48cc93888cfac6adb905367007cb275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 20:02:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"116137127d94d83f28be6cadff9b60891c9f880d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 21:48:28.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"d8412a3cb4a1fa9998723866a548678a1b11917f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 21:53:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fe940ad5c2f4c98f57f18d92739c3d997ef21950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 23:47:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2389269eb7234941abf85a8d23618ffe : SUCCESS in 5m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6243997d5891407ea395109ecf17751d : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abcfbad0cfa84c2c8d5b609dd72244a9 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa643a3f468427489b21b1703651dce : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc9da977969548199db3ad8a1ef67378 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62cb472bc7147ba85e9fcd9e8f63015 : SUCCESS in 8m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/efa949d0ffb04f4eadf63340abb6ddad : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8590c76c32fb412a91a7c0d7bfb2a901 : SUCCESS in 8m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3db9385c08dd4e04848bc93d584e0922 : SUCCESS in 34m 44s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5969cd44305d4533ac415b79533f8419 : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bd64feaadae342b6a8b9ccbde027d8a2 : SUCCESS in 1h 53m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/70e13f25d16247b687789c484699e357 : SUCCESS in 27m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/90f1fc2478d044159b4ce4aad04b2856 : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ddf75474fb4aa2f2740b9c200cabe5b1b1c60ac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 23:47:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c8be7a1018fbaaef437cc95faa082f474093f91b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 01:48:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e38028bd1ec2ed835869b41edbc7b06f6617801f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 01:48:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7edc3f98bb9c4e03be7675f7910149a6 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/564b9214ba8341bdb1d1cac3e4bf6d23 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/140233c936fe4fe0a7a47c6ce7f8acb4 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c2349a4ac7ee4dd9b0717c34ff3bd45b : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab579efbf4944fcb0176d92cdbda68d : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9fac3190e1b349bc97dfa9e73f4bfba7 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9851561fc2f343299b95a387ab9d3829 : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/536f5506727c4524b21adc632771738e : SUCCESS in 34m 58s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4d24a7d74f0240049896211a822953c2 : SUCCESS in 1h 11m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/eda11a6b9b0d4289b94321a95ca60c73 : SUCCESS in 1h 49m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b2a637a093cc4d1f8c241163f69e2e54 : SUCCESS in 25m 37s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4a7ac64133bb4da186ed750b41bc9def : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b30fa57a96ba5285eb8d9385c4a6ea6a41685767","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 01:51:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a20fd80def4e1d9e6c977b0a200897 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d3d9dbdb5e32e0a89e99768dd1cdec2dcd6d7b3c","revisions":{"82324a4875a86a9394c6cb5b15119a1e69f23a9a":{"kind":"REWORK","_number":1,"created":"2020-03-03 17:19:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/711062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/711062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/711062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/711062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/711062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/711062/1"}}},"commit":{"parents":[{"commit":"5f907b9d0a79869f60fa7ef6e1e3f69e7e3dc912","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f907b9d0a79869f60fa7ef6e1e3f69e7e3dc912"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-17 09:22:25.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 17:07:31.000000000","tz":0},"subject":"Remove activate-virtualenv and deactivate-virtualenv roles","message":"Remove activate-virtualenv and deactivate-virtualenv roles\n\nThese roles are no longer necessary now that Ansible supports setting\nansible_python_interpreter via a task- or role-scoped variable.\n\nChange-Id: I4121d01dc83ac028350d4d98d3e1158e15fdfd63\nStory: 2006574\nTask: 38824\n(cherry picked from commit 2e842ab3f5c7b737b547f4c2c718f96294f582d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82324a4875a86a9394c6cb5b15119a1e69f23a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82324a4875a86a9394c6cb5b15119a1e69f23a9a"}]},"branch":"refs/heads/stable/train"},"d3d9dbdb5e32e0a89e99768dd1cdec2dcd6d7b3c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-12 17:18:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/711062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/711062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/711062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/711062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/711062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/711062/2"}}},"commit":{"parents":[{"commit":"a14be33162c7b90fc023253f3283879179c89c08","subject":"Switch from shade to openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a14be33162c7b90fc023253f3283879179c89c08"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-17 09:22:25.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:35:49.000000000","tz":0},"subject":"Remove activate-virtualenv and deactivate-virtualenv roles","message":"Remove activate-virtualenv and deactivate-virtualenv roles\n\nThese roles are no longer necessary now that Ansible supports setting\nansible_python_interpreter via a task- or role-scoped variable.\n\nChange-Id: I4121d01dc83ac028350d4d98d3e1158e15fdfd63\nStory: 2006574\nTask: 38824\n(cherry picked from commit 2e842ab3f5c7b737b547f4c2c718f96294f582d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3d9dbdb5e32e0a89e99768dd1cdec2dcd6d7b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3d9dbdb5e32e0a89e99768dd1cdec2dcd6d7b3c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
