)]}'
{"id":"openstack%2Fkayobe~711063","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ic6a1c69a34241f4fbe617a0b12aec9b1528ba352","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Ic6a1c69a34241f4fbe617a0b12aec9b1528ba352","subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","status":"MERGED","created":"2020-03-03 17:19:20.000000000","updated":"2020-03-13 02:06:34.000000000","submitted":"2020-03-13 02:04:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711063-1584065095163-c6fdaf72","meta_rev_id":"379b5f4e7bfe0c7c79cd27bc16778f8f3c1f2e4d","_number":711063,"virtual_id_number":711063,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 02:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-12 17:18:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-12 19:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-12 17:18:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 17:30:16.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-13 02:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e14adb8da46ec856048dc84f01a871db64365f7b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81e89b43517e3cea676b8df78fab42f7fc374d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 19:52:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e01f1e00ae940cbaf8391327823d90d : SUCCESS in 6m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c95547a2ef3c4a17aa7e70788fbe4dd4 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44ed32d8649f42b18bc06f3db2dd9b47 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2399e41359c4f48b99838d9663b164f : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74478c36b2f64ade9f04481065727923 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1336a4a9cca547a8aab3795327e5af50 : SUCCESS in 8m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c899a739e7374036aa6e3b7f7d1034e0 : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c35c29804cb74fa399768337bd6d3876 : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e748fa1f498c4b34b165002c283c8522 : SUCCESS in 37m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/632f8745be264d608100f21eb21f38bb : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/eeb871c821244917b17db0513bc849f7 : SUCCESS in 1h 56m 42s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1db843285962424ca905605b9603c06d : SUCCESS in 29m 00s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/fceb6028c92748c996c247bb466bac0d : SUCCESS in 48m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2b166c2df9112045dc57a091739ffa8ec3a3eddb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 09:33:31.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"3f60254eeef939af2a53193f5c5f0b1abe96671c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-06 17:30:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d682152adcb1d95bc80bb9bb50eb93e2266c8edf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 10:23:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7bfa2aa80c60213ce101c28decdf1bd08a9ef3ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 10:24:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d88e0da4cbff12593337b723c8cbbd14ba85752","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 11:29:52.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"bc7add03c17f51dfb022b252c2517fb7297fa1c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 17:18:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"2b988a7e2fd655c4a818287d31f0e1a41a3ccc65","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 19:22:33.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7ede3415fc497cd28456d3533855cd75c93f43d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:28:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f206680296ff40d3b81a141bba8f1768 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/426968bd58694112bd709f9cf92cb328 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef10891137ba4aaabf76dd77329592ce : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20f8cf8b2284f24bd59cedff5995687 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/745b96d4797c40029901137e1eb18405 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e66bc37bed49b6a008689861fe5e4a : SUCCESS in 7m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/42995ee6f93a4036bff42a9e296f3a99 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/586468e421bd476ca3ec7ee0a2be9358 : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/759b5ce55d3d43bc9ce06efbb270b206 : SUCCESS in 34m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2e0dd81be2044a4e83d1d0237289cc2a : SUCCESS in 1h 11m 01s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7654fd816fa84536aceb5b65cf9d2765 : SUCCESS in 1h 48m 50s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/537b0d9f3ec149e5b8ce2d73676084c6 : SUCCESS in 21m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bd1b1d80dfe34599bd1d190ecf595929 : SUCCESS in 36m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e4ddf2658f0385d464ea991ebd3924a38ed3ec55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 20:02:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2fc45364ac95ec6e2033350f76f47076fc687de5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 21:48:28.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f1cec56f0c525a06fcf6346480e1fed82f245959","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 21:52:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1a9a9b57a26eb5304677bf418466a51f327111c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 00:06:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6175434ef2ce4e64967fa2245a244d61 : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ebd3f91a1816462b872a21d228844689 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc78a80d97674be4b13400230c023a12 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83bf8d60172b41c7b706763576faa59e : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ce0fe90d53b4806b7fd445577e2a3f5 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09839d03f144ae2824c85056d7b1809 : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d7bc80174f74809b8d96488e06d09a8 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0ed64ae3529c40a2b050c83c6075c231 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ec262c57ff242cf91c906fbb05b7aa9 : SUCCESS in 39m 47s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/35c33f57fa844350a1ba78f2fc86e55d : SUCCESS in 1h 22m 19s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2bca1abf699245d08d8245e1de14639c : SUCCESS in 2h 00m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6958d17c3650447bb4023da3e86e6c6c : SUCCESS in 22m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/dc7aefecbfb34db58b3e90b7ae520ee1 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b47c4c3d044405727c9d2d5cc8d691d2776e04cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 00:06:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4677b90ff30b8f38b7a6f4dbc557c6a9091bd4ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 02:04:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cf144595948f4516b1875b02e7e1fe0a : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39a0ecab8f334ed894419d19d39bc650 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8977e9d51041908c5069d0b580d387 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45818858362d4181a3b72b1fbfeac06f : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d9be44d5e724cacb9371cb638705b5f : SUCCESS in 6m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b092ceb4c644fe8b8ef6d8cc352b7bf : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b47b01988c74fc6b241b7dbb986023b : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad93986ceab946c396f3455a58c66da9 : SUCCESS in 35m 52s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c2cec51f6d2e43ee85402ddd52b576a2 : SUCCESS in 1h 19m 47s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d6bce40938ce49cda78f6cff2f320a50 : SUCCESS in 1h 54m 13s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e3eb8f5a4c0c4f20ba98d0e788fe4119 : SUCCESS in 21m 33s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/53cb20801c744a8996253755666f22c8 : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":2},{"id":"75a17a2d78d79250d79a87cb69006afbd8f34034","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 02:04:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"379b5f4e7bfe0c7c79cd27bc16778f8f3c1f2e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 02:06:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fbcf2440b2b4d1497c9d30971f8adf1 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a23b89e5ed66beca92784c156763348461e3bf8","revisions":{"8727416ede867e676042fd32226d4318c24d8589":{"kind":"REWORK","_number":1,"created":"2020-03-03 17:19:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/63/711063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/711063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/711063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/711063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/711063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/711063/1"}}},"commit":{"parents":[{"commit":"82324a4875a86a9394c6cb5b15119a1e69f23a9a","subject":"Remove activate-virtualenv and deactivate-virtualenv roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82324a4875a86a9394c6cb5b15119a1e69f23a9a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:05:06.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 17:07:47.000000000","tz":0},"subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","message":"CentOS 8: Use ansible_playbook_python for localhost dependencies\n\nCurrently we install python dependencies on the Ansible control host\neach time the ip-allocation and console-allocation roles are executed.\nThis is inefficient, particularly in the case of the ip-allocation role\nwhich is run serially for all hosts. It is also unnecessary since we\nhave these packages available in the Python environment used to execute\nkayobe.\n\nThe kolla-ansible role also has an implicit dependency on PyYAML for\nmanaging kolla passwords.\n\nThis change uses ansible_playbook_python as the Python interpreter for\nthe necessary tasks in these roles to avoid installing dependencies on\nthe system on CentOS 8 and Ubuntu. For CentOS 7 we still need to use the\nplatform Python, due to needing SELinux bindings.\n\nChange-Id: Ic6a1c69a34241f4fbe617a0b12aec9b1528ba352\nStory: 2006574\nTask: 38825\n(cherry picked from commit a91135179dc1ad2c3a657e19f7655207f8fdd0db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8727416ede867e676042fd32226d4318c24d8589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8727416ede867e676042fd32226d4318c24d8589"}]},"branch":"refs/heads/stable/train"},"2a23b89e5ed66beca92784c156763348461e3bf8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-12 17:18:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/63/711063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/711063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/711063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/711063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/711063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/711063/2"}}},"commit":{"parents":[{"commit":"d3d9dbdb5e32e0a89e99768dd1cdec2dcd6d7b3c","subject":"Remove activate-virtualenv and deactivate-virtualenv roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3d9dbdb5e32e0a89e99768dd1cdec2dcd6d7b3c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-20 16:05:06.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:35:49.000000000","tz":0},"subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","message":"CentOS 8: Use ansible_playbook_python for localhost dependencies\n\nCurrently we install python dependencies on the Ansible control host\neach time the ip-allocation and console-allocation roles are executed.\nThis is inefficient, particularly in the case of the ip-allocation role\nwhich is run serially for all hosts. It is also unnecessary since we\nhave these packages available in the Python environment used to execute\nkayobe.\n\nThe kolla-ansible role also has an implicit dependency on PyYAML for\nmanaging kolla passwords.\n\nThis change uses ansible_playbook_python as the Python interpreter for\nthe necessary tasks in these roles to avoid installing dependencies on\nthe system on CentOS 8 and Ubuntu. For CentOS 7 we still need to use the\nplatform Python, due to needing SELinux bindings.\n\nChange-Id: Ic6a1c69a34241f4fbe617a0b12aec9b1528ba352\nStory: 2006574\nTask: 38825\n(cherry picked from commit a91135179dc1ad2c3a657e19f7655207f8fdd0db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a23b89e5ed66beca92784c156763348461e3bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a23b89e5ed66beca92784c156763348461e3bf8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
