)]}'
{"id":"openstack%2Fkayobe~711064","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I2a7a82d7f576739c5516a0072f953712ffa5c233","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I2a7a82d7f576739c5516a0072f953712ffa5c233","subject":"Make Kayobe code compatible with Python 3","status":"MERGED","created":"2020-03-03 17:19:20.000000000","updated":"2020-03-13 19:00:42.000000000","submitted":"2020-03-13 18:59:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711064-1584125942639-f42c30c1","meta_rev_id":"24f53c3ec83cd69dba71b6f4e8f3e427612c51c8","_number":711064,"virtual_id_number":711064,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 18:59:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-13 12:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-13 17:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-13 17:07:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-13 12:41:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 12:42:48.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-13 17:07:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-13 18:59:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1a0a29a9c4f1115db5be02f550ef42984ffbe60","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e04449aa7f6a0808d7be708b78b3bd85f89f80c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 19:52:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19316466573a4790a39af734007161ff : SUCCESS in 9m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c22d3661c2ee424c900d62b055c6b772 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ea39facc11d42b0a3d70c54cb58b5fd : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c5ff2403e941adab607d68deebcc2c : SUCCESS in 8m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf42bf5f79594f89ac250e9c74a2e2c9 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dded64fbd4364c70afd05bb0dc110365 : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d51257be2854f1290bd63fe5fcea668 : SUCCESS in 7m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36f15b5235b24767ac24d33b9b81b22c : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/071932ef4f8c4c869f419a0e87ca4654 : SUCCESS in 31m 52s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/43a9c1e047e745b59f083d3ecee912d2 : SUCCESS in 1h 16m 46s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5b94f5e9b7c84eb1b40ff4c899891f30 : SUCCESS in 1h 48m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/20cd0de957754efd9daaf5de1a5504dd : SUCCESS in 24m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b4738e46871e486b84ced4e95c80aa98 : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4762e9afc4f535d6ff5cb4a62538ff7cce10d0c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 09:33:40.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"87eacf8e2a1ec1951f7e1ff994a933120bc7bd09","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 17:18:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"329af5e8e529db5cb9dfd6b67cb55b2feab5d198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:44:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b1623fe841a466083ce63df95152755 : SUCCESS in 5m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2e2740f7fa674cd8a4eaca58e181dab3 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09cc3db519074dfb8fae18a6088949fa : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fcf1be5f29c4c1ea835baa7ccdeb733 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7fc243412b4452186cbd13918e8aa4e : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6508e903358540ffb7a03a3c606c04ee : SUCCESS in 7m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bef49fe2b8bf4f8ba290f5ddf6e061f4 : SUCCESS in 7m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57ec4cb64f3d41029ff829c15a2e22d5 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f9017d326c94286bfb85ea812682efd : SUCCESS in 38m 00s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/06dd986dbc524840a253d86cd490c63f : SUCCESS in 1h 11m 34s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6dbd900eb30444259d0475225d897522 : SUCCESS in 2h 01m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4fe0f3c1bbce4dd1b6d5c27aeb34bb2d : SUCCESS in 22m 04s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f3f78dbb0e364d63a7841f43f91713ce : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a7c062a1ba043a3f6dd93bb0985cc8901b2d5d15","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 12:41:22.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"f1712946764cf9535af536f8e1278c61342b9d15","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-13 12:42:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0487d497ea3312e4ea74c3f32e4ae32a2fd1bfe8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 17:07:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"86c2b7d77d01934de431e56280ed6a1c3686f23d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 17:08:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc1f9f6f7c9dab78c8f5440b464486b5611a46d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 18:59:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/644946d86f6a4a319f995aa0fa650c70 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/705c351728fa401394ed5bac0ddd359a : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77106759002d4a7bbed914f484f96245 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97311734d8274a72a3ea60456880e2db : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdf5f97f2039473f86ec8f23be70ea42 : SUCCESS in 7m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d432ccfcbfab4c5a8d09bd91695293da : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a286968a3e2f4776bd1a5ea75e485b66 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a673e57d17d54b92bb3e150550afdb39 : SUCCESS in 31m 51s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d400dc8a31264cf687e6bc475baf1014 : SUCCESS in 1h 12m 06s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3fa2470e1d16435b82c13e13a2eca0cd : SUCCESS in 1h 48m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/151b4a3d51c8440dba3e0fc2108ae1f2 : SUCCESS in 22m 08s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d47b0ae194a34e53b83759a6c1cd5b90 : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a3dc3f0e399bb7cfdec66834d1ebbb473a9ee540","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 18:59:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"24f53c3ec83cd69dba71b6f4e8f3e427612c51c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 19:00:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e16c534178194af1b086a70ce2f944f6 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"20f45f2b46134ffe347b8a32efa327984565c152","revisions":{"370b7b6bc773d8515fe07a4e850d8267f0d346d0":{"kind":"REWORK","_number":1,"created":"2020-03-03 17:19:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/711064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/711064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/711064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/711064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/711064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/711064/1"}}},"commit":{"parents":[{"commit":"8727416ede867e676042fd32226d4318c24d8589","subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8727416ede867e676042fd32226d4318c24d8589"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 17:07:52.000000000","tz":0},"subject":"Make Kayobe code compatible with Python 3","message":"Make Kayobe code compatible with Python 3\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n(cherry picked from commit 15109ccb547e603c0aca2a73e43f4db523fa1bcd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/370b7b6bc773d8515fe07a4e850d8267f0d346d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/370b7b6bc773d8515fe07a4e850d8267f0d346d0"}]},"branch":"refs/heads/stable/train"},"20f45f2b46134ffe347b8a32efa327984565c152":{"kind":"REWORK","_number":2,"created":"2020-03-12 17:18:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/64/711064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/711064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/711064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/711064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/711064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/711064/2"}}},"commit":{"parents":[{"commit":"2a23b89e5ed66beca92784c156763348461e3bf8","subject":"CentOS 8: Use ansible_playbook_python for localhost dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2a23b89e5ed66beca92784c156763348461e3bf8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-02-07 17:41:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:35:49.000000000","tz":0},"subject":"Make Kayobe code compatible with Python 3","message":"Make Kayobe code compatible with Python 3\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I2a7a82d7f576739c5516a0072f953712ffa5c233\nStory: 2004959\nTask: 29392\n(cherry picked from commit 15109ccb547e603c0aca2a73e43f4db523fa1bcd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20f45f2b46134ffe347b8a32efa327984565c152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20f45f2b46134ffe347b8a32efa327984565c152"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
