)]}'
{"id":"openstack%2Fkayobe~711066","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~If5230854d7565c8b3c91a46da4795c63edf095e4","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"If5230854d7565c8b3c91a46da4795c63edf095e4","subject":"CentOS 8: Disable ntpd, enable chrony container","status":"MERGED","created":"2020-03-03 17:19:20.000000000","updated":"2020-03-20 18:44:53.000000000","submitted":"2020-03-20 18:43:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711066-1584729803456-eaa21e4c","meta_rev_id":"c8878bb64c9724ca73ebd75936cd402ebef48891","_number":711066,"virtual_id_number":711066,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 18:43:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-13 12:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-20 16:47:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-13 12:41:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 12:43:29.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-20 18:43:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49e1d5d134a8e0a60169ecb9f23dd31bd6e3e976","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d72560d1a612932f8080a69cd3dfc880430a852d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 20:13:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5787486a962492387268fb698caf2e8 : SUCCESS in 6m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d969c193981466787fc7fbcc9bfbf88 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5fbc53a58884e39b467a246fe650a29 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540908ec4b5f46c0b08072d2bc65a681 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/307967c742b34a4ba9a96833d59cc674 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684b37ee583c4217847ed33b07589cad : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fb02fcbd81b487584ebc2297070e3ba : SUCCESS in 5m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3778213e233a49b1881ece6491bca780 : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3289fac480b14e5f9dc5ac2a0badd4c2 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a4cb72279334008bee03ec0ba55d2b9 : SUCCESS in 38m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8d2870a7f241447fab134122556927f3 : SUCCESS in 1h 19m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9e45f8ed9c214a9082c9414a7bab66ae : SUCCESS in 2h 00m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/02dd8be7c36949cbaf450cc357860dfb : SUCCESS in 24m 37s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/194a38b9c3bd4ed5b08d97e758b674b3 : SUCCESS in 40m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8d0f67d3afe2d7b382d83bd889a2c7c8e52593c6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 09:36:21.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"85d947ef59db05b7e06a2ae275d4a3a16211a4f8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-05 14:12:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2332d8178d34e92a7add44316d016fd8a7d1536","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 16:01:24.000000000","message":"Patch Set 1: Workflow-1 -Review-Priority\n\nThis is a stable/train backport of https://review.opendev.org/#/c/709785/ which has not merged yet. Sorry, I\u0027ll remove RP+1.","accounts_in_message":[],"_revision_number":1},{"id":"6c405cd09dfd11e3fa1f6bdd3c149805c902d312","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 17:18:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef9fccab117db8c1d8eda5948c71219467a3240e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:59:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f3611d6123c449dbf5718f1900aa022 : SUCCESS in 5m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1032aad1b0384101ab4c2a27b3f6adef : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f2fcfd022144c44914d333bb728012c : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f447b5c8ea29497eb91d566a57274542 : SUCCESS in 7m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a43bf736e897477c94431df56fa9bd16 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ab8edf586f47fab6ece88472eae885 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7c59ae114db4afab223c8f75f9b3a44 : SUCCESS in 5m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f475b64ef0db4930add19c778431e7e6 : SUCCESS in 7m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/043e11d8c65c47e888f00c2dffd30fd6 : SUCCESS in 9m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/321040e3e5a64d6b908ab21cc188af90 : SUCCESS in 40m 25s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0c8afef8fd3c42588a8d35f166be39b5 : SUCCESS in 1h 29m 11s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2b2abd28ccac466784fb318c9cabce08 : SUCCESS in 2h 14m 05s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d3bf80c505d64c4a9e9f88825af058ad : SUCCESS in 22m 40s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bc78c772f1074ca78fec43973dbe348e : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a748f8ef94af004170270812c0517c597cc067a6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 12:41:06.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"386b50c05afd3d6cc746bbaa763309b1b6ebf1e3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-13 12:43:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"16f2da8af1f610201c20dcf9984fd1424738f914","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-20 16:47:49.000000000","message":"Patch Set 2: Workflow+1\n\nUncork a few backports.","accounts_in_message":[],"_revision_number":2},{"id":"27c1e97afc8ca9f693981ad19fd13a82e7358591","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:51:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c7b1f967103b3f60844f124817c7bb95b7517ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:43:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"75e58606e4cae5f1b0a7119c5de5e8a7a1c65f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:43:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/99b6c30211f54d42985bcd8a61e96519 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2e78b593efc4983b6f87ccbd57a4829 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f902c2ec5334bfcb8ef1fb2d6968869 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f9d988d629f4078babc28d774eb221e : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef1f95fd6fa4c1eaf4352db9342c37d : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a00c02c4b9544d3daa8dc5351ba67e07 : SUCCESS in 5m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ad2a9410baf47418070f23dd360a3e0 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/646dbedf921b46b3b7e648ff94b76930 : SUCCESS in 9m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/74b0eb9dce344a2888288ab44c4b9851 : SUCCESS in 40m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6537f4ef4b1642ac9f5f40658ba63a0c : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5a1cfd4e142b49e2a2d78e6d2c09d630 : SUCCESS in 1h 46m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4bef818169a64845bee78c734423c0fc : SUCCESS in 21m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a02ba8c1cca24bd2945aab239eb8d549 : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c8878bb64c9724ca73ebd75936cd402ebef48891","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:44:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6b7bcc4a3bb4572bf50ea0509d5f962 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a51bb68514624beaa3eca66851aee1be : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"612575a4a82a471fcd2e4c2233982c62efa496b3","revisions":{"b33f394b1b4d7bd121376576de14943ca322880b":{"kind":"REWORK","_number":1,"created":"2020-03-03 17:19:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/711066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/711066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/711066/1"}}},"commit":{"parents":[{"commit":"0177d152edcb5c40a5020c793f45e0ae07af2075","subject":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0177d152edcb5c40a5020c793f45e0ae07af2075"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 15:54:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 17:11:02.000000000","tz":0},"subject":"CentOS 8: Disable ntpd, enable chrony container","message":"CentOS 8: Disable ntpd, enable chrony container\n\nCentOS 8 does not provide an ntp package. Instead fall back to using the\nchrony container provided by Kolla Ansible by default.\n\nChange-Id: If5230854d7565c8b3c91a46da4795c63edf095e4\nStory: 2006574\nTask: 38866\n(cherry picked from commit dc8cd9d8a893be6121a846f2b0106f4222489d0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b33f394b1b4d7bd121376576de14943ca322880b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b33f394b1b4d7bd121376576de14943ca322880b"}]},"branch":"refs/heads/stable/train"},"612575a4a82a471fcd2e4c2233982c62efa496b3":{"kind":"REWORK","_number":2,"created":"2020-03-12 17:18:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/711066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/711066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/711066/2"}}},"commit":{"parents":[{"commit":"034e779152db603f437857a8df7bc009896040dd","subject":"CentOS 8: Bump MichaelRigart.interfaces to 1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/034e779152db603f437857a8df7bc009896040dd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 15:54:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:35:49.000000000","tz":0},"subject":"CentOS 8: Disable ntpd, enable chrony container","message":"CentOS 8: Disable ntpd, enable chrony container\n\nCentOS 8 does not provide an ntp package. Instead fall back to using the\nchrony container provided by Kolla Ansible by default.\n\nDepends-On: https://review.opendev.org/711666\n\nChange-Id: If5230854d7565c8b3c91a46da4795c63edf095e4\nStory: 2006574\nTask: 38866\n(cherry picked from commit 71d36cbe5e00be6e94ec28fdc1fc28126756e557)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/612575a4a82a471fcd2e4c2233982c62efa496b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/612575a4a82a471fcd2e4c2233982c62efa496b3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
