)]}'
{"id":"openstack%2Fkayobe~711067","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I7c1a885b36445e33d4db1b1c8533db28a644b4a1","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I7c1a885b36445e33d4db1b1c8533db28a644b4a1","subject":"CentOS 8: Add seed and overcloud CI jobs","status":"MERGED","created":"2020-03-03 17:19:20.000000000","updated":"2020-03-20 18:52:19.000000000","submitted":"2020-03-20 18:49:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711067-1584730170666-d1a6a7d1","meta_rev_id":"7c5b9d72d2c8e4453cff4ddb61e8b30c09b483b2","_number":711067,"virtual_id_number":711067,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 18:49:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-13 15:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-13 17:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-13 17:10:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-13 12:40:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 15:26:12.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-13 17:10:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-20 18:49:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d365744d2dfb3770022339fda9070073e244086a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 17:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3ce4467b65453694b72b3c91a1c8858d31083e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 20:22:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96fe35917cf34800ba2ca5b04cb8a22d : SUCCESS in 6m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ad4fbb9d69b643fd9e78200a65c32854 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc3a3d5ab01d43b78d1fede4311485fd : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d81c4263e176457395248b984686d4c3 : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26d9aa3b8c164415a9a91df33c88d38a : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443c004d0eef42d2aa67355ff5c3cf89 : SUCCESS in 8m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/811da877233d4f8ba7f93c02fa7bb374 : SUCCESS in 7m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f095634f88714426acb84685239878ea : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f9415792d0147359afe095b059556cf : SUCCESS in 36m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/74acd68e31bc4db990d07d45f3521022 : SUCCESS in 1h 17m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/723ebad6045d46b9a7dff7c3fc8272f6 : FAILURE in 8m 48s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/352a221af83d4a80ad927bf5147baa76 : SUCCESS in 1h 47m 53s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/26fc306b06194c6dbae1f0acc8c1a0f2 : SUCCESS in 23m 52s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/07c9036e957c46c6a17534bc700a6d20 : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6ee001df6ebb5ec33409e70d5557cb6f6d1880f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 12:51:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cbc89282242b3edb87b93e77e0eb28f6e1dc7d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 14:42:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdd1c552902346639da69d2d87aef1cf : SUCCESS in 5m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53627c2614604157b0fcf0dfcfdd1f9b : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5a1f5c7f91f420d82bf14113dbc6ca9 : FAILURE in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef9e9aaff6245ccacd4100a96cea50a : FAILURE in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1407ecdc68114a35a6526302a135a7b1 : FAILURE in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4919c08f28e40b994208b59f878de32 : SUCCESS in 7m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/081ccfd153e142cd981ce9f5c680b4cc : FAILURE in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8318d25a46494e6e8854b549d151f4c7 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a156a6066d9f4c2ead6c9b6c45059147 : SUCCESS in 41m 08s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/032ba68fc01744b7ba8887d69405f52c : SUCCESS in 1h 20m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5131ec35081e41f2b9b21a68ab0b31c8 : FAILURE in 43m 42s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bdf0700299af41aa9ac49bdd31e06045 : SUCCESS in 1h 48m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4d47b13ded6e492e9b8c8fbd000f6a56 : SUCCESS in 22m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/65560de666e946e5bdd3c70b6e46858c : SUCCESS in 48m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e72e7c08f6eee10fed98696b9a2c77a39cf46963","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 10:17:16.000000000","message":"Patch Set 2:\n\nrecheck\n\nk-a should be using correct images now","accounts_in_message":[],"_revision_number":2},{"id":"016c988fd4044f051b70ab880b147d094d784006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 12:15:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e48fabd89fd498384f4e5497697b45b : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e5aa8d149a2b44f4bddce03111d8c474 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7d0ceda98ba4362bd3471b51dcb8328 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbc9968aaf64023ab30658fc325bf3a : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7f493d989fe4b4f82a50e877439b30c : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13edf26f3abf465d92decf82d2bd9ade : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/acf7371c54f2483ca0e17aaf33119ec0 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19e8317c5a374e089faf32371253dd84 : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba0bbc944bd542d5bcd0ba91c468714a : SUCCESS in 39m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bb43a2b9cf954b55a8b02ce262914468 : FAILURE in 55m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fb3f1fe12e9543caacff9a1ba1e0a487 : FAILURE in 37m 39s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/976578011bb6421fbbc19de546d48c0b : SUCCESS in 1h 51m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/478bd117271d41398429e5228685dec2 : FAILURE in 22m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8a1aa0217db740a7acef5e92ceac48ac : SUCCESS in 34m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7bb8948ea77baec94b722adc84ef842e56651b68","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 17:18:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"648ab7b897e870eac0686a22a5de34312990b66e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:39:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f02ccd5d0a84bda8a15133177718cdc : SUCCESS in 3m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f145e2b1fc6540b29d89b2332619974c : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4136d040ab854dad971ceb48c3bb70b7 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ed261ed93a43febf8e15af00d72aa6 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ef1b89f943c48518e5f4b3dc1017915 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7378d6cfdcce41509e56e0f655a25929 : SUCCESS in 9m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e1c90d0fdf0428a9348967ee7b9b42d : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/14aced47fdea41a8be3eae9bb64939e8 : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03c5724d3d7f4868bd348f9714e19fb1 : SUCCESS in 36m 34s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/d9e73662d39e4e40a46afbaa227b4ed6 : SUCCESS in 1h 13m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/28af88fcbb1c4cd6908dec61fc5dc12a : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f7486f909b3f49b58a7e64a57192ccd1 : SUCCESS in 1h 47m 55s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5caf46a4dc5840fdaedb197af25fa305 : RETRY_LIMIT in 11m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2496f807e7714f2fb8606c06372448fe : RETRY_LIMIT in 1m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/17e43cd23c6244ecac7e2045e8e29399 : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":3},{"id":"611838378295272ed207f80a396f09b76ce94147","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 09:15:48.000000000","message":"Patch Set 3:\n\nrecheck\n\nnetwork issues in seed jobs","accounts_in_message":[],"_revision_number":3},{"id":"8b7981a79811e263702027e91a378ede1c473baa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 10:44:51.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5985ffe87f407bf49f2fbe6eb66c9eee6d0abcc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 12:36:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feaa4efcdf314cc8ba3a392f84e69eb1 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/060efad34f574fd68fa856ff7ea24236 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/340140910f1f4b748023e3b43739fcf3 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4a52b7425294236b0b6c839663dffe7 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28600a499279457893609ecb59ae9942 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a112d50d5ae4d5b9f6d597a99a37093 : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d130232a2ed24feabdfe8aa84df8cff7 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/604a4c37c3374304854dbb7a11b7e7bf : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4da575d3014423d802a7cc2ae0dd1ba : SUCCESS in 34m 06s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/669091721afc4ef1a4554ba6d382a048 : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/655d862941574d5ea2b013494a216b6d : SUCCESS in 1h 13m 49s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/dfe9552b6e7f4f6ab0a21e1b524b0aef : SUCCESS in 1h 46m 38s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/423233e28370424fa7de6833c9bf1cdd : SUCCESS in 26m 01s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/dcf6af89480f4561b10625b7eb3734bb : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":4},{"id":"155f2471603634984abaa8e218e7affd8cb18c12","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 12:40:52.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"f36b540db5c15ddd10933c11e02cf3adf5b9f386","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-13 15:26:12.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a472036e6fb22215d6e4bc1c6b76fcd7a60f9c62","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 17:10:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"11f5d673544e5db1b6bf974e20b9cc59c73cff35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:52:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2a8f36458eb79f784465825d9cf650d46c65e25b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:49:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9f0595ede68f111eb2523758495fb9245cc22660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:49:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/efabec99fe40428e898419372f50452f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8477be777eae4a78a5408d45854b685f : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb788fe840c248e4a0065dbddf0f9e70 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af5d28ceab6e4e04a7872c7498f107fe : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf14e496796144a68b8ab09dd271fa1b : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a87164d6a684608b0acc284e053adb0 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/59999ca9c1764b719441e6832767237f : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/384e59b541564c569096d505270401ec : SUCCESS in 28m 14s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9e437a95d94c4ad3bc2f1446cec58030 : SUCCESS in 1h 14m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fb39d4591aa24fe0954d8f99191faa63 : SUCCESS in 1h 08m 23s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/bd981cc8871a47b5b8767ff0f8215b2f : SUCCESS in 1h 48m 56s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/0a3c7ff0e519421a89ba30e826a5674d : SUCCESS in 25m 31s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/d9c05f1757f74f36a2e9a8e9e98452f0 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":4},{"id":"7c5b9d72d2c8e4453cff4ddb61e8b30c09b483b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:52:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/359efe3919b747628cf6b333526f45fb : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"13f5fa486bfc946c701eb89dd8f88544d11bb0f1","revisions":{"76d63be4210c61b116ad241da2a672a336809560":{"kind":"REWORK","_number":1,"created":"2020-03-03 17:19:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/711067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/711067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/711067/1"}}},"commit":{"parents":[{"commit":"b33f394b1b4d7bd121376576de14943ca322880b","subject":"CentOS 8: Disable ntpd, enable chrony container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b33f394b1b4d7bd121376576de14943ca322880b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:24:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-03 17:11:18.000000000","tz":0},"subject":"CentOS 8: Add overcloud CI job","message":"CentOS 8: Add overcloud CI job\n\nCentOS 8 removes interfaces from their bridge during ifdown, and removes\nthe bridge if there are no interfaces left. When Kayobe bounces veth\nlinks plugged into the bridge, it causes the bridge which has the IP we\nare using for SSH to be removed. Use a dummy interface in CI to avoid\nthis problem.\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: I7c1a885b36445e33d4db1b1c8533db28a644b4a1\nStory: 2006574\nTask: 38870\n(cherry picked from commit 4c390bb74355ca8d200b9854373d6ebafeece99b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76d63be4210c61b116ad241da2a672a336809560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76d63be4210c61b116ad241da2a672a336809560"}]},"branch":"refs/heads/stable/train"},"15e18d433bbb5daad27bd93255bf616099e9e2ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-04 12:51:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/711067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/711067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/711067/2"}}},"commit":{"parents":[{"commit":"8b924414746886a1a098082ee6d59042d1ba4a23","subject":"CentOS 8: Use same local python version for kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b924414746886a1a098082ee6d59042d1ba4a23"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:24:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 12:51:06.000000000","tz":0},"subject":"CentOS 8: Add overcloud CI job","message":"CentOS 8: Add overcloud CI job\n\nCentOS 8 removes interfaces from their bridge during ifdown, and removes\nthe bridge if there are no interfaces left. When Kayobe bounces veth\nlinks plugged into the bridge, it causes the bridge which has the IP we\nare using for SSH to be removed. Use a dummy interface in CI to avoid\nthis problem.\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: I7c1a885b36445e33d4db1b1c8533db28a644b4a1\nStory: 2006574\nTask: 38870\n(cherry picked from commit 4c390bb74355ca8d200b9854373d6ebafeece99b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15e18d433bbb5daad27bd93255bf616099e9e2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15e18d433bbb5daad27bd93255bf616099e9e2ea"}]},"branch":"refs/heads/stable/train"},"72ed6b9ff918d0719c1ef1e12cdc8de9d9951b6c":{"kind":"REWORK","_number":3,"created":"2020-03-12 17:18:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/711067/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/711067/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/711067/3"}}},"commit":{"parents":[{"commit":"047f84ad99598bb134b879b5a85fbcb6403752ab","subject":"CentOS 8: Use same local python version for kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/047f84ad99598bb134b879b5a85fbcb6403752ab"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:24:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:35:49.000000000","tz":0},"subject":"CentOS 8: Add seed and overcloud CI jobs","message":"CentOS 8: Add seed and overcloud CI jobs\n\nCentOS 8 removes interfaces from their bridge during ifdown, and removes\nthe bridge if there are no interfaces left. When Kayobe bounces veth\nlinks plugged into the bridge, it causes the bridge which has the IP we\nare using for SSH to be removed. Use a dummy interface in CI to avoid\nthis problem.\n\nUnlike the original version of this commit on master, CentOS 7 jobs are\nretained.\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: I7c1a885b36445e33d4db1b1c8533db28a644b4a1\nStory: 2006574\nTask: 38870\n(cherry picked from commit b0359e37dff7d3207627cfa39f28c167a81426b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72ed6b9ff918d0719c1ef1e12cdc8de9d9951b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72ed6b9ff918d0719c1ef1e12cdc8de9d9951b6c"}]},"branch":"refs/heads/stable/train"},"13f5fa486bfc946c701eb89dd8f88544d11bb0f1":{"kind":"REWORK","_number":4,"created":"2020-03-13 10:44:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/711067/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/711067/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711067/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/711067/4"}}},"commit":{"parents":[{"commit":"047f84ad99598bb134b879b5a85fbcb6403752ab","subject":"CentOS 8: Use same local python version for kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/047f84ad99598bb134b879b5a85fbcb6403752ab"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-25 16:24:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-13 10:44:48.000000000","tz":0},"subject":"CentOS 8: Add seed and overcloud CI jobs","message":"CentOS 8: Add seed and overcloud CI jobs\n\nCentOS 8 removes interfaces from their bridge during ifdown, and removes\nthe bridge if there are no interfaces left. When Kayobe bounces veth\nlinks plugged into the bridge, it causes the bridge which has the IP we\nare using for SSH to be removed. Use a dummy interface in CI to avoid\nthis problem.\n\nUnlike the original version of this commit on master, CentOS 7 jobs are\nretained.\n\nDepends-On: https://review.opendev.org/695881\n\nChange-Id: I7c1a885b36445e33d4db1b1c8533db28a644b4a1\nStory: 2006574\nTask: 38870\n(cherry picked from commit b0359e37dff7d3207627cfa39f28c167a81426b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13f5fa486bfc946c701eb89dd8f88544d11bb0f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13f5fa486bfc946c701eb89dd8f88544d11bb0f1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
