)]}'
{"id":"openstack%2Fkayobe~711219","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I96d8df005780e8d14dbdb80fe4557e8b0d037699","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I96d8df005780e8d14dbdb80fe4557e8b0d037699","subject":"CentOS 8: Use same local python version for kolla-ansible","status":"MERGED","created":"2020-03-04 12:51:20.000000000","updated":"2020-03-20 18:50:59.000000000","submitted":"2020-03-20 18:49:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711219-1584730166472-81340386","meta_rev_id":"973f068670fd46454dc22eec43d020c929d97419","_number":711219,"virtual_id_number":711219,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 18:49:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-13 15:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-13 17:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-13 17:09:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-13 12:40:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 15:14:52.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-13 17:09:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-20 18:49:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74860ae21da71b760577581bbb205b2f4c86380c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 12:51:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d51acdbad2c41dc4aa099c661205e56769b0bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 15:16:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5622526f4a4543b782b5d050eccd351f : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c004fe41b1245579ae054698f3758dd : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f362759719406fae83b48780830caa : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108d50be004d43a18628a76f7b5588cc : FAILURE in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fde67275abe4d61ad941ff44b6ab7fd : FAILURE in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/931cb52298d94d71b723ba7b8e5b7077 : SUCCESS in 6m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2bf53ccb49446b4835c257768585429 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ef912ee5a01f42b49f5432928a1b630b : FAILURE in 5m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28281d44f7024568a2bf37709fb93694 : SUCCESS in 37m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/916a53c4747e4626860dad4f41ba4a02 : SUCCESS in 1h 19m 46s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4470aa661ad24266905c5e95c5abb4b6 : SUCCESS in 2h 17m 07s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b22d57ac85bf44f39cfca7cf9628279c : SUCCESS in 24m 16s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a1fbd82ba5a1477590fac6e4fbc1ea5d : RETRY_LIMIT in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"fe4233db56f2ab2a5d2e180a9c5a4c2ad2e3cbc8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 16:04:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b291e1f922b099fdf7820fbfa3a301197298ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 18:18:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e953afdd1944658942424dda04da6f6 : SUCCESS in 6m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/94a027d7bcaa4d459d591716c57e7557 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d66c5c72d98c404594be8253f7819368 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fb85978cb840039bc1055bb22c08b8 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2214548df77477e95edd03dc8d05133 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4440861e99564e0bb17584ead7f13d8c : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba9913e65c6a4a76863e5bdbf556b7a7 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cf47a5c05eed49aca31445526d7df7c4 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd20a3e55fa74692b7bca5d37a18d929 : SUCCESS in 35m 58s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/13dc93eea23448ee98ed87c51dfcae24 : RETRY_LIMIT in 3m 03s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b904dc9a8afa46c097dd5f213b413e2f : SUCCESS in 2h 01m 20s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e9d57b3a6ded498ea9cbcdece102f627 : SUCCESS in 22m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c135ba2116eb4dc79a47e85a711aec89 : SUCCESS in 46m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a2fea0523dd58531bd15f6d0bd95f6d994fd8d00","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 09:32:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f15a562fdc83a375daef800ce8e73065d28afc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 11:44:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b0e607c405b49108af0dfbb4c0eb508 : SUCCESS in 5m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f42f137b877a4b2f8ec4f48618da49d3 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66046a78aa884e6ca90fe9b68259bf80 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9e9358a88c44b59a4bcc8db95d3024 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbe5b7f9a23841899fb1a683097031f9 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba64b9241b4a4244ae044caeca082d43 : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a3ba946939845d7ad80b3d338281a2c : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/97c04a08b801479092ea93960957be28 : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22c1fb1931a345deba23169480d4b012 : SUCCESS in 35m 26s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2fac6195dd834fab9a425de673a21a10 : FAILURE in 59m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e93bc4a1ee7a42058ad9a71284b2cd81 : SUCCESS in 2h 07m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/954284a1914a400f8a87bd0a88d76ba5 : FAILURE in 20m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8e4bcdcd23bb4740829b293b847aaa3c : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a96db7616058bf2fbaa53cc2d4a79949eb4e120e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 17:18:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a3611df7a1588768833af8836438c10a374ff4d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:44:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95c114968e5040fcb13b220e03982314 : SUCCESS in 6m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5eb6f789a0d44e5ebcafd76c9964176b : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8e0a9c47d3448dbc3c71b71fbf5007 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3648dc3bf12345d393841486282fc1ff : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/29f28a990aea4fce8c548819e8bb10f8 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/facba56d34414fac9c2c322602c1220c : SUCCESS in 10m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/50c7b7f854b34f639648e09f508d855f : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d612ae97b9c4b93a6da8ff4b8d4739b : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/570c49447af64633b464ff5d1cdd6ab8 : SUCCESS in 36m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1a1eef7b45a4420090db1482619a32fc : SUCCESS in 1h 16m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5da93dc294544aa9bb1deafe1a245fe9 : SUCCESS in 1h 50m 38s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/85a2c5654c73456cb58f785f3cacfb12 : SUCCESS in 23m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/fc05df3f5da44a24aa9df913850e1be5 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":2},{"id":"79588c181e93c2442dc879ac98bf8b180e1d6a1f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 12:40:58.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"df6821b627446f6abe325810208592a9d3800a4e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-13 15:14:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1bb2057a54913cfaef56a51268c6415a73f4b99a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 17:09:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"594a49c5ddad10c1cd6259f74b959263451f3a4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:51:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd19d4b49728d223488c585d5c02c35b14303989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:49:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fb98fa0f9ef84d4a9e9fbb3c5ce817f7 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d73178dc1884130ac6cd910e8548406 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb272d4d4144e49a2f0c7db9083c73f : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ed3a2a09b0343009c308fd77a3150fa : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf0b284ac784232af9aa01e9a3b1562 : SUCCESS in 6m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28429a86c5f94fb191996e7fcabd509e : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2af28d3af514ddc9db7c1f5e39ef2f4 : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea0e32b7d8f34b2584ebaa06086433be : SUCCESS in 37m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f6ba8577bb5541a08860ec7a20f87db3 : SUCCESS in 1h 11m 28s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f067a2a7bd334f3d92ec82d322e579ed : SUCCESS in 1h 52m 48s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a99f70c44f074963a5c3a53c159805b8 : SUCCESS in 25m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/036f2b2c34e44d1790688b4d046a39e9 : SUCCESS in 43m 30s","accounts_in_message":[],"_revision_number":2},{"id":"de1aca23a72589cd4ec80274cf591525944322d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:49:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"973f068670fd46454dc22eec43d020c929d97419","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 18:50:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ab0a69567944babf4e93ddb188b559 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"047f84ad99598bb134b879b5a85fbcb6403752ab","revisions":{"8b924414746886a1a098082ee6d59042d1ba4a23":{"kind":"REWORK","_number":1,"created":"2020-03-04 12:51:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/711219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/711219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/711219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/711219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/711219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/711219/1"}}},"commit":{"parents":[{"commit":"b33f394b1b4d7bd121376576de14943ca322880b","subject":"CentOS 8: Disable ntpd, enable chrony container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b33f394b1b4d7bd121376576de14943ca322880b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 12:50:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 12:51:05.000000000","tz":0},"subject":"CentOS 8: Use same local python version for kolla-ansible","message":"CentOS 8: Use same local python version for kolla-ansible\n\nTrain only.\n\nThe master branch switched to always use Python 3 for local\nkolla-ansible execution. For Train it is more appropriate to use the\nsame version used to execute ansible.\n\nChange-Id: I96d8df005780e8d14dbdb80fe4557e8b0d037699\nStory: 2004959\nTask: 38930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b924414746886a1a098082ee6d59042d1ba4a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b924414746886a1a098082ee6d59042d1ba4a23"}]},"branch":"refs/heads/stable/train"},"047f84ad99598bb134b879b5a85fbcb6403752ab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-12 17:18:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/711219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/711219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/711219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/711219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/711219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/711219/2"}}},"commit":{"parents":[{"commit":"612575a4a82a471fcd2e4c2233982c62efa496b3","subject":"CentOS 8: Disable ntpd, enable chrony container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/612575a4a82a471fcd2e4c2233982c62efa496b3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 12:50:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:35:49.000000000","tz":0},"subject":"CentOS 8: Use same local python version for kolla-ansible","message":"CentOS 8: Use same local python version for kolla-ansible\n\nTrain only.\n\nThe master branch switched to always use Python 3 for local\nkolla-ansible execution. For Train it is more appropriate to use the\nsame version used to execute ansible.\n\nChange-Id: I96d8df005780e8d14dbdb80fe4557e8b0d037699\nStory: 2004959\nTask: 38930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/047f84ad99598bb134b879b5a85fbcb6403752ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/047f84ad99598bb134b879b5a85fbcb6403752ab"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
