)]}'
{"id":"openstack%2Fkayobe~711484","triplet_id":"openstack%2Fkayobe~master~I535ab240b7ab8f3ab104b49170e4a9ee01fc482b","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I535ab240b7ab8f3ab104b49170e4a9ee01fc482b","subject":"Blacklist Ansible 2.8.9","status":"MERGED","created":"2020-03-05 15:38:38.000000000","updated":"2020-03-06 02:52:55.000000000","submitted":"2020-03-06 02:51:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711484-1583463078519-b70934f3","meta_rev_id":"ab38582e39f4e16b30b28b1fe8bb933b5d825c55","_number":711484,"virtual_id_number":711484,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 02:51:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-03-05 19:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-05 19:16:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 15:41:46.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-05 15:44:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-06 02:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4ffbcadccb42c487b529f1c7a6e8f52e47cc386","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 15:38:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c54e10b40dd78a10d81ca6aaa3d27f34af241bd2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-05 15:41:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e30b8ab8ef8dcf58787250a09792448cb246bfc3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-05 15:44:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8911838b02060d298d389c2b7b801742f9d19926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 16:52:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/455a218f0d8e48c2a474bb8baebf404f : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181015ac03cc4f50962fd6aa8bf4423c : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e4f82bfba0a4eecb6705b1ca62921c0 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed8d98c3ddbe484fa30a0d847cb7593c : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0396e1e22e5e4b12a52cae2281e76ca9 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7987205688e495489ea12b318f3017f : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7679def5cff148f7bad22a5b32a1e6bc : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/772d1cbced6a4ff586431e3dace7af16 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f23fb07244a048dc911a1dd850a69018 : SUCCESS in 37m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e91c511a0ada4b028bed701f5fb1f278 : SUCCESS in 1h 11m 23s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d954dc5b979d4c82b4d4c073a7657c3e : FAILURE in 1h 00m 16s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1e5edd8913ff401b9a3c164c18d00792 : SUCCESS in 21m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a31123e630ef47798bf07a6f255fb672 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0724375bb8a55e37f2353403934e3826339e7f54","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 17:00:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"00fbc39189964cfd9d966b917aeed8c5fb166cc3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 17:01:38.000000000","message":"Patch Set 2:\n\nHave to depend on a backport of this patch to get upgrade to pass.","accounts_in_message":[],"_revision_number":2},{"id":"b5a4a4ae650fa355ffdffb14935f61398a8c878e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 19:07:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3663146944964148a18b3100d857b65f : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ea6390860f45ee8e7efc04c0578f4f : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/995d0ea97b694efd8b6514e1e4f8e236 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6d8e641354e4fe4a037a55ce552928d : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01b09d5401434ac78d7c8bba12c15401 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd477ec259c40438a1d48f5b60be769 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/441961405fed474d89611acc2eecb343 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c7496619d1c4dd083070f011d2beaa2 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b4e25fbe6c144c5285c6686b50b8786b : SUCCESS in 28m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/875f6fa72a7346e5967df892e6c754cc : SUCCESS in 1h 11m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c47a66ad3fd34a57a81cfa9fa96526c8 : SUCCESS in 2h 04m 13s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b7fea9395f284abdb5427bcc94d7f50d : SUCCESS in 25m 48s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c90d00e2ea3d4c57829876882d8fd17a : SUCCESS in 42m 51s","accounts_in_message":[],"_revision_number":2},{"id":"eb1b2fe50a61fde6b46f6002742721e5e8ac5ed6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 19:16:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nPreviously approved, minor change.","accounts_in_message":[],"_revision_number":2},{"id":"ed0ca14f2f6773642dfca750c095089c5c41c952","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 19:16:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"687b94ec6242021b7cffa7210ca94f1711698b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 21:50:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d172974bfa44dc0b992cfd8699cf195 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33ac14fe6b8e417bae92c16685cf6bb2 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1353ac5d5a7b4c9eb97606cca1b1f8f5 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10ce38784ce454b88c75ceb45d9020d : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0437b3ade3ae47928272dc86cd49ca40 : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b60c33468580494585c068fbd47e795c : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e613e35449684dc1b796df87a638d278 : SUCCESS in 36m 40s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b990e9b077ea4a76912a9309ee82dd79 : SUCCESS in 1h 13m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/de11c0ac9d494dbca973c39b590e4381 : TIMED_OUT in 2h 31m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e994dcb6a75745ed8be08e64810a94bf : SUCCESS in 22m 16s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ef20bda125394203b7ea2b847bd5f261 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":2},{"id":"c492716f318632e9f1c8d36b34e690acaf5a7fb5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 22:14:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2909a358fe26c91e989b609c3d9c6f5b700e32bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 00:23:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfdc4048499944c6a082c2a5a66d7066 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93953823a97b45e19b03137f25acb203 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1d258321eab454eb0db34ba1e58754d : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0019bb3406c40508ddae6e4a7735f6a : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/477c5a17980340e18a0e73e90c7f3c75 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c926125f09441789665e3d2fb3d387a : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c4aaef6e9b5d422b81453655626833b9 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/53d37bd76a39483d8b8ce16ed9af6922 : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9100a8ec4b614754aef02007c34a202a : SUCCESS in 28m 05s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/837a9912ed8b4d269b48bfea1e8a3707 : SUCCESS in 1h 14m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6114113698984325a17df96799d982f3 : SUCCESS in 2h 08m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/209a05423fe24a2fa96a16d140567575 : SUCCESS in 21m 17s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/880dcf6644344e53a0bcc0a61eb8ccd0 : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":2},{"id":"26b58645ad71859570aa05bfb91068dbed57eec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 00:23:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c46d3e51947d1e3d4fdf6f408b1e430fa5034a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 02:51:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9022775bc6848a18a9dbafc826e8577 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25110b62cdd14f4fba62a624a3a67dcb : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e5aad0997fc47c0aa9926e24202b62a : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b11931e2dc9443c820b67948b81094f : SUCCESS in 6m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be7d15aa801a4ffc95204cd1dcf493ce : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d775c92922d848a79db17b319cedd381 : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b1d09fb946f44f5bccec29dff240f21 : SUCCESS in 36m 53s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a47609dfb032410b98b1f0babfc63901 : SUCCESS in 1h 19m 22s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bf8d8333ea42493781ea0aaceb24ab65 : SUCCESS in 2h 04m 13s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5e1d2236557049b982cd4f9acebaea10 : SUCCESS in 26m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/832e93212daf43fd8e02995b1d0fe4f7 : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3fa0022e140ad2d66038130c14eed190cbccda23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 02:51:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ab38582e39f4e16b30b28b1fe8bb933b5d825c55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 02:52:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30855d0e051a477eb4b2acd8776cff9e : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5507bba744165657cbfddd75a9f304b0bbb78be0","revisions":{"cdfb22e12c27b0a9df666ba46f1aeeda441ee6b0":{"kind":"REWORK","_number":1,"created":"2020-03-05 15:38:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/711484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/711484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/711484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/711484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/711484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/711484/1"}}},"commit":{"parents":[{"commit":"208c46e72c9be38bbb87dd44060a641c54163eb7","subject":"[ussuri][goal] More cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/208c46e72c9be38bbb87dd44060a641c54163eb7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 15:33:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 15:38:15.000000000","tz":0},"subject":"Blacklist Ansible 2.8.9","message":"Blacklist Ansible 2.8.9\n\nBackport: train\n\nOpenStack Ansible modules were broken in Ansible 2.8.9. This affects\nkolla-toolbox, and kayobe\u0027s ansible install for tasks that interact with\nAPIs. See https://github.com/ansible/ansible/issues/68042 and\nhttps://bugs.launchpad.net/kolla/+bug/1866181.\n\nThis change blacklists ansible 2.8.9 for Kayobe and Kolla Ansible. A\nseparate change will be made to kolla to blacklist ansible 2.8.9 from\nthe kolla-toolbox image.\n\nChange-Id: I535ab240b7ab8f3ab104b49170e4a9ee01fc482b\nStory: 2007383\nTask: 38959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdfb22e12c27b0a9df666ba46f1aeeda441ee6b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdfb22e12c27b0a9df666ba46f1aeeda441ee6b0"}]},"branch":"refs/heads/master"},"5507bba744165657cbfddd75a9f304b0bbb78be0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-05 17:00:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/711484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/711484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/711484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/711484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/711484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/711484/2"}}},"commit":{"parents":[{"commit":"208c46e72c9be38bbb87dd44060a641c54163eb7","subject":"[ussuri][goal] More cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/208c46e72c9be38bbb87dd44060a641c54163eb7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 15:33:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 17:00:46.000000000","tz":0},"subject":"Blacklist Ansible 2.8.9","message":"Blacklist Ansible 2.8.9\n\nBackport: train\n\nOpenStack Ansible modules were broken in Ansible 2.8.9. This affects\nkolla-toolbox, and kayobe\u0027s ansible install for tasks that interact with\nAPIs. See https://github.com/ansible/ansible/issues/68042 and\nhttps://bugs.launchpad.net/kolla/+bug/1866181.\n\nThis change blacklists ansible 2.8.9 for Kayobe and Kolla Ansible. A\nseparate change will be made to kolla to blacklist ansible 2.8.9 from\nthe kolla-toolbox image.\n\nDepends-On: https://review.opendev.org/711485\n\nChange-Id: I535ab240b7ab8f3ab104b49170e4a9ee01fc482b\nStory: 2007383\nTask: 38959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5507bba744165657cbfddd75a9f304b0bbb78be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5507bba744165657cbfddd75a9f304b0bbb78be0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
