)]}'
{"id":"openstack%2Fkayobe~711485","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I535ab240b7ab8f3ab104b49170e4a9ee01fc482b","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I535ab240b7ab8f3ab104b49170e4a9ee01fc482b","subject":"Blacklist Ansible 2.8.9","status":"MERGED","created":"2020-03-05 15:39:47.000000000","updated":"2020-03-05 19:20:18.000000000","submitted":"2020-03-05 19:18:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711485-1583435929561-1d1e633c","meta_rev_id":"b8acf01975f231024cd71f1ab9b35fb92c57f9e6","_number":711485,"virtual_id_number":711485,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 19:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-05 15:42:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-05 15:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-05 15:44:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 15:42:10.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-05 15:44:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-05 19:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0f763b0b710dca0c95d2188037b13c019424014","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 15:39:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70c6ac6223b8a16dfd2dd8785286b03a35080130","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-05 15:42:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79637c2288f8667f6944e18f0c4d13ef28e17212","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-05 15:44:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aedd41bd430e7344c8ed67bac61eb4244df263be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 17:31:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b3a84d5e2cc4bd28cdd5e874b2e086f : SUCCESS in 5m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4b275bcc339b4d4db2c309be5dcb3c9f : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4558cccf7944bd385461111ccf356ab : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64325514b20470faaa55781c91fc784 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6e4680487b9412ab14331e1a900f682 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87ed378725f04184ae56f8598e44e718 : SUCCESS in 7m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4db88e45e8fa45659c726dcdcd251e1b : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78e7718fa0ce47328986b5ad2002aefb : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d58f253b41d48749997743828b9ad32 : SUCCESS in 37m 51s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/80d94fb22a024813a528c5c3a1854254 : SUCCESS in 1h 17m 16s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/66276bc382954f6484ef6006fe0bfb96 : SUCCESS in 1h 46m 31s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e492b1ed34c44e5bbf3db452204f3377 : SUCCESS in 20m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e625a1edb2ea47f1b7d3cc89dfa9ddef : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":1},{"id":"af79e36e9f769f84f45aa9f3c2226bdc84ab3453","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 17:32:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e888c04ef53ff1e710a4ff85fcead21fbead36af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 19:18:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b4573f093a674117be4d362fb08546db : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed65b3b18b2d461e9cd213cc9e35208d : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21cc07e1bf364c02af8b92cad572e28f : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0cfc1a611f204b378fab870ae595b6f5 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8edcce423aa1490a82a228250d9400ea : SUCCESS in 11m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21e5ccaf58764a479b62ab76fe9deaa5 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49437c9c03364e6fb02a221647f501ac : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7381d555fda84ca5a91d4572a396510b : SUCCESS in 34m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b70d7c88970741ecadfbcbac8f14f84d : SUCCESS in 1h 05m 09s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d5f38f8092f2436e951712817e231a68 : SUCCESS in 1h 45m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e9f0a2fa330c4f308e9d5a03a8e8a12e : SUCCESS in 37m 25s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/1b569259a2aa420c9867c111d1c12d77 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a360d6e0c5a30e3f6a1f5602930d24dd0a70e8fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 19:18:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8acf01975f231024cd71f1ab9b35fb92c57f9e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 19:20:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451e587d6bd04f3da582f71481c925a5 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"61690b7e750dfe139cea62cb3760c20b85569dac","revisions":{"61690b7e750dfe139cea62cb3760c20b85569dac":{"kind":"REWORK","_number":1,"created":"2020-03-05 15:39:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/85/711485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/711485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/711485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/711485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/711485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/711485/1"}}},"commit":{"parents":[{"commit":"0208801f4a38617a34c71c92b7e8874dbff2adf4","subject":"[docs] Change CoreOS URLs to Centos URLs fir IPA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0208801f4a38617a34c71c92b7e8874dbff2adf4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 15:33:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 15:39:33.000000000","tz":0},"subject":"Blacklist Ansible 2.8.9","message":"Blacklist Ansible 2.8.9\n\nBackport: train\n\nOpenStack Ansible modules were broken in Ansible 2.8.9. This affects\nkolla-toolbox, and kayobe\u0027s ansible install for tasks that interact with\nAPIs. See https://github.com/ansible/ansible/issues/68042 and\nhttps://bugs.launchpad.net/kolla/+bug/1866181.\n\nThis change blacklists ansible 2.8.9 for Kayobe and Kolla Ansible. A\nseparate change will be made to kolla to blacklist ansible 2.8.9 from\nthe kolla-toolbox image.\n\nChange-Id: I535ab240b7ab8f3ab104b49170e4a9ee01fc482b\nStory: 2007383\nTask: 38959\n(cherry picked from commit cdfb22e12c27b0a9df666ba46f1aeeda441ee6b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61690b7e750dfe139cea62cb3760c20b85569dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61690b7e750dfe139cea62cb3760c20b85569dac"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
