)]}'
{"id":"openstack%2Fkayobe~711504","triplet_id":"openstack%2Fkayobe~master~I7ca5e5905887fb0ecdb34f9f44d0e669ca79906b","project":"openstack/kayobe","branch":"master","topic":"noip","hashtags":[],"change_id":"I7ca5e5905887fb0ecdb34f9f44d0e669ca79906b","subject":"Add a method to prevent an ip being set on an interface","status":"ABANDONED","created":"2020-03-05 16:37:43.000000000","updated":"2020-05-01 13:03:19.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"593c0d2af0a58191cb26f22cedd503e9919874be","_number":711504,"virtual_id_number":711504,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"date":"2020-05-01 11:35:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-27 13:52:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-05 16:49:12.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":-2,"date":"2020-03-05 16:57:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2020-03-05 17:05:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":-1,"date":"2020-03-05 16:57:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 16:49:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-05 17:05:43.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-27 13:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-01 11:35:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8798cf2a3d59930a77374fee9f06e9f1d5ef8377","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-03-05 16:37:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5417756718211a005beeacffbae6c41b3bb7b1f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-05 16:39:50.000000000","message":"Patch Set 1: Code-Review+2\n\nNo more dummy networks.","accounts_in_message":[],"_revision_number":1},{"id":"f66543a0d5007560659cf55da186ffff8b70f2f9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-05 16:41:14.000000000","message":"Patch Set 1:\n\nnit: Do we need to update docs and add a release note for people that have used dummy networks?","accounts_in_message":[],"_revision_number":1},{"id":"c8a6859abbdb4e8bac2e91305d2c0477819e1b8e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-05 16:49:12.000000000","message":"Patch Set 1:\n\n\u003e nit: Do we need to update docs and add a release note for people\n \u003e that have used dummy networks?\n\nDocs are needed anyway to explain how to use it. Release note would be nice too.","accounts_in_message":[],"_revision_number":1},{"id":"c9d93c5a02d3b7366353700410d7901e268f157d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 16:52:11.000000000","message":"Patch Set 1:\n\nIs the real issue here that the interfaces role should apply an MTU to slaves as well as the master?","accounts_in_message":[],"_revision_number":1},{"id":"d545dca72b0104f8099ae5cb54304059c8dca007","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 16:55:24.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e No more dummy networks.\n\nI don\u0027t think it avoids dummy networks always. They\u0027re useful for plugging bonds into bridges, and other things.","accounts_in_message":[],"_revision_number":1},{"id":"cab3bce265bb1151016efbd10a1d470af3acf3dd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-03-05 16:57:46.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\nThis didn\u0027t seem to work. Maybe we do need to set the mtu on the bond slaves in the interfaces role.","accounts_in_message":[],"_revision_number":1},{"id":"277fb0f01297b8e18bd3d898f303f96cc01c7197","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-05 17:04:21.000000000","message":"Patch Set 1: -Code-Review\n\nI suppose this problem goes away if you use a native VLAN for one of the VLANs you want on the bond, and then you can set the MTU for that VLAN on the base interface. Could be a pain if you wanted different MTUs.","accounts_in_message":[],"_revision_number":1},{"id":"2c84ff2e187e6c2609686195d0452d70ec0e1157","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-05 17:05:43.000000000","message":"Patch Set 1:\n\nWell, not quite sure if that will work yet, but for a single interface it does.","accounts_in_message":[],"_revision_number":1},{"id":"a5614f0dcb966498fd55617f56753131f2a56706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 18:11:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a51fafc98254476af0e14eb63a4c201 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9885755d94e44f23b3d615c2fb520a79 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0195cf1711e743119b13305bce9e33bd : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61eea11a27924ed3acaeefb641bef4c2 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd341a0b72f1449e9aee22e0b690201e : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ddbcce0d944eb5a6a5967c32547d18 : SUCCESS in 8m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f26b5b0f106456794974d08bf02bc5e : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/114980bce1ef41a0bd083914e646a277 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a096603d05a454688ee24096b983998 : SUCCESS in 34m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/54079e3ee6b44498afb9551b2ae29cff : SUCCESS in 1h 27m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9cc97d4f688040e0a78613fdf5630721 : FAILURE in 1h 02m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e4a721b5ee8049a193ca3667481f0cd2 : SUCCESS in 28m 48s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7b8f9c998838435a86bf0f0f4a2e9cad : FAILURE in 45m 36s","accounts_in_message":[],"_revision_number":1},{"id":"54e6d325fbba0f901ea8fd7b69948550f75e9535","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-27 11:58:32.000000000","message":"Patch Set 1:\n\nrecheck - this did seem useful for the case for bonds where there was no native vlan.","accounts_in_message":[],"_revision_number":1},{"id":"b83114ea456013d152a8822714211180dd674552","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 12:07:48.000000000","message":"Patch Set 1:\n\n\u003e recheck - this did seem useful for the case for bonds where there\n \u003e was no native vlan.\n\nStill needs docs :)","accounts_in_message":[],"_revision_number":1},{"id":"c207e08b1521c6379f7f94750f709da0ca6a1740","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 12:12:33.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nShould this flag also affect IP allocation, to avoid allocating an IP address?","accounts_in_message":[],"_revision_number":1},{"id":"0763d8a625917daba07d63618fe7aa86d36b9716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:52:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23e75387d28d4f8096caeb8c574a066e : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23d3c7fe30f14ce1ad3e750d6de64d44 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06b31218247b421d86af67edcd38d0b5 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/290f45b17cd54b8b83edbd2b9999f26b : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fc8e6b91aba454dbfbbe8520e927122 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc075f780534c1e8e92861ba202a22e : SUCCESS in 7m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b3d73f17223451381d363846ece5dde : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a877d45b766d41349d19be95467e24c3 : SUCCESS in 9m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/028ee34ebc9e4102ac976f75c2ce2d08 : SUCCESS in 28m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/349ea6c64e65420c8f4b7ec9513b4207 : SUCCESS in 1h 06m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9c3bff47b7b04d9b8cad3842d25f78b1 : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4237360115de4417b4339c38290a6a4d : SUCCESS in 1h 52m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cd662abb749b45d88ee5bcd9114bb01b : SUCCESS in 17m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f86b703874bb4456bf1dc005604468a7 : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d1521216a75c551c706172798e45966a48bee30e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-29 15:46:41.000000000","message":"Patch Set 1:\n\nHere is my config for setting mtu for a bond:\n\n  bonded_storage_mtu: 9000\n  bonded_storage_noip: true\n\n  storage_extra_network_interfaces:\n    - bonded_storage\n\n  bonded_storage_interface: bondstr\n\n  storage_slave1_interface: enp95s0f0\n  storage_slave2_interface: enp95s0f1\n  storage_slave3_interface: enp216s0f0\n  storage_slave4_interface: enp216s0f1\n\n\n  bonded_storage_bond_slaves:\n    - \"{{ storage_slave1_interface }}\"\n    - \"{{ storage_slave2_interface }}\"\n    - \"{{ storage_slave3_interface }}\"\n    - \"{{ storage_slave4_interface }}\"\n\nIs this a reasonable solution?","accounts_in_message":[],"_revision_number":1},{"id":"0d3cdd7393ddd082ad98c2b8f658fc944f5bd1c8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-29 15:53:05.000000000","message":"Patch Set 1:\n\nJust to clarify, then all the network interface defintions that use this bond are vlan subinterface i.e there is no native vlan:\n\n  storage_net_interface: \"{{ bonded_storage_interface }}.{{ storage_net_vlan }}\"\n  storage_net_mtu: 9000\n\n  storage_mgmt_net_interface: \"{{ bonded_storage_interface }}.{{ storage_mgmt_net_vlan }}\"\n  storage_mgmt_net_mtu: 9000","accounts_in_message":[],"_revision_number":1},{"id":"e6f53fdec0bdec48d842841f15af66c689b841a3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-29 15:56:00.000000000","message":"Patch Set 1:\n\nMaybe you are right. We could set the MTU on the bond as the max of the slave MTUs in the interfaces role. Would that work?","accounts_in_message":[],"_revision_number":1},{"id":"1ab778e055d1c9026570ef0b39d8c2744a4a7d34","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 19:52:41.000000000","message":"Patch Set 1:\n\nI just tried to reproduce your setup using a seed hypervisor.\n\ngroup_vars/seed-hypervisor:\n\nvlan42_interface: bond0.42\nvlan43_interface: bond0.43\nbond_interface: bond0\nbond_bond_slaves:\n  - dummy-slave-0\n  - dummy-slave-1\n\nnetworks.yml:\n\nvlan42_vlan: 42\nvlan42_cidr: 10.42.0.0/24\nvlan42_mtu: 9000\n\nvlan43_vlan: 43\nvlan43_cidr: 10.43.0.0/24\nvlan43_mtu: 9000\n\nbond_mtu: 9000\n\nseed-hypervisor.yml:\n\nseed_hypervisor_extra_network_interfaces:\n  - vlan42\n  - vlan43\n  - bond\n\nI created the dummy interfaces for the bond slaves like this:\n\nsudo ip l add dummy-slave-0 type dummy\nsudo ip l add dummy-slave-1 type dummy\n\nThey had the 9k MTU applied:\n\n8: bond0: \u003cNO-CARRIER,BROADCAST,MULTICAST,MASTER,UP\u003e mtu 9000 qdisc noqueue state DOWN mode DEFAULT group default qlen 1000                                                                                \n    link/ether 12:44:3d:0b:57:f6 brd ff:ff:ff:ff:ff:ff qdisc noqueue state LOWERLAYERDOWN mode DEFAULT group default qlen 1000                                                                   \n    link/ether 12:44:3d:0b:57:f6 brd ff:ff:ff:ff:ff:ff\n11: bond0.43@bond0: \u003cNO-CARRIER,BROADCAST,MULTICAST,UP\u003e mtu 9000 qdisc noqueue state LOWERLAYERDOWN mode DEFAULT group default qlen 1000                                                                   \n    link/ether 12:44:3d:0b:57:f6 brd ff:ff:ff:ff:ff:ff\n12: dummy-slave-1: \u003cBROADCAST,NOARP\u003e mtu 1500 qdisc noop state DOWN mode DEFAULT group default qlen 1000                                                                                                   \n    link/ether 0e:01:c3:ff:d7:69 brd ff:ff:ff:ff:ff:ff\n13: dummy-slave-2: \u003cBROADCAST,NOARP\u003e mtu 1500 qdisc noop state DOWN mode DEFAULT group default qlen 1000                                                                                                   \n    link/ether de:8e:8e:6a:5c:c4 brd ff:ff:ff:ff:ff:ff","accounts_in_message":[],"_revision_number":1},{"id":"ce4bdd7e80e293e07b6d014ca9e8953279053a57","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-01 08:36:19.000000000","message":"Patch Set 1:\n\nBut the bond has the address 0.0.0.0, right? From what I can remember if you have more than one such bond on the network. If you do a systemctl restart network, it will detect the duplicate IP and refuse to start.","accounts_in_message":[],"_revision_number":1},{"id":"15d08850a197a77b4e12aec1f4905cf16fb5aa4c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-01 11:15:40.000000000","message":"Patch Set 1:\n\nIt just doesn\u0027t have an address:\n\n20: bond0: \u003cBROADCAST,MULTICAST,MASTER,UP,LOWER_UP\u003e mtu 9000 qdisc noqueue state UP group default qlen 1000\n    link/ether ae:45:a1:8c:5d:53 brd ff:ff:ff:ff:ff:ff\n    inet6 fe80::ac45:a1ff:fe8c:5d53/64 scope link \n       valid_lft forever preferred_lft forever","accounts_in_message":[],"_revision_number":1},{"id":"42b5f43a25f930541bd82ed5173646c0b9648be4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-01 11:35:46.000000000","message":"Patch Set 1:\n\n\u003e It just doesn\u0027t have an address:\n \u003e \n \u003e 20: bond0: \u003cBROADCAST,MULTICAST,MASTER,UP,LOWER_UP\u003e mtu 9000 qdisc\n \u003e noqueue state UP group default qlen 1000\n \u003e link/ether ae:45:a1:8c:5d:53 brd ff:ff:ff:ff:ff:ff\n \u003e inet6 fe80::ac45:a1ff:fe8c:5d53/64 scope link\n \u003e valid_lft forever preferred_lft forever\n\nMade a second bond, no problemo\n\n28: bond2: \u003cBROADCAST,MULTICAST,MASTER,UP,LOWER_UP\u003e mtu 9000 qdisc noqueue state UP group default qlen 1000\n    link/ether 72:a3:91:7f:90:3c brd ff:ff:ff:ff:ff:ff\n    inet6 fe80::70a3:91ff:fe7f:903c/64 scope link \n       valid_lft forever preferred_lft forever","accounts_in_message":[],"_revision_number":1},{"id":"f997d79f242ca6d96e1bc5e4e8ad2a3796c537e2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-01 13:03:19.000000000","message":"Abandoned\n\nOK, thanks Mark. Seems this is unnecessary then. Not quite sure how I hit the issue I did, but with the time pressure I forgot to collect any logs *facepalm*.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d54a2ab14d70565845ec3af0aa2c448e040b116d","revisions":{"d54a2ab14d70565845ec3af0aa2c448e040b116d":{"kind":"REWORK","_number":1,"created":"2020-03-05 16:37:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/04/711504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/711504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/711504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/711504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/711504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/711504/1"}}},"commit":{"parents":[{"commit":"208c46e72c9be38bbb87dd44060a641c54163eb7","subject":"[ussuri][goal] More cleanup for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/208c46e72c9be38bbb87dd44060a641c54163eb7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-03-05 16:36:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-03-05 16:36:19.000000000","tz":0},"subject":"Add a method to prevent an ip being set on an interface","message":"Add a method to prevent an ip being set on an interface\n\nThis can be used to set the MTU on interfaces that make up\na bond.\n\nChange-Id: I7ca5e5905887fb0ecdb34f9f44d0e669ca79906b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d54a2ab14d70565845ec3af0aa2c448e040b116d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d54a2ab14d70565845ec3af0aa2c448e040b116d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
