)]}'
{"id":"openstack%2Fkayobe~711666","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I640873c11ceae5008030dc03984c089a410a0cee","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I640873c11ceae5008030dc03984c089a410a0cee","subject":"Skip resmo.ntp role if ntp_service_enabled is false","status":"MERGED","created":"2020-03-06 17:06:45.000000000","updated":"2020-03-19 12:04:59.000000000","submitted":"2020-03-19 12:02:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711666-1584619376081-51d1296f","meta_rev_id":"2f16477abf0e2971fb633a562b2bedeea4c131df","_number":711666,"virtual_id_number":711666,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 12:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-19 10:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-13 17:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-03-19 10:10:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 17:08:47.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-19 10:10:59.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-19 12:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b488fce83194a1d2f90a590f8ad5d4aa67330ae7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 17:06:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"95c319e3e64a1a8b1e0f14ca45e6478c56f60829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 19:18:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cda69e262b845ee9ed65b35f45b9a54 : SUCCESS in 6m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/361cec64bcf6420d9252a725635a3332 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59ca1065cb5f43faac3ea4ee348560fb : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69703f249d294e779a95643b49053f2d : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82cdee1251fd47ee9c98034e75ec0837 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04bd20629674cb79a9e6cba5479dde0 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44a259c57eb944ccaefbd5bbed5a3cb6 : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73d529fea7134462a5ba7f5b402f2bc8 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae106d616271429c87651abf9a759faf : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6aee07bf365f486db1a4d921462a1e2c : SUCCESS in 35m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bd3238f24cf144aeae74938e16a4edc8 : SUCCESS in 1h 12m 13s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8bbe0e69499248dfb52ea06e77bf7703 : SUCCESS in 2h 06m 11s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4276743c98d34ab3ac5a6b11793831e6 : SUCCESS in 27m 28s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0f46d470b8bd4a088df44a62d3ad1b22 : SUCCESS in 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b7554e6970ed03612659ead1cc4670e787bbc152","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 17:08:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f8652b6d5c76a62d069be2848c49f99dce196b9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-19 10:10:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81d903ef63a1251fae819e0e9cc27bf9efc8d467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 10:11:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09488c3fda2cc7e6703f287bf86fe3a90bc56add","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:02:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1f3a1a6b539241c191d34549950a1daa : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f914b5ae4704913b23222e7705474c8 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d53eb659bfcb49deba53d4f3e82e5c24 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f01604055fd84754962c3a085e607757 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e748c0f3154df3b0a32f7d8c4061fc : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38619f9c08cb44389d8b98ca1573fd73 : SUCCESS in 5m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/325a28263903456c9763eff1f3da7c3d : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5d873e237d74b1a8d5e7b29c7c0dcc0 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7696547112074bfeb1fd117ab0b503c4 : SUCCESS in 36m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5b29776da34d4297aa62c0caca11a3e5 : SUCCESS in 1h 12m 11s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/844fa549187d468f9b46313c7888ad62 : SUCCESS in 1h 47m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/80245e17293c48a1ba86dcd951ce1d43 : SUCCESS in 26m 53s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f9e68c1b93b042b7a1308ba0ca8b85f5 : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":1},{"id":"703dc935353c468b30721ba5c6eb9b70aa98652b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:02:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2f16477abf0e2971fb633a562b2bedeea4c131df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:04:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbc834faef94554b89815a6adb4e3d2 : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cefaac0090ec4e5c92b820ad61d1d6cc : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"369fc3a86aa282f3b925bb0bd6b38b07947661d6","revisions":{"369fc3a86aa282f3b925bb0bd6b38b07947661d6":{"kind":"REWORK","_number":1,"created":"2020-03-06 17:06:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/711666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/711666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/711666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/711666/1"}}},"commit":{"parents":[{"commit":"61690b7e750dfe139cea62cb3760c20b85569dac","subject":"Blacklist Ansible 2.8.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61690b7e750dfe139cea62cb3760c20b85569dac"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 16:46:05.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 17:06:45.000000000","tz":0},"subject":"Skip resmo.ntp role if ntp_service_enabled is false","message":"Skip resmo.ntp role if ntp_service_enabled is false\n\nWe enable ntpd by default, and provide a variable to disable it -\nntp_service_enabled. It is also automatically disabled if the user\nenables the chrony container (kolla_enable_chrony).\n\nHowever, setting ntp_service_enabled to false will cause the host\nconfigure commands to fail due to a bug in the resmo.ntp role. This is\nbecause it tries to configure the ntpd service in systemd, but it will\nnot exist so the task fails.\n\nThis change fixes the issue by skipping the resmo.ntp role if the NTP\nservice is disabled.\n\nChange-Id: I640873c11ceae5008030dc03984c089a410a0cee\nStory: 2007384\nTask: 38968\n(cherry picked from commit 35e50bb6b8f149d9edac6718121407b08ff402d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/369fc3a86aa282f3b925bb0bd6b38b07947661d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/369fc3a86aa282f3b925bb0bd6b38b07947661d6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
