)]}'
{"id":"openstack%2Fkayobe~711667","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I640873c11ceae5008030dc03984c089a410a0cee","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I640873c11ceae5008030dc03984c089a410a0cee","subject":"Skip resmo.ntp role if ntp_service_enabled is false","status":"MERGED","created":"2020-03-06 17:06:57.000000000","updated":"2020-03-19 12:01:55.000000000","submitted":"2020-03-19 11:59:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711667-1584619168412-4c3f8529","meta_rev_id":"01923ffc01dfec348840bef5c5836d100605e29c","_number":711667,"virtual_id_number":711667,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 11:59:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-19 10:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-13 17:09:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-03-19 10:09:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 17:09:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-19 10:09:50.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-19 11:59:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdcf59296590524bd915164a028bdbb4c0acc684","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 17:06:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c6c7065fd452286c19c32cbb57954612860ef266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 19:07:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1e5dd36840d4e2298f66c67fe0ed07c : SUCCESS in 4m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/54ca693f080c4019b534199599a9f563 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b4aa334087b40b289b15302837b1ac5 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff3292beb49e4db2be17e66b9d8bbb70 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9d271c0d316f446190f0eff01c3ae395 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47b0c84b27947288d2b298ae8308a04 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b8b71af25140adad663bc1f2fb6763 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a9f0dada71384e51a3b651359bbfa708 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/865c918c26844f539e7eb0f51802ee0b : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e7ebeeb349b4cd4a00d4711a84ca15f : SUCCESS in 34m 16s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/09995b711fe4433891b433258cef7fdc : SUCCESS in 1h 05m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/34738304571e49f0a0a4eeee20af7344 : SUCCESS in 1h 54m 32s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2f24166eeab5474c82e8595c0dcbbc14 : SUCCESS in 23m 44s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/06a697b10e6a4d0e915228a0c2735dff : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2be3a0dafc9189c5826762fff45c268f69e44998","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 17:09:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"586d8ec15d5533edbc406f6bd5d6cc005141e491","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-19 10:09:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2e1545d7f7fe679c316fefed6851f6d0bf2a8f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 10:10:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c170ba6053fb18f6c5059416275f4693c7d272b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:59:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"01203a4054ced4d903a2e4d5f42cdb51e4bd19bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:59:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a98ab6bb90264b5ead921e5b9c857b4e : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28e55b782ec54adda19dee6c6868f6b0 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1f20e45dd9483e9b7f7e45c25715bb : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89cac274e13742799f150cfe710ef30b : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1ab7caf00ba4743b82bbdd578258f7c : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fdad0cef67148ad8d82bcd2a59b9727 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/580c1a92981d4f9b862f5dd55b69115f : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/053ca496d15a4c1a996d1cdd575a1a7c : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3cb7bc9df9fd4757a44c39d8669b0396 : SUCCESS in 34m 37s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b7034ef7e1f940729ecbedff7974c151 : SUCCESS in 1h 02m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/86e80c337e064918be43975ff8674750 : SUCCESS in 1h 45m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6fe0956df2a54442a263cf57949d9967 : SUCCESS in 20m 44s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/07a441e13161479c99e4c431b2751df7 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"01923ffc01dfec348840bef5c5836d100605e29c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:01:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88bc886a6ccb4f8bb863972a95b173f3 : SUCCESS in 1m 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3eeb5b49d0d54c2f8034f52ec54f88ef : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d9383a6d95e51803d27dd61dc7df3969f054e16","revisions":{"8d9383a6d95e51803d27dd61dc7df3969f054e16":{"kind":"REWORK","_number":1,"created":"2020-03-06 17:06:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/711667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/711667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/711667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/711667/1"}}},"commit":{"parents":[{"commit":"26c8abc6642af90e1436b68c4fedad27c40f7cae","subject":"Merge \"Fix use of IPA images in Bifrost without a checksum\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26c8abc6642af90e1436b68c4fedad27c40f7cae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-05 16:46:05.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 17:06:57.000000000","tz":0},"subject":"Skip resmo.ntp role if ntp_service_enabled is false","message":"Skip resmo.ntp role if ntp_service_enabled is false\n\nWe enable ntpd by default, and provide a variable to disable it -\nntp_service_enabled. It is also automatically disabled if the user\nenables the chrony container (kolla_enable_chrony).\n\nHowever, setting ntp_service_enabled to false will cause the host\nconfigure commands to fail due to a bug in the resmo.ntp role. This is\nbecause it tries to configure the ntpd service in systemd, but it will\nnot exist so the task fails.\n\nThis change fixes the issue by skipping the resmo.ntp role if the NTP\nservice is disabled.\n\nChange-Id: I640873c11ceae5008030dc03984c089a410a0cee\nStory: 2007384\nTask: 38968\n(cherry picked from commit 35e50bb6b8f149d9edac6718121407b08ff402d8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d9383a6d95e51803d27dd61dc7df3969f054e16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d9383a6d95e51803d27dd61dc7df3969f054e16"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
