)]}'
{"id":"openstack%2Fkayobe~712133","triplet_id":"openstack%2Fkayobe~master~Ie0edf6a924a914395c6502e2d5cf1139bce14a48","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ie0edf6a924a914395c6502e2d5cf1139bce14a48","subject":"CentOS 8: seed VM \u0026 bifrost","status":"MERGED","created":"2020-03-10 17:35:31.000000000","updated":"2020-04-15 16:31:31.000000000","submitted":"2020-04-15 16:28:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712133-1586968127697-170bf913","meta_rev_id":"642e396a2b509e5566872d91f90d98573f930a3c","_number":712133,"virtual_id_number":712133,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 16:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-04-15 14:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-09 16:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-04-15 14:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-04-15 16:31:31.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 17:13:47.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-09 16:16:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-15 14:19:39.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-04-15 16:28:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f6086ebb79afa354d503d3f17a01dc46341a84a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-10 17:35:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cd76925c17120ba596a50a5bc1aacd64adef0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 19:03:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7654d14da98e48f587ded1efeaa54fb3 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/675dfbdd4e064bb6b2d0c117f40e56eb : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc670748ddb042a480e55a07420746b5 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a76fc25238c4ace9154afe48040342a : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d9380ad382b4a9da530e0fad85ef488 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a3654d15e14af0a2cab093daf1de07 : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f8c62d08e6a14a0e83470ab7770064ea : FAILURE in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6df68d40e6e8413894cbf4ba8edb9e29 : FAILURE in 5m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/980c69732f7a4f2b902b1ada3a048d7b : FAILURE in 8m 09s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/65574d88efdc4477b4b0023ac3e66397 : FAILURE in 5m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f02df7db2d294a4f9f76b339fe4c7189 : FAILURE in 8m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6070177f67de466bb5c5d97f9701edd6 : FAILURE in 1h 17m 38s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/437c08ebbb434f299180edcc735b8317 : FAILURE in 5m 12s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e4e40ada4cd34f45beef2990b64972a6 : FAILURE in 22m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6e0be337830a364e8a14013faa331db67fd0be55","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 10:53:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"110b595d22894a1cfe5ba5072daea56ed1af3d28","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 10:54:19.000000000","message":"Patch Set 2: Workflow-1\n\nTenks broken by virtualbmc 2.0.0","accounts_in_message":[],"_revision_number":2},{"id":"85c6404f3aefca050bb89df63b13d9c74f242ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 12:29:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77b569ac7d224196af66d0bcd1fc1dce : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316a323475d041f298b94f68db2965fd : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9198be2507cc413e80f8eb652421d71e : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/380631fbd9ab4ad489cea16b9e9d628b : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28bb0090ad74490fa1b8571325a3496f : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf74c76b2004ca7b498ca2dc78b2af6 : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4cf9409fe03248728686ef4840770474 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/673208159ef940a9a1fe0f31b22175f0 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5eaf19fab3d64c129f835e02975c792d : SUCCESS in 36m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/79ceb4022067468e935a825672bf9833 : FAILURE in 1h 00m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0cc15d0273de481c95b89174bcf29936 : SUCCESS in 1h 24m 13s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4afb545b4ac548c9870af142290d5298 : FAILURE in 1h 16m 25s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3c667ccc70bf47a2984b1b872e0d2191 : SUCCESS in 27m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/19b0418ffd19483980699875ddc487e7 : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2d6b69418b23e4ffd9afde2b651d78f8245dcf28","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 17:10:47.000000000","message":"Patch Set 2:\n\nrecheck\n\ntenks fixed","accounts_in_message":[],"_revision_number":2},{"id":"5b1cc88bed64563c8c8ad3cd699fd0baa81d489e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 17:11:04.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9009916c4a4617040b8e384b43892a703cf6033e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:21:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f275af312fd44ef9ae9cecb4b2b0e0d6 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2b59ac6c4b4da3bc841469608f30e1 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98980668a49d4ac891866b7d2e896b73 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4c301098cc04fa1b4e3f660718dfab0 : SUCCESS in 13m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13c4c23498d14e5a8cd3bbee1371036f : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3050841df4cc468a854127c2b828dc93 : SUCCESS in 10m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/333f4495fe0e4a5d9fd07fc59d50f97b : SUCCESS in 10m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/227a7b4d84dc41c6935418c806fba658 : SUCCESS in 17m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4eaa234368274fa89a52083a8c362109 : SUCCESS in 39m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/185216aeae134eb9a0ee018f99424ba7 : SUCCESS in 1h 10m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8a6087e3c0c64d82b932840652383de6 : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":3},{"id":"fd8149279a625d6521dbadae7122aac2933f519b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-18 10:25:16.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"45e24a77e21aad400037466bb3196e5da6d1f111","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-18 17:13:47.000000000","message":"Patch Set 3: Code-Review+2\n\nnice!","accounts_in_message":[],"_revision_number":3},{"id":"f52c6d8693080e49f439ff0869f54ff1ec5d021a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-03-27 21:46:07.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nMostly looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"3efb1e47757d068db88876ca9b1a021795e540d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 14:10:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c9cad1cbb64d11055b67d3c2cd7752b9ebd29707","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 14:11:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b5b164fad38583701d8ca912cafda4841efe9d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:49:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f345545def5f447892106b9a82f69f10 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f770d6ba6f430c80637984f52fe509 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6615503945324bcba41455bb22ab8d83 : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97683d31782d439dabe5171d9a6e48cd : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ea90d47ad346efac9200021594294e : FAILURE in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fc2bc1cd9c46c289f9324cce3dde5b : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d1de8ce2fef418a877d4cb1ae80f03b : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d241fe6c5d6b4f459dbc9b4a846012f0 : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa491e9aeb824873b5583629dbfeb498 : SUCCESS in 39m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b5e0e98c97ac4936b465b9483fade282 : SUCCESS in 1h 24m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7631b17b62d64be1b991058736bafd12 : SUCCESS in 1h 48m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1b33b9c3f6bc4db997b101ab728ae6ef : SUCCESS in 17m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0b703abd65314dfbb6a6097615776c81 : SUCCESS in 28m 37s","accounts_in_message":[],"_revision_number":4},{"id":"183a010673e4ee048c11ce099bd5a3bf373fd0a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 10:27:01.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"45a461e5f04d39d7b8c78b8514240ca40a46c5a1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-07 10:36:41.000000000","message":"Patch Set 4:\n\nI guess the py38 job is fixed by https://review.opendev.org/#/c/717266/","accounts_in_message":[],"_revision_number":4},{"id":"4e8489312ca0f4be642c954ab7d8ab0439f6078b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 16:54:26.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ffec30f5bb26e8883cdb1929ce0338ef68baf9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:20:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4bf9f5fdabb466a806af2b870daf713 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aac73a1a9b04119bfd3c5ed65283a8e : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd0533060f949e99ed6d42e379dbfad : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b588d60b241c405dbe601e3fef9cc837 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2e16a67b0b3459099c615923cd45fd3 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4bb3e7658044849b83005819df58e5 : SUCCESS in 9m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/beb327b377a24a4a92d7a7c853763511 : SUCCESS in 8m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ef580725249452f87be6ee39c27c16f : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6348736762f54c67b5580e6b82547946 : SUCCESS in 28m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/33fe3dc971a3480c928e6b4a9d0dc654 : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3d49b633b1584387b7c50cad0276b9d1 : SUCCESS in 1h 56m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8deff90e086a42729f05aeee197b151d : SUCCESS in 20m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3ed8fc15e1454a9a8c515ffc6e7fdf26 : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":5},{"id":"70aa919a366009ce462686989b80ee546ab564a9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-09 08:48:38.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7f003457d100445acd55254fb7849ad9c194c9da","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-09 13:17:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6d78921c6ec8e472e21656be390c5ad1a814faae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-09 13:25:17.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9592e554143ee9ac44f7c454da7cb70a9dc8e107","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 14:02:41.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a3689da9149c83401c84ca71e9eb387f582ec937","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 14:02:47.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"cfe626aaf5edbb789583d1b1573823d222233d32","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-09 14:04:22.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"d225e2f42570d28fecd847803f31dc10ce0f76be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 16:09:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc26c006445847fc91e85edb2344a339 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42c46ccdc5a4505bc5f26efc36a70e0 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee6bffb4086440a8b37aaad7ba296e3f : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/39138b0c33b94727baf57eefcc3efe1d : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63b2aee7c3cf454a9fbffa7221318a68 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2152d4a9865d49498a0761eee897633d : SUCCESS in 10m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/efd465963b064a70a560f31cafca6a2c : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff9aaef1a1e845fe9b4e5af2f6c3a638 : SUCCESS in 13m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/05746306f1cf416d8ddaae474b13078a : SUCCESS in 39m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/49e2b5a10fb245778048597249ae2da3 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/80f2969f0427429fadc707ecf7283130 : SUCCESS in 1h 56m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/abf0831fa8ae47ec91c7c1861f438045 : SUCCESS in 20m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b1c7a2f09ff84c25933e0235a41b4472 : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":7},{"id":"2bf41131a5ad439cf884091d5d3df081109019e8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-09 16:16:11.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"326ecd5246b4522d19af1051fc6994e52c1f13ec","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-15 14:19:39.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e9f1292ae90bb42c0b280546fbc7d7e4a61c6b61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:19:51.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f127ef7b31785a2f094d1e07ad546616aea2a1c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 16:28:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d12ababcea94e39488458f45a5a839613a7688bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 16:28:47.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0a0e78202274278ba6d1f4b52eb6dad : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7744003b9bb492bbf679ae68fd65f87 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0aced32cefea4ed0af6c59fba49a9382 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3e88847cfe45c4917c0c79632ba870 : SUCCESS in 8m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d85928178ca4be6a70b502dba69381c : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af324d117cde4938a5122604456033c8 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a42e8dfed2f341b9b90250ffec39eef8 : SUCCESS in 30m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/628f8974e23b41f48ae500436e4ed907 : SUCCESS in 1h 12m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/99de74b9abc94f4ba1bbb338894b8bff : SUCCESS in 2h 07m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/84a6157533af42b5abeeee5d1ca9f2b1 : SUCCESS in 21m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fca5bd1d82ca4e3fa6cd8ba48a2bfb83 : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":7},{"id":"e63a315f1dc3222e3935245a09efc09de6f2bbe7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 16:31:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e8e7d4f8bcf4722b81a83e5fb7c4637 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"51b84b6001e90435aa76aa09a763edf8063ed085","revisions":{"0cc437e26b7ddacbec248df187b62fa0ca9361f5":{"kind":"REWORK","_number":1,"created":"2020-03-10 17:35:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/712133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/712133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/712133/1"}}},"commit":{"parents":[{"commit":"a8b6527e275dab7b6802e76f0ae44dda967d6613","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8b6527e275dab7b6802e76f0ae44dda967d6613"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:46:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 17:21:16.000000000","tz":0},"subject":"CentOS 8: seed VM \u0026 bifrost","message":"CentOS 8: seed VM \u0026 bifrost\n\n* Change default seed VM image to CentOS 8\n* Change default bifrost deploy image to CentOS 8\n* Workaround DIB bug\n  https://bugs.launchpad.net/diskimage-builder/+bug/1866847 by setting\n  DIB_DISABLE_KERNEL_CLEANUP to 1\n* Install iptables on seed for SNAT - missing on CentOS 8\n* Fix provider network MTU lookup for empty string\n* Bump stackhpc.libvirt-host to 1.7.0 for CentOS 8 support\n* Bump stackhpc.libvirt-vm to 1.13.0 for CentOS 8 support\n* Bump jriguera.configdrive for Python 3 support\n\nChange-Id: Ie0edf6a924a914395c6502e2d5cf1139bce14a48\nStory: 2006574\nTask: 39000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cc437e26b7ddacbec248df187b62fa0ca9361f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cc437e26b7ddacbec248df187b62fa0ca9361f5"}]},"branch":"refs/heads/master"},"2e8cde9e4cc9fcdd024dbd0f2d139692de16c21a":{"kind":"REWORK","_number":2,"created":"2020-03-11 10:53:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/712133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/712133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/712133/2"}}},"commit":{"parents":[{"commit":"a8b6527e275dab7b6802e76f0ae44dda967d6613","subject":"CentOS 8: Add overcloud CI job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a8b6527e275dab7b6802e76f0ae44dda967d6613"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:46:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-11 10:53:53.000000000","tz":0},"subject":"CentOS 8: seed VM \u0026 bifrost","message":"CentOS 8: seed VM \u0026 bifrost\n\n* Change default seed VM image to CentOS 8\n* Change default bifrost deploy image to CentOS 8\n* Workaround DIB bug\n  https://bugs.launchpad.net/diskimage-builder/+bug/1866847 by setting\n  DIB_DISABLE_KERNEL_CLEANUP to 1\n* Install iptables on seed for SNAT - missing on CentOS 8\n* Fix provider network MTU lookup for empty string\n* Bump stackhpc.libvirt-host to 1.7.0 for CentOS 8 support\n* Bump stackhpc.libvirt-vm to 1.13.0 for CentOS 8 support\n* Bump jriguera.configdrive for Python 3 support\n\nChange-Id: Ie0edf6a924a914395c6502e2d5cf1139bce14a48\nStory: 2006574\nTask: 39000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e8cde9e4cc9fcdd024dbd0f2d139692de16c21a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e8cde9e4cc9fcdd024dbd0f2d139692de16c21a"}]},"branch":"refs/heads/master"},"4a8fd94a6eead4da78c89714ba57fa11f4b71386":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-11 17:11:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/712133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/712133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/712133/3"}}},"commit":{"parents":[{"commit":"b0359e37dff7d3207627cfa39f28c167a81426b5","subject":"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0359e37dff7d3207627cfa39f28c167a81426b5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:46:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-11 17:11:04.000000000","tz":0},"subject":"CentOS 8: seed VM \u0026 bifrost","message":"CentOS 8: seed VM \u0026 bifrost\n\n* Change default seed VM image to CentOS 8\n* Change default bifrost deploy image to CentOS 8\n* Workaround DIB bug\n  https://bugs.launchpad.net/diskimage-builder/+bug/1866847 by setting\n  DIB_DISABLE_KERNEL_CLEANUP to 1\n* Install iptables on seed for SNAT - missing on CentOS 8\n* Fix provider network MTU lookup for empty string\n* Bump stackhpc.libvirt-host to 1.7.0 for CentOS 8 support\n* Bump stackhpc.libvirt-vm to 1.13.0 for CentOS 8 support\n* Bump jriguera.configdrive for Python 3 support\n\nChange-Id: Ie0edf6a924a914395c6502e2d5cf1139bce14a48\nStory: 2006574\nTask: 39000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a8fd94a6eead4da78c89714ba57fa11f4b71386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a8fd94a6eead4da78c89714ba57fa11f4b71386"}]},"branch":"refs/heads/master"},"b4e7fddb43ec77647bd6f0d64401bc0c93d2596d":{"kind":"REWORK","_number":4,"created":"2020-04-02 14:10:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/712133/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/712133/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/712133/4"}}},"commit":{"parents":[{"commit":"35ccc0d2ccea88efbdf6671df303ad8edce6f2d2","subject":"Merge \"CentOS 8: Enable seed upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ccc0d2ccea88efbdf6671df303ad8edce6f2d2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:46:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 14:10:31.000000000","tz":60},"subject":"CentOS 8: seed VM \u0026 bifrost","message":"CentOS 8: seed VM \u0026 bifrost\n\n* Change default seed VM image to CentOS 8\n* Change default bifrost deploy image to CentOS 8\n* Workaround DIB bug\n  https://bugs.launchpad.net/diskimage-builder/+bug/1866847 by setting\n  DIB_DISABLE_KERNEL_CLEANUP to 1\n* Install iptables on seed for SNAT - missing on CentOS 8\n* Fix provider network MTU lookup for empty string\n* Bump stackhpc.libvirt-host to 1.7.0 for CentOS 8 support\n* Bump stackhpc.libvirt-vm to 1.13.0 for CentOS 8 support\n* Bump jriguera.configdrive for Python 3 support\n\nChange-Id: Ie0edf6a924a914395c6502e2d5cf1139bce14a48\nStory: 2006574\nTask: 39000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4e7fddb43ec77647bd6f0d64401bc0c93d2596d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4e7fddb43ec77647bd6f0d64401bc0c93d2596d"}]},"branch":"refs/heads/master"},"68dc0cd469295dbf8becf47ce4f03bbdb9962f79":{"kind":"REWORK","_number":5,"created":"2020-04-08 16:54:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/712133/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/712133/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/712133/5"}}},"commit":{"parents":[{"commit":"35ccc0d2ccea88efbdf6671df303ad8edce6f2d2","subject":"Merge \"CentOS 8: Enable seed upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ccc0d2ccea88efbdf6671df303ad8edce6f2d2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:46:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-08 16:54:23.000000000","tz":0},"subject":"CentOS 8: seed VM \u0026 bifrost","message":"CentOS 8: seed VM \u0026 bifrost\n\n* Change default seed VM image to CentOS 8\n* Change default bifrost deploy image to CentOS 8\n* Workaround DIB bug\n  https://bugs.launchpad.net/diskimage-builder/+bug/1866847 by setting\n  DIB_DISABLE_KERNEL_CLEANUP to 1\n* Install iptables on seed for SNAT - missing on CentOS 8\n* Fix provider network MTU lookup for empty string\n* Bump stackhpc.libvirt-host to 1.7.0 for CentOS 8 support\n* Bump stackhpc.libvirt-vm to 1.13.0 for CentOS 8 support\n* Bump jriguera.configdrive for Python 3 support\n\nChange-Id: Ie0edf6a924a914395c6502e2d5cf1139bce14a48\nStory: 2006574\nTask: 39000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68dc0cd469295dbf8becf47ce4f03bbdb9962f79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68dc0cd469295dbf8becf47ce4f03bbdb9962f79"}]},"branch":"refs/heads/master"},"5f02e38e86fb03d36cdd2b9be874f208da3bd57a":{"kind":"REWORK","_number":6,"created":"2020-04-09 14:02:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/712133/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/712133/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/712133/6"}}},"commit":{"parents":[{"commit":"35ccc0d2ccea88efbdf6671df303ad8edce6f2d2","subject":"Merge \"CentOS 8: Enable seed upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ccc0d2ccea88efbdf6671df303ad8edce6f2d2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:46:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-09 13:32:43.000000000","tz":0},"subject":"CentOS 8: seed VM \u0026 bifrost","message":"CentOS 8: seed VM \u0026 bifrost\n\n* Change default seed VM image to CentOS 8\n* Change default bifrost deploy image to CentOS 8\n* Workaround DIB bug\n  https://bugs.launchpad.net/diskimage-builder/+bug/1866847 by setting\n  DIB_DISABLE_KERNEL_CLEANUP to 1\n* Install iptables on seed for SNAT - missing on CentOS 8\n* Fix provider network MTU lookup for empty string\n* Bump stackhpc.libvirt-host to 1.7.0 for CentOS 8 support\n* Bump stackhpc.libvirt-vm to 1.13.0 for CentOS 8 support\n* Bump jriguera.configdrive for Python 3 support\n\nChange-Id: Ie0edf6a924a914395c6502e2d5cf1139bce14a48\nStory: 2006574\nTask: 39000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f02e38e86fb03d36cdd2b9be874f208da3bd57a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f02e38e86fb03d36cdd2b9be874f208da3bd57a"}]},"branch":"refs/heads/master"},"51b84b6001e90435aa76aa09a763edf8063ed085":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-04-09 14:04:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/712133/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/712133/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/712133/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/712133/7"}}},"commit":{"parents":[{"commit":"611111b8bd112124311370f7cccbd8a53b9901c6","subject":"Use unittest.mock instead of PyPI mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/611111b8bd112124311370f7cccbd8a53b9901c6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:46:50.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-09 14:04:22.000000000","tz":0},"subject":"CentOS 8: seed VM \u0026 bifrost","message":"CentOS 8: seed VM \u0026 bifrost\n\n* Change default seed VM image to CentOS 8\n* Change default bifrost deploy image to CentOS 8\n* Workaround DIB bug\n  https://bugs.launchpad.net/diskimage-builder/+bug/1866847 by setting\n  DIB_DISABLE_KERNEL_CLEANUP to 1\n* Install iptables on seed for SNAT - missing on CentOS 8\n* Fix provider network MTU lookup for empty string\n* Bump stackhpc.libvirt-host to 1.7.0 for CentOS 8 support\n* Bump stackhpc.libvirt-vm to 1.13.0 for CentOS 8 support\n* Bump jriguera.configdrive for Python 3 support\n\nChange-Id: Ie0edf6a924a914395c6502e2d5cf1139bce14a48\nStory: 2006574\nTask: 39000\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51b84b6001e90435aa76aa09a763edf8063ed085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51b84b6001e90435aa76aa09a763edf8063ed085"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
