)]}'
{"id":"openstack%2Fkayobe~712134","triplet_id":"openstack%2Fkayobe~master~I8e1d877ee91a07b86cb858d25b841f8bfcd50e21","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I8e1d877ee91a07b86cb858d25b841f8bfcd50e21","subject":"CentOS 8: separate kolla build tag from deploy tag","status":"MERGED","created":"2020-03-10 17:35:31.000000000","updated":"2020-06-08 10:03:39.000000000","submitted":"2020-06-05 21:13:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712134-1591391632005-7797b1a4","meta_rev_id":"cb99aed59c6e112f7e7cce22345517f05074ee25","_number":712134,"virtual_id_number":712134,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 21:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-05 09:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-28 07:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-05 09:00:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-04-27 16:43:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 07:36:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-05 09:00:14.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-05 21:13:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f97d3558a2ed7ccc495206990e68a551fc61cf1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-10 17:35:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a46d0580259342207844e8247c1613480ea4928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 18:52:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd4774e209e44336831968c0e4ed2f94 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32af3dd499846b7859a99010b8ecbf9 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/883d0a42626b42f992767345e691a3d8 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c185b01facf3439fbabe2aae6b742ddb : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97d4ad6ea3f7453dbf0ea5ad6654d7dc : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f36c2bbfd8444a7817f9f528c9c1cd0 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b902d6b538f74636b568f9522e3d7d5b : FAILURE in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e67acf2f34014bebbb4c7c58a3d198f0 : FAILURE in 6m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c26aa13b2c564b8d846fb1b4ae3ea2ce : FAILURE in 6m 20s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d7231fc3ac414729b3f888e8f0b2050c : FAILURE in 5m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/508532d0835d4acfb796521c60aff660 : FAILURE in 5m 19s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5d6d5bceb7874be1a0f089ca07f92e0b : FAILURE in 1h 01m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/93c6f9474de74b9bb0dde5807c7c53a3 : FAILURE in 6m 03s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/736e18d87f19466e97b6d0dc278114f9 : FAILURE in 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"02a8103a6f966e6f6bd3c3b0ceb4db7953fe68b4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-20 16:00:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"855f3dda391e9858276fde04e38a9952f59256df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 19:31:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e6385964c134653a5812b3c3d8ca0ef : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1e93d3285d4a948c45074261977842 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/416ddfe77128477080e3e6a33f84913e : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4062bbc2fe0f439ea503f60e72f69a82 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f46b2194792e4de4a6c13a19765d90f8 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c507f8f1ff47bb9cf64f9c2e3c0a36 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0edf4338fedd46b3bedd7fdc33b9774f : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f8ca75c52094df9aa6410afcabec36a : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/40cc7230af68462d99a3926f2ca3baea : SUCCESS in 8m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6db99b414701480d82d0d0980727875d : SUCCESS in 33m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/059861776a134226b10530f358c21629 : SUCCESS in 1h 04m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/28c75b2470ac4e29912b80c760265cc5 : SUCCESS in 1h 51m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/716bacf57d32473cad856d92d4170d26 : SUCCESS in 16m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f7e69ec56fce45c2ba83781f6b97d67b : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2ca86ff59a901e096924f6d82abba6cc3157fc1f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:02:28.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"0dd73807452931c78eadd8cd5d92568eccb85ea2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-27 15:58:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39637a65ea18f62a1144383a27800fc4acc9af1b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 16:42:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d1db9a270de18fca72e489566374e9cb7bb48633","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 16:43:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8d3f828a6bbfab4a8fa19820128ee21bab07590c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 16:43:27.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"9f877400ab9b464073a9d821a50ccc7f0cfdf17e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 19:36:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d53e3fb071d74ad5b5fb8eb73b49e9d7 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726c272a9fd0413fba5b6ca8ad77c3a9 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de1676ec18cc4da29c6cc2c9ed06b495 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6ed4f7a837e42d1aa0cd67402db9737 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c3a1bdf3be842118500644589a17369 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974540e63f9d423e936a2132a3fc60c9 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/449eda4964514ba4b933b447932a8adc : SUCCESS in 5m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0997b0f21142441a9b3ea940ffbeb014 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/354ce34c7dbf4de9904667a979e0c424 : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d0fa7293f494a66abf57a07dc253746 : SUCCESS in 38m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b6add55e6f234286890f7eea9ef00797 : SUCCESS in 1h 41m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a7078b24c8294921aa96137a61adcc76 : SUCCESS in 16m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5faf2d5148d944fb9546b2ee42f86f1e : SUCCESS in 2h 22m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/611e6b44728c481aa5d041f85e004132 : SUCCESS in 19m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/66f239cfa4a4464d88ac5e8f12e073db : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":3},{"id":"4176db8fe224ca9666f7287d01bc2a4028367909","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-28 07:36:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"359de863623b5ed814e7b41044b78610405255bb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-05 09:00:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6cdf64d458b57c17c8da9593c109a43b59bda520","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 09:00:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b23e6929e988c46627b3c343771c1a376fe38569","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 11:04:51.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d070d1043a84ea19a47af510d931307 : SUCCESS in 10m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cb91d6cfadf4afbbd424ea57b748997 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f419701c8b4e457fb4b7bd9cdbad5ced : SUCCESS in 10m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462ce372a1c74135a789aa9097156572 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/433117ca3c86483689b063befe4ade2f : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb3b710028d64632afda3e7e39e22e61 : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33eb9a40d26a4d6a9f7f3c9138b6c663 : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1ef221e900384bf3a6fce9104ceb4f94 : SUCCESS in 36m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/894d6c271d3547be9520bde0d8c89b8b : FAILURE in 1h 45m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5902a74899b340d3817c1e45c7f31290 : SUCCESS in 20m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c40cd8c82b5b4a1993349349b6166be2 : SUCCESS in 1h 59m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c4871e0f0f9b4103812732b4e9e57dad : SUCCESS in 20m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7fa4b1b705ab4a74a8bb82ce08473662 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":3},{"id":"2699a0f33b1b59b1c2231b71c402cb104cd5f0f5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 11:09:11.000000000","message":"Patch Set 3:\n\nrecheck\n\nMaybe an issue with a specific infra host?","accounts_in_message":[],"_revision_number":3},{"id":"68cc5c4343634a99dee3ea46e63408ea592a4589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:20:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a1cdaad82a5488bbac8cef915d0e0a1 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284d0de5b4784852b046607e0084d7c4 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbf0b5a4d1be46f38eb0bec649227eaf : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46adce84385c447284fa5662ccd8166a : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2662d6aa7a364c5cb76e15049b168248 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/889c07d4f2854492b2339dd9cfd91633 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b14fc25c990f482880e294628e4e77ba : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/230f09f6187d41f2b45ac6d096953bcc : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc3e3eaa16b341d48896150c6966930f : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd4466f0c3624da7b743fefd78238efd : SUCCESS in 39m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0c26efcfcc964f258207be824a474001 : SUCCESS in 56m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6dc1f780c6ce419ea0925935d2cf8e0d : SUCCESS in 13m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3b3f904267414b49b0965096f102af54 : SUCCESS in 1h 54m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9f1dc4466dca42f288e1e1e0a3108546 : SUCCESS in 19m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d207bfb19e644f45bb35bbffb3430089 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":3},{"id":"dedd24024cdbb705f20225c72cf9741c830600cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:20:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"28474915723b2fb9e13b3fb7425e2baf219cfa9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:54:47.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894f2c41be684b03ba206d916333b238 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0dc2d5b216446b8a09c701dad524865 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c1fea838fd849979d22ee4948af4d49 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f137774ba07b40cf82a8902aaaa460d7 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e63171b53f13435cb8119f7fa17e332f : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ebd39d644c5840eabb03decfa695213f : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/94b10d4c4a194db188ad94b08e462a63 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7dbb20b02ece437ca6b54131b845174b : SUCCESS in 33m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a605549ca3074810920887fdd796dd59 : SUCCESS in 1h 13m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8e5bd69d66b84c78b13f80009c923914 : SUCCESS in 17m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ece9ece076904d92b62aff3626b868fe : TIMED_OUT in 2h 33m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dadf0a5baf7a41c89bfe3a785a7ad5dd : SUCCESS in 22m 28s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c3db399dc1604b0581de258d96d2190a : SUCCESS in 32m 10s","accounts_in_message":[],"_revision_number":3},{"id":"eed8b86ea2ac91cb91631245db22bbc2550fbd41","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 17:11:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"44069edec158c7436547475f6de737ce174f3fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:17:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/564c997131ed4bf6a319dafc4f573d52 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b7ee5415d74407a5160a82403ccee5 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9a0369075bc4c659d2b2027aa377ed2 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81d149e7207c4eb2a460b853735a305d : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b43aff2b420456eadd81415ba95f8a7 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d87fe7039ed94d4386dea576daa473a8 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/388ca219342d461586a57be703f07432 : SUCCESS in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4553f73014744c3c89074624c020fc11 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f7e79a54865847a2931d464785aabc27 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9ee88df3427f40548d8772d8e703a4ba : SUCCESS in 34m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c7f0aae88cc14215acb2f404b2e0f1ff : SUCCESS in 1h 17m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e7b0bc0a158142cda866f9281fcda38a : SUCCESS in 19m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/01f549eb5be94a8b96b47210e7a742d0 : SUCCESS in 2h 03m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6f4a277b744e4e55852fe549dd72ad14 : SUCCESS in 17m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0157c571d60b4519ae6a8081a1de4c69 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":3},{"id":"a82be7e031e63028d466df2d6bc550abf4c7e5bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:17:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94ef97f902e8a565fd0f591a57601cae9b58e0a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:13:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf29a9ee79464a37b8d3d688a378469a : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6776de9e20c44108bade808be951e73a : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2eedaf9d9dfd4d62b58fc030351b1e97 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e98cc3eb06491f80179da884717f60 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18a854f9992f4085b5dd418e1bbf920f : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ad4bb2d7ff3e4df6a5c3cb33a29b2e97 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5de3dcffa14e4508ad60459d11a713d0 : SUCCESS in 9m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/894b6010cf834b8fa04bbc0dc5991992 : SUCCESS in 33m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fa7ada647f4540b0b4898e1f1a03e588 : SUCCESS in 1h 13m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a083bc6049d4468db670f197065468f2 : SUCCESS in 16m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3e4929b0fd274795b3354dbdc424161a : SUCCESS in 1h 51m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/91e8fd1101fb4d22a2ca3660b325e15b : SUCCESS in 17m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7bd3b081873149b288bc706c6299691b : SUCCESS in 27m 08s","accounts_in_message":[],"_revision_number":3},{"id":"75a9d782702a967bc19bb99492e37ef6d301eafa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:13:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"546fc329c78edbae2aaa79eaa533f0a4013016ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:15:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c68e4908597a45c8bc9d7c7ccffef676 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e06425dc75114726a349c015287f8011 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3},{"id":"cb99aed59c6e112f7e7cce22345517f05074ee25","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 10:03:39.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 33724452dd82e529a4852aa1bbdddafc829c12e7","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c563f45fb7e90dc202814df9f746bc909f2d0ffd","revisions":{"f03c8d6b214c92ad7dd3192e1d0b2676b761fa4d":{"kind":"REWORK","_number":1,"created":"2020-03-10 17:35:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/712134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/712134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/712134/1"}}},"commit":{"parents":[{"commit":"0cc437e26b7ddacbec248df187b62fa0ca9361f5","subject":"CentOS 8: seed VM \u0026 bifrost","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cc437e26b7ddacbec248df187b62fa0ca9361f5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 17:33:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 17:33:32.000000000","tz":0},"subject":"WIP: CentOS 8: support base tag and tag suffix for kolla","message":"WIP: CentOS 8: support base tag and tag suffix for kolla\n\nThese are required for the CentOS 8 transition. We may wish to only add\nthis on stable/train.\n\nChange-Id: I8e1d877ee91a07b86cb858d25b841f8bfcd50e21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f03c8d6b214c92ad7dd3192e1d0b2676b761fa4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f03c8d6b214c92ad7dd3192e1d0b2676b761fa4d"}]},"branch":"refs/heads/master"},"8556cc29e9863c599b79f162007002d30fcafb57":{"kind":"REWORK","_number":2,"created":"2020-04-20 16:00:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/712134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/712134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/712134/2"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 17:33:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-20 15:57:00.000000000","tz":60},"subject":"CentOS 8: separate kolla build tag from deploy tag","message":"CentOS 8: separate kolla build tag from deploy tag\n\nPrevously the container image tag applied to built images was configured\nvia \u0027kolla_openstack_release\u0027. This variable also controlled the tag\nused for deployed images. This could cause problems during the CentOS 8\ntransition, where we need to build two sets of images, and Kolla Ansible\nmay apply a \u0027-centos8\u0027 suffix to the tag we specify on CentOS 8.\n\nThis change separates the tag applied to built images into a different\nvariable - \u0027kolla_tag\u0027. The default is still \u0027kolla_openstack_release\u0027.\n\nChange-Id: I8e1d877ee91a07b86cb858d25b841f8bfcd50e21\nStory: 2006574\nTask: 39487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8556cc29e9863c599b79f162007002d30fcafb57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8556cc29e9863c599b79f162007002d30fcafb57"}]},"branch":"refs/heads/master"},"c563f45fb7e90dc202814df9f746bc909f2d0ffd":{"kind":"REWORK","_number":3,"created":"2020-04-27 16:43:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/34/712134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/712134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/712134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/712134/3"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 17:33:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-27 16:42:55.000000000","tz":60},"subject":"CentOS 8: separate kolla build tag from deploy tag","message":"CentOS 8: separate kolla build tag from deploy tag\n\nPrevously the container image tag applied to built images was configured\nvia \u0027kolla_openstack_release\u0027. This variable also controlled the tag\nused for deployed images. This could cause problems during the CentOS 8\ntransition, where we need to build two sets of images, and Kolla Ansible\nmay apply a \u0027-centos8\u0027 suffix to the tag we specify on CentOS 8.\n\nThis change separates the tag applied to built images into a different\nvariable - \u0027kolla_tag\u0027. The default is still \u0027kolla_openstack_release\u0027.\n\nChange-Id: I8e1d877ee91a07b86cb858d25b841f8bfcd50e21\nStory: 2006574\nTask: 39487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c563f45fb7e90dc202814df9f746bc909f2d0ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c563f45fb7e90dc202814df9f746bc909f2d0ffd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
