)]}'
{"id":"openstack%2Fkayobe~712443","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ieb1577f859d388a0f326cb6933c953f486dfca15","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Ieb1577f859d388a0f326cb6933c953f486dfca15","subject":"Stop gzipping logs in get-logs.sh","status":"MERGED","created":"2020-03-11 13:50:22.000000000","updated":"2020-03-19 12:08:36.000000000","submitted":"2020-03-19 12:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712443-1584619582191-b7a9cff9","meta_rev_id":"a699f2f3c13ef65f0bd6d29e25837bafe4432214","_number":712443,"virtual_id_number":712443,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-03-11 17:09:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 12:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-19 10:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-13 09:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-03-19 10:12:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 09:54:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-19 10:12:27.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-19 12:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b42088075a43f0b940333853965f174b9c109811","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:50:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2baf4f6b63a845e031b44951f2a9d495ff6b4548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 15:26:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/263f1948edfd41078e42ddb6dcf41c9c : SUCCESS in 6m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a095d6e974d4952a4ef13db8eece4b4 : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35d1fe44c90442efba497cc75172d951 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118127556a1c4e13b5e7e946c5c56ea7 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba415812a0fe43db96d98e3293aaf4ea : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c31cb19eaa47eda3d09918e8620e0b : SUCCESS in 6m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9f2aad0f18447bba3372b1c6619487c : SUCCESS in 9m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f37ec96873454d7684bb83069089fec5 : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ada8fc84f4240fb88a29280e9cba6a3 : SUCCESS in 40m 07s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/17f91a48b1c44f0b99ce4b6724e3b9f9 : FAILURE in 56m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/48e3489b43de4e9481c7dbe85414142a : FAILURE in 56m 32s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0dda823aa1f64a52bcc018152e584aa1 : SUCCESS in 23m 58s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b9960b80a5284d46b867d83dea8e4448 : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2dc528c78d0e950c3dcc140fd8ad469c115eb05e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 17:09:22.000000000","message":"Patch Set 1:\n\nrecheck\n\ntenks fixed","accounts_in_message":[],"_revision_number":1},{"id":"676d485f7ba0b57672b0c299a799cb09cecf8bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:31:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f63cc348fab74807b7054775aab487e5 : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe54cd5c1ecf40d8ad6d546777d57838 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1df45bf6af04a1e9f6a7e02fa95a750 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ab747c487f4532b0257b05e1afb378 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/638228accea04c7fabce27484b432e93 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ba3412885744c6aab596ee9d01519b : SUCCESS in 6m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d06ede4a37744af9662112038f4713d : SUCCESS in 7m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63f03be56f6c4dd890c383ce97b11399 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/727f2f63b69142738bbb584532391b8d : SUCCESS in 29m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a099d04e390a47369f5f3e6963a7917d : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d1b0d38e850b4811aef209dd7949abf4 : SUCCESS in 2h 09m 18s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3c279de3529048b8be9d0e144c1c2483 : SUCCESS in 28m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8f265ef45d6e4db6bb1872d618719eff : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fa7d5ccbd0904d4ddf1948e2eefb1ca5d90620b4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 09:54:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce7319bbfd86c0f6f39bb81ce47ca1aeb3c0313d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-19 10:12:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0402b7d454750e2a6ad27b9a7e6f16c6f6ed60a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 10:12:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf35119e2ed2b015c34656d681ed497d8bad097e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:06:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3ee27119488c482c9f37ef08d8fd99ba : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ede65af81b374378b2a026988e046ef0 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41990a774e4a48b792c36f3ba18ba1b0 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/064f9187437440dca088e867c3d32383 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5622dcacc94394b5bddd0048ac3e60 : SUCCESS in 4m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b73ae37317744f78c1c980fef164fda : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/773dbaba96024bc0bbdf35df30c91955 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c65d67d07b9b4a609e2a7c9293bf66fd : SUCCESS in 34m 46s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/309cba8fd62b4dca88da8c5b718ba81b : SUCCESS in 59m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ff2e33a818bd4e81a62cf56b5742ac30 : SUCCESS in 1h 49m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/50e4dcaeb11d4c018ed7b02b3274f087 : SUCCESS in 20m 50s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/2c522cd0d2a345b1b8827628cf9e20a2 : SUCCESS in 44m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8ec2efb99925ad6ec3d6ffdcb6eac7b607413948","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:06:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a699f2f3c13ef65f0bd6d29e25837bafe4432214","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:08:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f5cfbb8e6a64108b5b3663d5583ec56 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c74a9d0800ebb982820567eb06f125099e0ae89","revisions":{"8c74a9d0800ebb982820567eb06f125099e0ae89":{"kind":"REWORK","_number":1,"created":"2020-03-11 13:50:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/712443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/712443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/712443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/712443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/712443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/712443/1"}}},"commit":{"parents":[{"commit":"26c8abc6642af90e1436b68c4fedad27c40f7cae","subject":"Merge \"Fix use of IPA images in Bifrost without a checksum\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26c8abc6642af90e1436b68c4fedad27c40f7cae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-11 13:04:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-11 13:50:22.000000000","tz":0},"subject":"Stop gzipping logs in get-logs.sh","message":"Stop gzipping logs in get-logs.sh\n\nAs per [1] we should stop compressing the logs sent to swift in order\nto get them back readable via a browser.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-January/011875.html\n\nChange-Id: Ieb1577f859d388a0f326cb6933c953f486dfca15\nCo-Authored-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit c657ead4f25148d7f6dd5825ae11c29827ed3201)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c74a9d0800ebb982820567eb06f125099e0ae89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c74a9d0800ebb982820567eb06f125099e0ae89"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
