)]}'
{"id":"openstack%2Fkayobe~712444","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Ieb1577f859d388a0f326cb6933c953f486dfca15","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"Ieb1577f859d388a0f326cb6933c953f486dfca15","subject":"Stop gzipping logs in get-logs.sh","status":"MERGED","created":"2020-03-11 13:50:31.000000000","updated":"2020-03-19 12:03:00.000000000","submitted":"2020-03-19 12:03:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712444-1584619380923-fcd7e5da","meta_rev_id":"8c16d5280eb8d9469433accee4597ff895b7b493","_number":712444,"virtual_id_number":712444,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"date":"2020-03-11 17:09:11.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 12:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-19 10:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-13 10:16:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-13 09:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-03-19 10:12:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 09:54:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-13 10:16:50.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2020-03-19 10:12:09.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-19 12:03:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e72ddd6e018e44d419d95160d5438cd843aa53ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:50:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c2e252ef0a380ed8ba8c3ba82fb5ccba544f7f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 15:43:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/59038e1f30db40d5bbc55a51a7f36c51 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72fc9d8f0118400fa65794c900ce327d : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6978c234ade44ff39502969c5145dedb : SUCCESS in 8m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0f39e42f15464db8be5e1589fe9e9ae5 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ba4fe57b4847d595619c08ae224c29 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32bc27f169ef4d5cbc01d3966f6dfe72 : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d79a94e658d5456ca6bfdf9595911ac1 : SUCCESS in 8m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e96e959d13a4483bf3c5727dc55b66d : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3189861d6d4f4c9697f24b452bceab4b : SUCCESS in 36m 19s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d707362438c64952bbcca6ca16c7f17b : FAILURE in 52m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b59ff1dbb08b42bfad55a3feaef05da5 : FAILURE in 54m 13s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9041fca3c57248f3a92e8be62022f7cb : SUCCESS in 20m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/edff1f76bccd4458bede4255ac1fc7a2 : SUCCESS in 36m 53s","accounts_in_message":[],"_revision_number":1},{"id":"35b6b8deaa0bda6d5c14a58d4f215f29bc68d998","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 17:09:11.000000000","message":"Patch Set 1:\n\nrecheck\n\ntenks fixed","accounts_in_message":[],"_revision_number":1},{"id":"632eff7c6351f885d9f8ad9fcfa95d17a69ab29a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:17:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ad21eca013c44269a7d6fe48d8c77854 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0082cac536a2416bb6a805c0b8af8be4 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34bb63f39bbd4480a96767feec7a4e8c : SUCCESS in 6m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/82b1726e146e45e6aa8f673c8e0486dc : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3853c6563bfb4133a9a8168210a22b4d : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2f1d0dffa554f1e89a2e36fec49bfd7 : SUCCESS in 5m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f1aa4da0c0e45a3ae92763a46a796f4 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/46ec7c92387d40e0ae83bfff2e6604cb : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af736f8bd69646b4b90c36552600c5a2 : SUCCESS in 28m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0bee18df736e4f87974dd899dce2b745 : SUCCESS in 1h 10m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a1ee0b9f6d0640f78d66a23c20102a90 : SUCCESS in 1h 59m 24s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dce4579f735f4bd3a4e572f1ea9dda7b : SUCCESS in 26m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/384e8c661e7340a8ac5560ec7a780796 : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8919fade002af35425a11f9b16b78e1f351f8f5a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 09:54:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"074342f65c067c35d84f40cc46273429e6286e04","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-13 10:16:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b383b6942b76d2ce6e0dc49e65453499e8657dd8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-19 10:12:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2cb801c5016dcfb4b369b039f8617d8ca8923c43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 10:12:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"444d74eafe6b454383a7f93628e817e793bc2909","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:03:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c16d5280eb8d9469433accee4597ff895b7b493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:03:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0cfc7f3a120448a686d665e62e45ce29 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1b5273dfd341f7982478c450c345d9 : SUCCESS in 3m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/75aa444f0e1e4e49b1c70edccbab4475 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99c36fa6ef474c678ae63f8198d695fb : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5bacae502d548a5a1c3aac781072e54 : SUCCESS in 5m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ba7706f5c49843bc806cb5234fa6e93d : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/becfc168265d448cafad4c70ba753662 : SUCCESS in 8m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9298f941d33c43e8bece2c0ff2ed3a33 : SUCCESS in 31m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/98b0b3ee9f0245a2b740cbd4c8af43ef : SUCCESS in 57m 47s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c750d2ba34da41498752fce0eeed83d2 : SUCCESS in 1h 45m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8016df84f0a24d8092705e2de9e2adb9 : SUCCESS in 23m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9240a80dbfa1441c966d628728cf171c : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef232aa14e5a128bb73b5a01fe58fe07234e854c","revisions":{"ef232aa14e5a128bb73b5a01fe58fe07234e854c":{"kind":"REWORK","_number":1,"created":"2020-03-11 13:50:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/712444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/712444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/712444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/712444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/712444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/712444/1"}}},"commit":{"parents":[{"commit":"755bc38ce1d4867855dae0cdee81269a1e14edf2","subject":"Merge \"Don\u0027t list tasks for config dump\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/755bc38ce1d4867855dae0cdee81269a1e14edf2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-11 13:04:37.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-11 13:50:31.000000000","tz":0},"subject":"Stop gzipping logs in get-logs.sh","message":"Stop gzipping logs in get-logs.sh\n\nAs per [1] we should stop compressing the logs sent to swift in order\nto get them back readable via a browser.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-January/011875.html\n\nChange-Id: Ieb1577f859d388a0f326cb6933c953f486dfca15\nCo-Authored-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n(cherry picked from commit c657ead4f25148d7f6dd5825ae11c29827ed3201)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef232aa14e5a128bb73b5a01fe58fe07234e854c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef232aa14e5a128bb73b5a01fe58fe07234e854c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
