)]}'
{"id":"openstack%2Fkayobe~712694","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I6bfb0736350cbc07b1e7b98dcc6e76d3d1c8bfc7","project":"openstack/kayobe","branch":"stable/train","topic":"dnm-base-test-virtualenv-workaround-drop","hashtags":[],"change_id":"I6bfb0736350cbc07b1e7b98dcc6e76d3d1c8bfc7","subject":"DNM: use base-test for kayobe-base to test drop of bad workaround","status":"ABANDONED","created":"2020-03-12 14:22:29.000000000","updated":"2020-03-12 16:37:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e8fc5ca14d28d2e99cffc0f2e801bfb278781bd","_number":712694,"virtual_id_number":712694,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-12 16:22:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-12 16:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c08d63100dcadcbf887425bd553aa4a2c82bf02e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 14:22:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b90ab8763d331469dd7ee9d89fdc14755031973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 16:22:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/334e918aba5143c4a5825340bfca2eb7 : SUCCESS in 7m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f0d4a4189a7d4afeb37fce288bbee646 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebfee187b75d44e0a1d3d51a11ee8095 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe7d009128dc47f3aa694c024f78f1f6 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6b34c65af6e470e9152d2e865f0b5a3 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c955daf96ad642f9a206bcdb8b339362 : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6587818a33d4439bb4040bf9f36f3420 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ef7e0065baff4e218a118bcbdd5dcc75 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70f89b4ca23348c28367d4515d522303 : SUCCESS in 28m 45s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/cc503fb36af343e69a8a463f59f39b77 : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/610e8e5c7d374596a66255760fbde9c4 : SUCCESS in 1h 55m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1030e0470c8d470d96624c9c3e0a6b12 : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a14c442c023a4c22a63cfab792e646cf : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6bcf019252c344457ef80447b745d2c6f55c54ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 16:37:30.000000000","message":"Abandoned\n\nFix now merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04992f5fec708e753440fd4f03445ed6373b03cc","revisions":{"04992f5fec708e753440fd4f03445ed6373b03cc":{"kind":"REWORK","_number":1,"created":"2020-03-12 14:22:29.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/94/712694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/94/712694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/712694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/712694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/94/712694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/94/712694/1"}}},"commit":{"parents":[{"commit":"61690b7e750dfe139cea62cb3760c20b85569dac","subject":"Blacklist Ansible 2.8.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61690b7e750dfe139cea62cb3760c20b85569dac"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-12 14:21:16.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-12 14:21:16.000000000","tz":60},"subject":"DNM: use base-test for kayobe-base to test drop of bad workaround","message":"DNM: use base-test for kayobe-base to test drop of bad workaround\n\nChange-Id: I6bfb0736350cbc07b1e7b98dcc6e76d3d1c8bfc7\nsee: https://review.opendev.org/712689\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04992f5fec708e753440fd4f03445ed6373b03cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04992f5fec708e753440fd4f03445ed6373b03cc"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
