)]}'
{"id":"openstack%2Fkayobe~712723","triplet_id":"openstack%2Fkayobe~master~I0e88683f775769c1a80879685b0e7a2983599b08","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I0e88683f775769c1a80879685b0e7a2983599b08","subject":"CentOS 8: Enable overcloud upgrade job","status":"MERGED","created":"2020-03-12 15:58:00.000000000","updated":"2020-03-23 14:31:12.000000000","submitted":"2020-03-23 14:24:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712723-1584973495186-81340386","meta_rev_id":"5a7bd7742fa849c59c99741401f5b5380c1afb37","_number":712723,"virtual_id_number":712723,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 14:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-03-23 12:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-23 12:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-03-23 14:31:12.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 17:12:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-18 17:10:44.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-23 14:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e0a15382fd16475b312369b8da376fe9fbede10","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 15:58:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bc283ac6484e7a88d318aafeff98fd66f43f20d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 15:58:18.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"28439f01da56bc89f788a74974fa10f8bbcc0786","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 19:29:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b660e2560e5466351b02f5c2814e2a42963e1d64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 21:25:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26646d19e22c4825a7dcfb2697f1a497 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09b14a676e45423d854c85b45752c364 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c215b39621e645138f0f11bdcb6d4b42 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5347327806046e2b5e3f9cacd3aaa7e : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28f753702a4c427d88dda3e6edb7571e : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad45c22ac4554ad3b1cadc8075dec8f0 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/11b365da30d44b0eabd650effe795370 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/825b4ccd330b48d687872bfeda09ca43 : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f74e2a1787c49eaace7ddd9386e9bb3 : SUCCESS in 35m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0dc88a9018df47d5b787c5e240b796f9 : SUCCESS in 1h 04m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0867c1df2f704f1789375272597794d0 : SUCCESS in 1h 50m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8943acdc187145298b94c7b8595ac610 : SUCCESS in 17m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/731ad252cf744e09adaecbaadcb885e9 : FAILURE in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e1473bfc1875e7fc98e507b3099ff9ea50fa4ca9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 10:42:57.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"98b02bd507ddbfefc0cc2b5eb2b533e51e4bf526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 10:45:04.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"c30bf39bab2d83d75a5582249e9b44b8caf06494","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 10:47:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e7e54b0e28354402e3912a5a527822be74cfde1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 12:40:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e54b3b5c3e8415aacbf7e06d2814e1c : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e4f84a940c545de8779708aea184ea5 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e63e2369c1a143769c8add581ac7182d : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/313e78b2d77a49c480795d9ce1a58acc : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93a39f4a3f9047cf9e93e59361d74463 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c653e04a7d543728d06035e8ff49660 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dfd11af5593e4963b7ef3745e3190f49 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f17fdb03a1dd4fa6917e655077ebb228 : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0a2b472dc2fd492698b0c319855098ef : SUCCESS in 34m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/681520c67f7a40eb84d7dab15c4d6dcf : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/811dc850de17464eb801984a7f8891e6 : SUCCESS in 1h 49m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f80b4cbd230e4369a834b501174822d1 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":2},{"id":"073cdbfd5359b5b8ebe3d50063a3c3d0f69e5661","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 12:41:35.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"7f84d6a13ebdda7d6d440c01fe9c35b19198d5b4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 17:12:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c0fe077b8f1e1b16cdcf8129aeebea37f7709d17","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-18 17:10:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d788d7709d865791cf9f0c0dbab9289c090dad4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:52:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c800107b86e024efad2b0be5d721a9b3926b7352","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 19:20:02.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d01341e397f4b808349e08707304cba : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/975d3212bfa74530840db67027dfe778 : SUCCESS in 7m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b76540124a1947f5b43ee5ab312a43cd : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65767445f1654d1ab999c9dc72c81378 : SUCCESS in 7m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/604c2bb911524eceab13803d07a247d6 : SUCCESS in 9m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1491f9ae4b84223bfae19a4919a8c79 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9259902dada24fc89db1535e034b80be : SUCCESS in 42m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6452a77776f64b4c90fd1af745e3de60 : SUCCESS in 1h 25m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1abc0707724f4714a8d602bbcb7dea3d : FAILURE in 2h 24m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4c5b224ade6c4be68110a6600bc594ae : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":2},{"id":"74cc728fc767f175964f7c7cbed23ebeb093e988","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-21 10:19:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c308330825ce3efe1999ddf1b2e8e358651ee194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 12:23:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9d7f55342fd4ad1a7a09fa97011de9f : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c3d7f91270b42bbb2190d391b4691df : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5cde6e4e83b428da10ceca7d2eb9ebf : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd2bcb75051e4bb69cd6157598a386c1 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/404f00d080bb43719395e05171f9519c : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78bb5051f044ae7a2e777cc4efeeb42 : SUCCESS in 5m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb361b5521a544e0b65f039cddf92571 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/927adcf398ff4b7d8cf6498106cc5687 : SUCCESS in 8m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/74c989bb3b5e40d794b78cc92655de94 : SUCCESS in 34m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4e62c3a0bed340d2859a7ee056966a53 : SUCCESS in 1h 05m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/287a1362ee87450f84edcd42acd96105 : SUCCESS in 1h 59m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/72c8c45e665d47ec89ce51c444cad9a0 : SUCCESS in 17m 50s","accounts_in_message":[],"_revision_number":2},{"id":"cac7f712db220485b4d3b8e316db254e4ee64257","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 12:23:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4cd0ca06a1f61a7214c099230be697b0215a35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 16:52:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99931c54f1843d2868014b53d709a82 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e363e2a0a2d84c969b76df6fb560161e : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f728f080a144aa0927f3fc9de7a2bce : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980651c0a8024899b3ea9ae9516a28d0 : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/98eef0c37c864b668877948f4d80db9a : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/df40d6d84b8c4189b53a21725bfdcfc1 : SUCCESS in 9m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e5b67df9ad1d47bf8ccecb5eb214ab10 : SUCCESS in 34m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/70e880c246cd4c22ba21e84049c6a5c7 : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/efb6501a3aec4c30b8e63eb59a2b4e2d : FAILURE in 2h 19m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d7ddf751f61d463fb440ec9735a0be5c : SUCCESS in 18m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e9e9f984ceecaf2dd46f8bc873c55fbdb9c48e24","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-23 10:33:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84d79488440e1233f716986f9acc8b7f13334636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 12:27:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b90c2df912c48c3a1318d6ccf96e989 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e67b4b412b4ebba138ed5ae997ec52 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfd87c39ab5c4a459350df53401b7a05 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20d65af4f7da4a959e2b41d712d6321d : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/669dbc08f1974d40b8141521d4a99d88 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad27a60605cf42448aed34059e85f048 : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce7fa8728b9b4a10bca36f980b24a87b : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cbec8db1af6b443e9d00598f0d660228 : SUCCESS in 9m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c06f06f08c05454d9bb7a404410efa70 : SUCCESS in 33m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/eac3b76bfbc8411dbecc79cb3a1b4b59 : SUCCESS in 1h 05m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/182b239381094f8a9aef1202341b44b2 : SUCCESS in 1h 53m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5206ea0ecfc84940a33e3f2b4a07364b : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":3},{"id":"aa20acfc0709600bd4e6eb82ecbf5c599037371f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-23 12:39:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nApplying previous approval after minor change.","accounts_in_message":[],"_revision_number":3},{"id":"3f9d057a7070ed54b2751f5ffe83b992c6e4a5aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 12:39:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dbbbe26b36d40827556d56207bfac7fea2f7825f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:24:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52dc5c48e8e943a484b541e2891902a2 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ffa971a48044c9cbda2f3ca4b9ee95a : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb3058643ac640f89f37e3e7b8b2f624 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9860d192544c6fb7f0d826b5c6f33f : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/12d59c2023004882ae4e16f99c564f21 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e25f1c16b59447c8b2fa34b197d37f65 : SUCCESS in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41baad74c2dc4576a30e468163f1456a : SUCCESS in 26m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ce7c20d63b554e9482fd37badfecfb85 : SUCCESS in 1h 02m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7ede7e23da264e8abbb975508669af30 : SUCCESS in 1h 44m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0cf65ba97a50459cb86f75a2b4468ecf : SUCCESS in 16m 21s","accounts_in_message":[],"_revision_number":3},{"id":"4c4d3580254074749d2bc6f75e75767c913dcfa9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:24:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c27adc670c226f88209f20e1454d498996e62dea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:31:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce81034445f74f74a7cf010d1f620d50 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a28fd66b6e0eee1e79f07f8b2ecab8853958ef47","revisions":{"07a365552d3ade45b909b1ca1f2e79eadebb665d":{"kind":"REWORK","_number":1,"created":"2020-03-12 15:58:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/712723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/712723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/712723/1"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:54:33.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:54:33.000000000","tz":0},"subject":"CentOS 8: Enable upgrade jobs","message":"CentOS 8: Enable upgrade jobs\n\nDepends-On: https://review.opendev.org/711067\n\nChange-Id: I0e88683f775769c1a80879685b0e7a2983599b08\nStory: 2006574\nTask: 39047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07a365552d3ade45b909b1ca1f2e79eadebb665d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07a365552d3ade45b909b1ca1f2e79eadebb665d"}]},"branch":"refs/heads/master"},"bc094b27bf3d9c0cc12afa090f0cd52fc6376e35":{"kind":"REWORK","_number":2,"created":"2020-03-13 10:42:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/712723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/712723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/712723/2"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:54:33.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-13 10:42:53.000000000","tz":0},"subject":"CentOS 8: Enable overcloud upgrade job","message":"CentOS 8: Enable overcloud upgrade job\n\nMissing bifrost image for Train, so no seed upgrade job yet.\n\nDepends-On: https://review.opendev.org/711067\n\nChange-Id: I0e88683f775769c1a80879685b0e7a2983599b08\nStory: 2006574\nTask: 39047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc094b27bf3d9c0cc12afa090f0cd52fc6376e35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc094b27bf3d9c0cc12afa090f0cd52fc6376e35"}]},"branch":"refs/heads/master"},"a28fd66b6e0eee1e79f07f8b2ecab8853958ef47":{"kind":"REWORK","_number":3,"created":"2020-03-23 10:33:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/712723/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/712723/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/712723/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/712723/3"}}},"commit":{"parents":[{"commit":"5ee815b0e9507427b80ce94291d32640db666cfb","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs, drop CentOS 7 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee815b0e9507427b80ce94291d32640db666cfb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 15:54:33.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-23 10:32:57.000000000","tz":0},"subject":"CentOS 8: Enable overcloud upgrade job","message":"CentOS 8: Enable overcloud upgrade job\n\nMissing bifrost image for Train, so no seed upgrade job yet.\n\nDepends-On: https://review.opendev.org/711067\n\nChange-Id: I0e88683f775769c1a80879685b0e7a2983599b08\nStory: 2006574\nTask: 39047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a28fd66b6e0eee1e79f07f8b2ecab8853958ef47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a28fd66b6e0eee1e79f07f8b2ecab8853958ef47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
