)]}'
{"id":"openstack%2Fkayobe~713841","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I57fbf91ae3440d3e4e6a64cd7d05151e299c9322","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I57fbf91ae3440d3e4e6a64cd7d05151e299c9322","subject":"Fix Kayobe overcloud introspection data save","status":"MERGED","created":"2020-03-19 11:07:42.000000000","updated":"2020-03-19 20:17:45.000000000","submitted":"2020-03-19 20:16:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713841-1584648974731-140d9962","meta_rev_id":"624d08f647bf389fcdafa101e64ebd291b2e22fc","_number":713841,"virtual_id_number":713841,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-03-19 15:49:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 20:16:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-03-19 11:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-19 11:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-19 11:20:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 11:13:51.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-19 11:20:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-03-19 20:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1414f313aaf85ac50424af942100886027c456b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 11:07:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0ebaa45e5fd115518b7c920a3944be49f3f78b0f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-19 11:13:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d9b929e6c24a079137939945c247d612d2db03b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-19 11:20:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f640ad54343c3b901c858543fb264905cd4bb422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:49:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55ea3efb31f7438dbd1d1e407e819998 : SUCCESS in 4m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/03352540c919479fa416f2e702bf240f : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dab7e7cfb1e4f998d00eec4e7cf3060 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd9ebeabff346a4a19f6fefa80f2a98 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff35f15bb96d46e094133c675213f64f : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c9208feb4e847708c0b6a9a05738017 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/383f19ab17224424ba861dbd578675e6 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5509d6862e4341b2a7fa311b904d03a2 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d1abc95cf0f4e14805e531191141b37 : SUCCESS in 8m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/833d2879de2642c798f7c22035c3be3f : SUCCESS in 28m 33s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0641c19852b64771804dded3b7251dd5 : SUCCESS in 59m 11s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c300b9373a1b4111a8a4d0e195b5c07a : SUCCESS in 1h 40m 17s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8f010991495345a697d308545c4c1fbb : SUCCESS in 21m 31s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f2bcbce342294a29924e483574d5b048 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a0426040479290c0e297def3fd87200fd0ebdecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:50:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2528215453258175cd1956919f3f30947d07d62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:53:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a78a76e3fd2049dd85513c18b4cb9312 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3200fd84f86247f3928c60b533407c6d : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b86731c7db4beb8540d1d3ae688990 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d6a558deb644ea88d605a02ef1be0fc : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10afc0ef85e4453fa32642261a64a056 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1b677aef4c24055877d2bee86937f56 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab37cd7f2d1447da9d1b17d0efd69ac6 : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6d171032c3e41e3aaaa2eddcecfc179 : SUCCESS in 13m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7902624b7bb84884bc00feae7b223e1f : SUCCESS in 33m 05s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/402f77d8db88412ab4d1fb17aa4fba4b : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/03dffce0e4c746f7b88642815da0fce7 : FAILURE in 52m 02s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/37757d1424d749b0b343dd75236c686e : SUCCESS in 23m 27s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3941ac5b8c704f2687d3579edae0dfcf : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"15d561e301ed71303cb88d999870cf2529845a0d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 15:49:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2cc6c1db59660f70ecbe7b293f75123c30b15054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:27:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76368dbf1d0c4f39a46ba2d6f021100c : SUCCESS in 6m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6bf0b34c1d4e48abbd80c9c6726467b0 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca67584253504113a1c6f6ba40bc9ebf : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/689b6f9bb3ed49618944f2c870857fea : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b79d18e918c40f0946a6a60d886da06 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ce343e300543abb342df07532b842b : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ab3917ef8d042f4ab36447f09dd0c94 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/251ec8e12696412b8f987a862b466d75 : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91a5563d656044dfb47f16b26db2527f : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1c3e48e0b92491b8aa06f9fcf134766 : SUCCESS in 35m 01s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/775ce226f1964ae685fb0701203d1571 : SUCCESS in 1h 02m 26s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ffc8e111a17f4cef880c34733197dfe8 : SUCCESS in 2h 20m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1c2bb32004fb485d8116796ce979cd6b : SUCCESS in 25m 05s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e00994c951b0494e9f2e583dad7334cc : SUCCESS in 33m 47s","accounts_in_message":[],"_revision_number":1},{"id":"12067b8b26725cfe1385be7f1cdecec72ed1ec90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 18:27:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"251e234d0023577c210c1f9391677d99934be77e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:16:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7fff9230c3d94586815b9eb306679432 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d92978a2989c41d6b7c9bfed780a69a2 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87fce9413cf4981add06d3bc3f09cff : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b07e6fb37773497b9a80aeea8551c257 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9b1f0e88ec4a48b2694b0fdeab2786 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4770e4d8fdb8496f94e4965b806cb368 : SUCCESS in 4m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b54d67ff92194016bbebbc0beca15f82 : SUCCESS in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56f4a25cdd384cb28473eb0fb3263895 : SUCCESS in 9m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f44570d706ce491bbe93eeff17813cbf : SUCCESS in 35m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6f991a563fde46329ee06c53fd10132e : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a9e43d8df5984d8ab74a46974ad9706c : SUCCESS in 1h 43m 16s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9b17c4588a104e208c89b17d121d6eb8 : SUCCESS in 24m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ac494fc58fa647fe96412213853c7d74 : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":1},{"id":"31daffbc3764342d30bf2310d0c678a7cd4abe4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:16:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"624d08f647bf389fcdafa101e64ebd291b2e22fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:17:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657890db3cb04adfbda9ca569ce5b039 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abe4a97cc43c477ca326850189816e47 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2b5b4b2b3b1ccb48d992c19aa731bbb950a7678","revisions":{"a2b5b4b2b3b1ccb48d992c19aa731bbb950a7678":{"kind":"REWORK","_number":1,"created":"2020-03-19 11:07:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/41/713841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/713841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/713841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/713841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/713841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/713841/1"}}},"commit":{"parents":[{"commit":"26c8abc6642af90e1436b68c4fedad27c40f7cae","subject":"Merge \"Fix use of IPA images in Bifrost without a checksum\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26c8abc6642af90e1436b68c4fedad27c40f7cae"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-21 15:43:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 11:07:42.000000000","tz":0},"subject":"Fix Kayobe overcloud introspection data save","message":"Fix Kayobe overcloud introspection data save\n\nKayobe overcloud introspection data save fails because the dynamic\ninventory script siliently breaks causing Ansible to parse it as a\nstatic inventory file. The failure occurs due to OS_TOKEN being set.\nThis change works around setting OS_TOKEN before running the dynamic\ninventory script by setting OS_CLOUD before querying inspector.\n\nConfirmed on Stein and Train, and verified in both environments.\n\nStory: 2007326\nTask: 38846\nChange-Id: I57fbf91ae3440d3e4e6a64cd7d05151e299c9322\n(cherry picked from commit 167e2ff3db19f32544e324bd02f4586e36486650)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2b5b4b2b3b1ccb48d992c19aa731bbb950a7678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2b5b4b2b3b1ccb48d992c19aa731bbb950a7678"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
