)]}'
{"id":"openstack%2Fkayobe~714963","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I57d67c51572315d20475b10e8853bb76b5f44a65","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I57d67c51572315d20475b10e8853bb76b5f44a65","subject":"CentOS 8: Enable seed job","status":"MERGED","created":"2020-03-25 14:55:21.000000000","updated":"2020-04-09 08:13:22.000000000","submitted":"2020-04-09 08:07:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714963-1586419658220-441aa25d","meta_rev_id":"2a4bdf8c5aa6e3e0746a47e926bf9dd8f2c679ee","_number":714963,"virtual_id_number":714963,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 08:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-04-08 20:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-04-08 20:08:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-08 20:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-09 05:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"date":"2020-04-08 20:08:32.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":1,"date":"2020-04-08 20:08:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-27 14:41:29.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2020-03-27 20:04:14.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-03-27 21:28:34.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-04-09 05:51:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-09 08:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"152a698c5d28640c1681aa80b22e4f06f6356370","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-25 14:55:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"902256fcd71bd9ecd761c1d26a6e2c87bc9f89f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 16:41:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fee3bec920b4fb58a99935ba1121e3c : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/24cbfe87b9c64930b3f5fa46c2596703 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed54c3acc9574ef7b28aaec11c5f7023 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bc9ada04ef455db1efaa47c2d495bc : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/967001873b7443ebbc105c1ab0234a00 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f902d9ecf51044259ac88c7fd2543d0d : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/96f31c6ad6de43ab836631579353b79e : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/710edc770097497b828b36c7854a36c1 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12e349664adc4451be3dfd25eef6ccd6 : SUCCESS in 37m 03s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/f3f9be60c2dc4d6bb8ccfecf92508873 : SUCCESS in 1h 17m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ff987efe93704bd68479f4caf924269a : SUCCESS in 1h 11m 38s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3bca728c55304fbb8271727d37c88ac2 : SUCCESS in 1h 40m 54s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/f2e3b1fbd778414dba2b46e5a18cf692 : SUCCESS in 21m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/24faf27b336c46dfa9204059343f8cd9 : SUCCESS in 20m 06s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/889d68ed198845fd848d0c8083aa3cd7 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d179a60b88e425b036e9444f6a3619ddd6cb8166","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-25 17:11:10.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"796876ff9aad47f5a5f3534008815978e8b6c52a","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-27 14:41:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"866627c33041e52b3b31edf723c8d1a79c9efa5b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-03-27 20:04:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3df2757bd22f616e1e7dd211aa713ba3b55519fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 20:04:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b67b441d5026b61baaa454b59f69f2a0dbe7205","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 21:10:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/12bc174309d54b9da0681142c723c9ba : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba070b128434629bd229876033351e3 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d3fdafc985345df80af229d0f0ec804 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54ff2c23ffe4481eaff9fa1b106b6866 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb977ebacd3c4ccf9b34fe51ff2adc4d : SUCCESS in 7m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/333bbe80342b4edea9a60a8c84e5f301 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/85a449711e55480a8f1a8eee31a3047f : SUCCESS in 10m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/caaeb3a2dfb8411c8ced79443a494a38 : SUCCESS in 30m 52s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/f099f40e6b104aacbff563d16efca3f1 : FAILURE in 1h 04m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ce65a23b882d41c38931a6ed957a69ba : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f8113d985ddf4b6d8ab47093123b15cc : FAILURE in 53m 54s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/60bdf7c0939945808ca04134ec3c8f2e : SUCCESS in 23m 48s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3ddc1de9679344579038f632c150d636 : SUCCESS in 20m 36s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2614df3739a947a181ca103e1c2ed19d : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e759894ab0abd06b8cd1cdb920a1a4e92d69a0a4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-03-27 21:28:34.000000000","message":"Patch Set 1:\n\nReply looks like a python2 -\u003e python3 bug \n\nimport Queue -\u003e import queue","accounts_in_message":[],"_revision_number":1},{"id":"2e528472bf965cb3ff24e05776e2424214ffff52","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 14:21:32.000000000","message":"Patch Set 1:\n\n\u003e Reply looks like a python2 -\u003e python3 bug\n \u003e \n \u003e import Queue -\u003e import queue\n\nFix proposed: https://review.opendev.org/#/c/717008/","accounts_in_message":[],"_revision_number":1},{"id":"4d6a36c6e1be30ac635020bccc85ad254b8519ef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 14:24:43.000000000","message":"Patch Set 1:\n\n\u003e \u003e Reply looks like a python2 -\u003e python3 bug\n \u003e \u003e\n \u003e \u003e import Queue -\u003e import queue\n \u003e \n \u003e Fix proposed: https://review.opendev.org/#/c/717008/\n\nFrom logs it seems we hit this error in Kayobe, which I am hoping https://review.opendev.org/#/c/715880/ would fix (needs to be backported to Train).","accounts_in_message":[],"_revision_number":1},{"id":"03a4cc454c4e6c6e685338748e83e3a240d7586b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 14:41:08.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e Reply looks like a python2 -\u003e python3 bug\n \u003e \u003e \u003e\n \u003e \u003e \u003e import Queue -\u003e import queue\n \u003e \u003e\n \u003e \u003e Fix proposed: https://review.opendev.org/#/c/717008/\n \u003e \n \u003e From logs it seems we hit this error in Kayobe, which I am hoping\n \u003e https://review.opendev.org/#/c/715880/ would fix (needs to be\n \u003e backported to Train).\n\nYou\u0027re right. The fix I linked is for Stein \u0026 earlier.","accounts_in_message":[],"_revision_number":1},{"id":"be41c3074b7b4b7cf66ce87015311d25867965d6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-08 20:08:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"50b6fd9e1b999acdd59f415c858fdbb3b3c44827","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-08 20:40:17.000000000","message":"Patch Set 2: Code-Review+2\n\nHolding on W+1 until jobs succeed","accounts_in_message":[],"_revision_number":2},{"id":"40d437fc206e57bc389a1c9d691914187db8f9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 22:18:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe74ac9b39674b4a8849ebb9bf3152c5 : SUCCESS in 7m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e4e6231038640a091c1edc17e1c3bfe : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a06ea938e3247be8941860b9e121eb1 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a61baf9cc38f4af6bea51f6d63bcbdb6 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f3892623126425197a6fe90755cd2df : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765171cabe07403b8d59e6bba67c95e3 : SUCCESS in 12m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb7d02e1b4084a9e8482800462fc981b : SUCCESS in 7m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c93667a45bbd4c32be4369d4c1787d9a : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/490ab28c5d7145c0ae686a949ea2f1c1 : SUCCESS in 29m 58s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/5193dbc197ba46478cb3d03996ddfe5b : SUCCESS in 1h 09m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8083d4e2d1d34fc580afb01af011d744 : SUCCESS in 1h 11m 13s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3d20e529305b467bb306c076a57161d8 : SUCCESS in 1h 46m 58s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/05ff6183269f4623ae48ee86c4705fd0 : SUCCESS in 21m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/680be26622be424895c777e978ea9dbc : SUCCESS in 19m 14s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/be793005a6b04fb5a276c1898b9821fe : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b94470d111f78ce57bca26e0261c5facd420a8fd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-09 05:51:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"53e26750db7dc31719ffb82403b367ecad18a448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 05:51:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dce292188e75a31dde141016a01bd2eb381cd633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:07:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6cfa7c6f590845269fdf523745e0f35b : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34d2c59cd7be4470985cb2b85e79287f : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e111682ac3948cebc2025a32d31dc59 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f658eaef14494738906bccf3972c41f8 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ce7f129d214d28bbedbaf26a4a405a : SUCCESS in 8m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/645ed9d95c774f039ec9f500b4a6c722 : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/639d8041d26a49a1bb32f2a3a0005323 : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef7a4f68e7b34b45874f24ddf9fa1bca : SUCCESS in 29m 49s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/431636f5c6a2484c8c542b24c22a8d98 : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1e3cdb36c06546f9be7c4b34aadd9e76 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/42c1cdaf743c46a49fe77bc7ffb4ec02 : SUCCESS in 2h 05m 52s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/7b3445ec6ffc449f86bd7c051a89b2e9 : SUCCESS in 21m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e35e60830c4d4b5a8a47352c20ea9943 : SUCCESS in 22m 13s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0778d4b7f5704827b4cfbfa5231a2517 : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":2},{"id":"84f767149c86c7d12f282f35649930256f9f4015","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:07:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2a4bdf8c5aa6e3e0746a47e926bf9dd8f2c679ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:13:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825b9597885147f3a52eb9319692ecfc : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de8f5592e9840f0bab71a09b83f1d5b0271a9345","revisions":{"14a06d4533c75d880c29a28aa5349419ce6a2df8":{"kind":"REWORK","_number":1,"created":"2020-03-25 14:55:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/63/714963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/714963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/714963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/714963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/714963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/714963/1"}}},"commit":{"parents":[{"commit":"f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-25 14:54:54.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-25 14:55:07.000000000","tz":0},"subject":"CentOS 8: Enable seed job","message":"CentOS 8: Enable seed job\n\nThe Kolla CentOS 8 Bifrost image is now available for Train.\n\nChange-Id: I57d67c51572315d20475b10e8853bb76b5f44a65\nStory: 2006574\nTask: 38870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14a06d4533c75d880c29a28aa5349419ce6a2df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14a06d4533c75d880c29a28aa5349419ce6a2df8"}]},"branch":"refs/heads/stable/train"},"de8f5592e9840f0bab71a09b83f1d5b0271a9345":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-08 20:08:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/714963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/714963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/714963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/714963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/714963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/714963/2"}}},"commit":{"parents":[{"commit":"6255b3829771826d6ad66942ed37d8064404b040","subject":"Fix multiple CI failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6255b3829771826d6ad66942ed37d8064404b040"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-25 14:54:54.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 20:08:32.000000000","tz":0},"subject":"CentOS 8: Enable seed job","message":"CentOS 8: Enable seed job\n\nThe Kolla CentOS 8 Bifrost image is now available for Train.\n\nChange-Id: I57d67c51572315d20475b10e8853bb76b5f44a65\nStory: 2006574\nTask: 38870\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de8f5592e9840f0bab71a09b83f1d5b0271a9345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de8f5592e9840f0bab71a09b83f1d5b0271a9345"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
