)]}'
{"id":"openstack%2Fkayobe~715880","triplet_id":"openstack%2Fkayobe~master~Ie44508fe3d3b08190afa5a43748e43548a63ff82","project":"openstack/kayobe","branch":"master","topic":"osc-upper-constraints","hashtags":[],"change_id":"Ie44508fe3d3b08190afa5a43748e43548a63ff82","subject":"Install python-openstackclient using upper constraints","status":"MERGED","created":"2020-03-30 08:53:53.000000000","updated":"2020-04-02 20:10:00.000000000","submitted":"2020-04-02 20:07:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715880-1585858072810-87f092a5","meta_rev_id":"69a49abc5500504a92a55b3c3b65b0801240678b","_number":715880,"virtual_id_number":715880,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 20:07:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-04-02 16:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-04-02 16:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-04-02 17:54:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 16:00:00.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-04-02 17:54:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-02 20:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"453a0a1d480557f7cdc811ed80786784cbf36569","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-30 08:53:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"210c2f5f32561a73df90c8b5a9209f2c1f5f4986","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-30 09:06:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6cc472499b00dbbbf43ea9e5a0be0907de756c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-30 09:06:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5129945c36e22bbad882d3cd5111f00445e5f09c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-03-30 10:23:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2cb41604b6af9fee3e23a46bf410411358f2a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 11:11:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/910c47a23a55492fa3a8e72e9392a556 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0a1b2b6d8494c2493d0d48c6a6c1334 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ecbed772a0954250ab336997046f7f18 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f104e910793d43c3a653708b7fe238e9 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c577935298a415487409657027e5c2a : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc6ad2cce4845ff875808cb3f6a714a : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d9172c0182c4a3eaa5b8dfbf46c8692 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/203e68020b4f4d25b55da47e65f0ba31 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22748dce13174c498738a5066df4990d : SUCCESS in 33m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9aa9f0a9897f466788bbf52d50ffe2d8 : SUCCESS in 1h 17m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4a84822d16ec43dba823dee4473e828c : SUCCESS in 2h 11m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/89db305c5d284013bf9d812c830345ae : SUCCESS in 18m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a273a7d75bc948a596cc2aafd0ac958f : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":1},{"id":"35097ccc3ca36f83b5ce3284c0858c26dadfa89a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 13:27:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2399bd327727a625439cda7d4968996c9212a32a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 13:29:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec68f64effabca61490c0808ad3959ac1a965237","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 13:31:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60e599e76e54f03993727da6d48dc62518cd8b0d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 14:40:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8e18a16a77f2fdfc681a41392270e80b926f055e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 14:56:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1e78124b64254ed798c1ca174fdee06052a7c922","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 14:59:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c9f47ec659c1df98900d16582cc6eb3d2ad2ff1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-02 16:00:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0848912dcd459a615a1058ce764eb770d4c075c6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 16:04:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"954623fbf606a519ee44c6f5b631d585f4b62fd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:09:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c0d50c9b0314af6a0243797c9e57a2e : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d556ea8f79ee4315a934017e279e0621 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b3aa35188c94c72aae7d01228353d59 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc92d0c3e7c9448ab16161f11a205cea : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37a0ec5bd2da495cad368e403d01ea5f : FAILURE in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499b5a19cc5e409995003532c51bf992 : SUCCESS in 13m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c7ad0f38bcf4355b64512514e9cb621 : SUCCESS in 9m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8de7d43e28e54ff19f6460e69b90dd90 : SUCCESS in 13m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/647e8f855c5c4382a358706cab9fec65 : SUCCESS in 34m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e3d089741d23420292e3f0c2998a12c2 : SUCCESS in 1h 08m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5dc2e16742ca4a53855fcd787c05452e : SUCCESS in 2h 02m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3c645d67b1a940acaa423e2dd9558da8 : SUCCESS in 23m 15s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3ab93710b35747cf82d43db356ec6c59 : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":5},{"id":"b7717468a00ca7549aaaa623d41f84fe233f0820","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 17:54:33.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"afbb2d660421f5a86cf00c3e3301f96ffa11a928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:06:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"31eebcd7ef424821281a88dc3bbe2b9e83ff2ac3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 20:07:52.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b5262c8cc54c91b3fe0da43709c4f5 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a931ec833f204564a7a88bdf9b2acbb7 : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/83a7e87930154b62b442e89695f7faa4 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1fc03c89002489fa999ae0d2f5cdf44 : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/24e8f6ab7b6d42afa560a4e942c4589a : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f5745ec96d474b55b0bfcec1b8f93385 : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f9d5e32c9bba4cdb851baea1f1df1a0a : SUCCESS in 35m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/91964d14fe754a589840d5ae7716ddf0 : SUCCESS in 1h 08m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/57ff35238bee4ac0bfcc5d09cf86ec92 : SUCCESS in 1h 54m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e8904330dda24d9d922f6c117b2eb127 : SUCCESS in 19m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/418abbcdb48e4334b5c89f3867bf576e : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":5},{"id":"cc632f9ac4296c574551afaa9342788811c484c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 20:07:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"caed8f9079d668a4b597529dae8b82f8af576f0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 20:10:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a349a8746b974fb18cb6869ec6c28a75 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ef33e6ecb770f852bb4433d6b0d2dec7778937a4","revisions":{"735b1e9c103193b86f9626d023485416f986382b":{"kind":"REWORK","_number":1,"created":"2020-03-30 08:53:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/715880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/715880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/715880/1"}}},"commit":{"parents":[{"commit":"35ccc0d2ccea88efbdf6671df303ad8edce6f2d2","subject":"Merge \"CentOS 8: Enable seed upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ccc0d2ccea88efbdf6671df303ad8edce6f2d2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-03-30 08:53:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-03-30 08:53:34.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nUse upper constraints to guarantee compatibility of\npython-openstackclient with the Python version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/735b1e9c103193b86f9626d023485416f986382b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/735b1e9c103193b86f9626d023485416f986382b"}]},"branch":"refs/heads/master"},"49422836f74c0e2bd7ee92d049e6a86150ffcad7":{"kind":"REWORK","_number":2,"created":"2020-04-02 13:27:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/715880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/715880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/715880/2"}}},"commit":{"parents":[{"commit":"35ccc0d2ccea88efbdf6671df303ad8edce6f2d2","subject":"Merge \"CentOS 8: Enable seed upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ccc0d2ccea88efbdf6671df303ad8edce6f2d2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 13:27:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 13:27:15.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49422836f74c0e2bd7ee92d049e6a86150ffcad7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49422836f74c0e2bd7ee92d049e6a86150ffcad7"}]},"branch":"refs/heads/master"},"bbf7f150b0663880eed6c61b5ed08f3249b7f4b2":{"kind":"REWORK","_number":3,"created":"2020-04-02 13:31:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/715880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/715880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/715880/3"}}},"commit":{"parents":[{"commit":"35ccc0d2ccea88efbdf6671df303ad8edce6f2d2","subject":"Merge \"CentOS 8: Enable seed upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ccc0d2ccea88efbdf6671df303ad8edce6f2d2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 13:31:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 13:31:08.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbf7f150b0663880eed6c61b5ed08f3249b7f4b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbf7f150b0663880eed6c61b5ed08f3249b7f4b2"}]},"branch":"refs/heads/master"},"cf984c3bb620744dbecdcde776fa5e7ed812cf49":{"kind":"REWORK","_number":4,"created":"2020-04-02 14:56:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/715880/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/715880/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/715880/4"}}},"commit":{"parents":[{"commit":"35ccc0d2ccea88efbdf6671df303ad8edce6f2d2","subject":"Merge \"CentOS 8: Enable seed upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ccc0d2ccea88efbdf6671df303ad8edce6f2d2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:55:30.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:55:30.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf984c3bb620744dbecdcde776fa5e7ed812cf49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf984c3bb620744dbecdcde776fa5e7ed812cf49"}]},"branch":"refs/heads/master"},"ef33e6ecb770f852bb4433d6b0d2dec7778937a4":{"kind":"REWORK","_number":5,"created":"2020-04-02 14:59:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/715880/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/715880/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/715880/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/715880/5"}}},"commit":{"parents":[{"commit":"35ccc0d2ccea88efbdf6671df303ad8edce6f2d2","subject":"Merge \"CentOS 8: Enable seed upgrade job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35ccc0d2ccea88efbdf6671df303ad8edce6f2d2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:58:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:58:50.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef33e6ecb770f852bb4433d6b0d2dec7778937a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef33e6ecb770f852bb4433d6b0d2dec7778937a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
