)]}'
{"id":"openstack%2Fkayobe~717031","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ie44508fe3d3b08190afa5a43748e43548a63ff82","project":"openstack/kayobe","branch":"stable/train","topic":"osc-upper-constraints-train","hashtags":[],"change_id":"Ie44508fe3d3b08190afa5a43748e43548a63ff82","subject":"Fix multiple CI failures","status":"MERGED","created":"2020-04-02 14:28:43.000000000","updated":"2020-04-08 20:34:00.000000000","submitted":"2020-04-08 20:23:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717031-1586377399183-b33c4e60","meta_rev_id":"3271c4d054b5cdd1872a3cfd8625e70c97b1f1af","_number":717031,"virtual_id_number":717031,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 20:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-08 16:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-08 18:18:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 18:18:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-08 20:23:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a59fcc41319e753c27c6c7cdc104981980cc3448","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 14:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19f8eefa5a4fb6f9dc6f867c2ac2f777a47424b4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 14:59:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cca4bbcc302571ac32d97b25e8bf57cdda50690e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:49:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33282ac81b684719804bb34604f1024d : SUCCESS in 9m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b48fd41a05e64bbe952cfd9a17939f00 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e281646c66345b28e26ba5f49435722 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135e9b307d51419dba5263ce36623b08 : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8eb737b9b56f42ccbbb5c133bccdfd2d : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b0f653a0fa4df8b185667023589114 : SUCCESS in 15m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f529eed58e4742638f44d3c4cad49023 : SUCCESS in 7m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a55f7cb24054f1a9cfa2d8f21775adf : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/58e64c18bef249b494b2499d3b7bc102 : SUCCESS in 37m 38s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/d8694e90a0fc4184899faeb628ac3095 : SUCCESS in 1h 20m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c405f6cf24cc45f3add3dd4258a7fac8 : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/40f8c408659f4821b103cd7033d1deb0 : FAILURE in 56m 39s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/f414cd2cb9184b23ba0579a9a616ec2a : SUCCESS in 30m 16s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/bfa1f5d0abe4411fb3039b1fdc29ad4a : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":2},{"id":"7ce038276aac6d48a1a2828e28c98fc63db8d995","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-02 17:19:33.000000000","message":"Patch Set 2:\n\nIt fails in the overcloud job, we might need to apply to earlier branches first?","accounts_in_message":[],"_revision_number":2},{"id":"7b17ec010bafbf2c0f0225b43f1d652cbfca2f09","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 18:10:27.000000000","message":"Patch Set 2:\n\n\u003e It fails in the overcloud job, we might need to apply to earlier\n \u003e branches first?\n\nSeems so. I suppose we are using python 3 on master so didn\u0027t hit it.","accounts_in_message":[],"_revision_number":2},{"id":"1d974eba9c38796b629094341ea16d06941d4daa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-08 12:45:28.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1e476e9f70795212e749550c8d1fbf3d88d0f6a6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 13:11:29.000000000","message":"Patch Set 3: Code-Review-1\n\nWe\u0027ll need tenks 1.x here too for CentOS 7","accounts_in_message":[],"_revision_number":3},{"id":"c6dd22e512cfbaf97566d04d3706eb1b5806f07c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-08 13:17:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e5d8ef7c388520d66780d56575146acfd5a45165","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 13:28:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9fad24ea7f2983d885c62ca1affa0ce2fef62fa3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-08 15:08:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"27af27e7524f53808338efd2cca0eff3e11853bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 16:17:49.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"77262cf54d2297f5576f9d41e8b08f58f2d87b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:36:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd35b12d6ec447a4b252b1f18e78514a : SUCCESS in 9m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3f947f6988ca45ce8b13980874fc6b68 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2c5a8aba8b84a20ad6997f802321fa0 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/245794da565146288a3865565217b0c5 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f0d342e0882418ab432c1947c79c802 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2f33a893a74b76a79587ece84888a0 : SUCCESS in 7m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b57cad92938408a851369b993006cd7 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/414b238a49e94343ab82579c5218903d : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bd555ffbbe442c39958fd1825c07f7a : SUCCESS in 35m 15s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/60ea7585a44045028c76a0abbbc110f7 : SUCCESS in 1h 16m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/43c1f84551d0400cbec7b14ad900a1af : SUCCESS in 1h 34m 29s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/68b34e6839c24c578a544c2484c12116 : SUCCESS in 2h 16m 26s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/c1299632f3284c099ad3eb4ac4dafa7e : SUCCESS in 23m 51s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1edc940a3a4c4b63bea9114605835ad5 : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":5},{"id":"bed7146f9d69a1c7aebce4f2227df143b790d8de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 18:18:24.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"68fd8cb02d7d840b02fe596adb20ac20121544a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 18:19:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"20422697da0c04fd624f12ddfffbda535ff45b67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:23:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b0c4b2d4761b4e6287aa943ac0de8c35 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4a8cec7a5124e6e846bb2a0fb5a3bac : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed3772b69d0411d98c6f595da19be57 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff8777acef8b49d6af34afb2a15764ed : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e80d7c5aba42c7a4c9c3817c008250 : SUCCESS in 8m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4422bba9d1894d1587eaccf07a8bac74 : SUCCESS in 6m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb483001caa9457f82476278d057844c : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/55cb944bc21e480a8f36f9f82de5d093 : SUCCESS in 28m 29s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e50e3eb2e91c43389861341a493bde7f : SUCCESS in 1h 11m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5205d35617fb47e88d7a88054bed6eaf : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/87316aa96b2b416db1375e7a26ececae : SUCCESS in 1h 56m 19s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/dc72e4026a994cfca6d27de02b6bd7ed : SUCCESS in 23m 11s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1ca67a864a294ce9ace1fcbc9efb6444 : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":5},{"id":"997dfe4703f8c3a28ee1f242e707f3100b60662d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:23:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"e3db65b1379621c6cd9fbb7f2ef19578ef80a15b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:34:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202aac0d25324e1987d5129866689881 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6255b3829771826d6ad66942ed37d8064404b040","revisions":{"9adf2ee1bf05b0d6e67f824f2deb33dd476e9f50":{"kind":"REWORK","_number":1,"created":"2020-04-02 14:28:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/717031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/717031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/717031/1"}}},"commit":{"parents":[{"commit":"f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 13:31:08.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:28:32.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\n(cherry picked from commit bbf7f150b0663880eed6c61b5ed08f3249b7f4b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9adf2ee1bf05b0d6e67f824f2deb33dd476e9f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9adf2ee1bf05b0d6e67f824f2deb33dd476e9f50"}]},"branch":"refs/heads/stable/train"},"ed4e789329ebca385fdecdd741eaa36d38d707d6":{"kind":"REWORK","_number":2,"created":"2020-04-02 14:59:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/717031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/717031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/717031/2"}}},"commit":{"parents":[{"commit":"f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:58:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:59:01.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\n(cherry picked from commit ef33e6ecb770f852bb4433d6b0d2dec7778937a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed4e789329ebca385fdecdd741eaa36d38d707d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed4e789329ebca385fdecdd741eaa36d38d707d6"}]},"branch":"refs/heads/stable/train"},"f2da5f46dafed845713cf1d9b6b449023f6fedb3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-08 12:45:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/717031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/717031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/717031/3"}}},"commit":{"parents":[{"commit":"f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 12:45:03.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 12:45:03.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\nDepends-On: https://review.opendev.org/718424\n(cherry picked from commit ef33e6ecb770f852bb4433d6b0d2dec7778937a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2da5f46dafed845713cf1d9b6b449023f6fedb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2da5f46dafed845713cf1d9b6b449023f6fedb3"}]},"branch":"refs/heads/stable/train"},"e3f4996684657968986fa3e49eb8c13f39294ddf":{"kind":"REWORK","_number":4,"created":"2020-04-08 13:17:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/717031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/717031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/717031/4"}}},"commit":{"parents":[{"commit":"f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 13:16:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 13:16:34.000000000","tz":120},"subject":"Fix multiple CI failures","message":"Fix multiple CI failures\n\n1. Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\n(cherry picked from commit ef33e6ecb770f852bb4433d6b0d2dec7778937a4)\n\n2. Use tenks stable/1.0 for Python 2 support\n\nTenks dropped Python 2 compatibility in its master branch.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\nDepends-On: https://review.opendev.org/718424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3f4996684657968986fa3e49eb8c13f39294ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3f4996684657968986fa3e49eb8c13f39294ddf"}]},"branch":"refs/heads/stable/train"},"6255b3829771826d6ad66942ed37d8064404b040":{"kind":"REWORK","_number":5,"created":"2020-04-08 15:08:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/717031/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/717031/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/717031/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/717031/5"}}},"commit":{"parents":[{"commit":"f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf","subject":"Merge \"CentOS 8: Add seed and overcloud CI jobs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f4cbdaf4d6c8ea52aa2305fe0dc5886b3e0c13bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 15:07:59.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 15:07:59.000000000","tz":120},"subject":"Fix multiple CI failures","message":"Fix multiple CI failures\n\n1. Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\n(cherry picked from commit ef33e6ecb770f852bb4433d6b0d2dec7778937a4)\n\n2. Use tenks stable/1.0 for Python 2 support\n\nTenks dropped Python 2 compatibility in its master branch. Use Tenks\nstable/1.0 for CentOS 7 jobs.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\nDepends-On: https://review.opendev.org/718424\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6255b3829771826d6ad66942ed37d8064404b040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6255b3829771826d6ad66942ed37d8064404b040"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
