)]}'
{"id":"openstack%2Fkayobe~717082","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I1f09f7c11767226e89b34687dab1553e87be76ba","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I1f09f7c11767226e89b34687dab1553e87be76ba","subject":"Switch to stackhpc fork of resmo.ntp","status":"MERGED","created":"2020-04-02 17:29:08.000000000","updated":"2020-04-09 09:28:15.000000000","submitted":"2020-04-09 09:25:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717082-1586424304727-441aa25d","meta_rev_id":"0461e1349657897e9a9f27be306d50256548ba11","_number":717082,"virtual_id_number":717082,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 09:25:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-08 16:59:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-08 20:04:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-08 20:04:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 16:59:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-08 20:04:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-09 09:25:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66cc92cb5ad89b821f7b4a0e2ebcef2bf774e8bc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-02 17:29:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a81f02900cb815e984cf1ff2bba638ab3bf65973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 19:34:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b052887460b4f56b9b4d592def58a08 : SUCCESS in 7m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/92e58af440fb421cb1f470a67c15ccd7 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/095919a6f2c24c308ac1288d248bf60a : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/755f7aef0bb445b79c399ba8d372febf : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4eb19f14a1e447f8a5b898334541c1b2 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3135651bc54910b53deeee09aef795 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/765943f3e4fc4fd4a747b9a58a475e59 : SUCCESS in 6m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/20e3551cf26f40a293c68ab84bd3a8e1 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f6c69bbf0e904c8083cf51915f93fa99 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d4bd4f385644996815f325900147611 : SUCCESS in 37m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1fed09b5be214f878188d57e20ee649c : FAILURE in 1h 03m 23s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c028897dfe154800b60d32bfe4ec9037 : FAILURE in 45m 21s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0b9b833400554a0288d384df6dce949b : SUCCESS in 25m 44s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5d45a92c1cce4fc5bfce214360a2f96c : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fc3d55fe620ecbe2dab2b89d2b60b36f5e2454ae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-08 14:58:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"dac98f4d6a868700976ba9a19211ec14802bb015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:52:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19a257c7b04c45209070ae1b08675d8e : SUCCESS in 7m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7e3bab3d7a1f443495c066f9fefacba9 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f90a4212b1d5460ebf8aba7e7a7c2a77 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceeea492e7e3440e93c0fd6af3b2e1df : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7717f815f6304fc4912f23a48ac26791 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2269ba5ad6294013a92f69a66d62624d : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0645c08a7b934579afaec21168d071eb : SUCCESS in 5m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25edcf28346543cf93eedbf405acf806 : SUCCESS in 8m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03503b1073504b24ad57ee6f265a3523 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6595e66c03d6460a92febc2fec5020b1 : SUCCESS in 33m 26s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ae8eea25eb0c49f1af9ff4bd6feb8e83 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5451e9e5350141c7a3e6112d56588bf2 : SUCCESS in 1h 45m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b773ce7531784b03875afd6a2a86fba8 : SUCCESS in 22m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7aa4cbc59d3d471689a0cbbb4bcc1a30 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":2},{"id":"32de254c065d9c881c6deee6387b38b15e25b7aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 16:59:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9d83ca19a89a5b5182b00a57dd561756dd819511","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-08 20:04:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"505e5889dbeb038fbc7ee89ab91cbd1437cbfcb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:05:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"086a90b0fc8ed1bdb72763c120ac38a162dec4d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:57:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/46bdee7c6693419f9047e4f784979ac6 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7543064343ce488f9e664aea9dbdd52e : FAILURE in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8458a40e02e74ca7a0bfdd1cb374bb83 : FAILURE in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9207cd849b664df5b205456acdcbfa9e : FAILURE in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888e920041e54783900a0e4dd26d399d : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff358f5d1494f7393964f76e8d861fe : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ad2ad3d0c31d4818aaf26737804a55d0 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5725c75a6f374ddd971f7d8597712acd : SUCCESS in 9m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09f1b3786ae94a3d9bfbdff964c163b7 : SUCCESS in 28m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b247fb58aa954bf2bb74830f0c5c5c04 : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5d93ea9ce5c94dfb8345f15ca42f9faa : SUCCESS in 1h 49m 31s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b04ec94191794623b219e5637241d5bd : SUCCESS in 22m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8e1e24050ed543e996885e9468651eb2 : RETRY_LIMIT in 1m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e2c168dad192612f25949ba9a008328e8d21caca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-09 05:53:56.000000000","message":"Patch Set 2:\n\nrecheck\n\nName resolution issues","accounts_in_message":[],"_revision_number":2},{"id":"8c67f95a96f8dff752297a53610ed5a4c4449368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 07:44:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6280b6e64626484189ec8add0a1e6587 : SUCCESS in 6m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/506dfcd089f845579235b98ff2bd58b5 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0252f395b774e8fbed63d84850a3905 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d08f30eefa4832997dde4be8bd2a5d : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8eeef2bc8d2b42c0ab8cabda0b3876b9 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a5b297022446149f5bcf85ae703059 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a22894e24644d54ba8cde588f3f077b : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2df58f75bad4e34b592bed26f5907cd : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/abf4634db6df4d45a1580325d7fc47b4 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/68e916ec071647c3a3bffe501930d63d : SUCCESS in 31m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e8c64d92e5d34eda9617c93335331955 : SUCCESS in 1h 09m 28s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6a5712bf7118407d8cf7fdaaa1df8c81 : SUCCESS in 1h 48m 31s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ca02c089ea3049f78c97dac94b0375ef : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3f2d8b2b5be24485a3c1a4c39fa794f1 : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":2},{"id":"54bccee4df4b6382cffb9c2a5c6791a8645195ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 07:44:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e11a0cbb54f4e4aef2d674ab872c05317cbe6ba4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:25:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/39587cf1eadb4111a756f6b2fb459c6a : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ac46bcd8284588a61aca21c16f1ab3 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe041757e9584771b37cb5ba6e973188 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f19891b022b94da891af9061eee883a5 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b19696df5594c828afcd7b813063d3a : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cd1958bea6d4caf9ab461558281dcd2 : SUCCESS in 5m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/80b792613bf74280955e0b8fb6874025 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79774e089faf498d8af21de167a9238b : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e4f3381eb6e47c1b40fda1246e5da83 : SUCCESS in 31m 36s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c331e9e156964c598b93519a8940bf0c : SUCCESS in 1h 00m 35s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f8e10b7ea83441c1bf8bd0308e93fd81 : SUCCESS in 1h 39m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8e23966cd25d4c93a3873b4c1e8a1362 : SUCCESS in 21m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0ca26034204d49edb0c90d2922db4c73 : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":2},{"id":"6e03e030d9557a8fa84b65786b449a481b421d4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:25:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0461e1349657897e9a9f27be306d50256548ba11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:28:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0592d62513e4402a9227c6502f7fc9a8 : SUCCESS in 2m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5e71d8e7557416c843cc049d09254c8 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"152d549bee253a4da4c7518644544e54f5c4acae","revisions":{"0c943de674f8beb45345a28968d857f550c47c43":{"kind":"REWORK","_number":1,"created":"2020-04-02 17:29:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/82/717082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/717082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/717082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/717082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/717082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/717082/1"}}},"commit":{"parents":[{"commit":"78d6d5709b2d1ff6643c72be1aa0ae76ae225ed4","subject":"Merge \"Fix Kayobe overcloud introspection data save\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d6d5709b2d1ff6643c72be1aa0ae76ae225ed4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-03-30 09:39:26.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-02 17:29:08.000000000","tz":0},"subject":"Switch to stackhpc fork of resmo.ntp","message":"Switch to stackhpc fork of resmo.ntp\n\nThere is no activity on the resmo fork of the role and it seems\nimpossible to get any patches merged.\n\nChange-Id: I1f09f7c11767226e89b34687dab1553e87be76ba\nStory: 2005272\nTask: 39197\n(cherry picked from commit 31ad22abc46bae49728013c3635fc6d1329aead0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c943de674f8beb45345a28968d857f550c47c43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c943de674f8beb45345a28968d857f550c47c43"}]},"branch":"refs/heads/stable/stein"},"152d549bee253a4da4c7518644544e54f5c4acae":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-08 14:58:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/717082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/717082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/717082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/717082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/717082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/717082/2"}}},"commit":{"parents":[{"commit":"aaf8b769231a6f9292d056e53717e3cc114012b6","subject":"Fix multiple CI failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aaf8b769231a6f9292d056e53717e3cc114012b6"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-03-30 09:39:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 14:58:37.000000000","tz":0},"subject":"Switch to stackhpc fork of resmo.ntp","message":"Switch to stackhpc fork of resmo.ntp\n\nThere is no activity on the resmo fork of the role and it seems\nimpossible to get any patches merged.\n\nChange-Id: I1f09f7c11767226e89b34687dab1553e87be76ba\nStory: 2005272\nTask: 39197\n(cherry picked from commit 31ad22abc46bae49728013c3635fc6d1329aead0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/152d549bee253a4da4c7518644544e54f5c4acae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/152d549bee253a4da4c7518644544e54f5c4acae"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
