)]}'
{"id":"openstack%2Fkayobe~718088","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Ie44508fe3d3b08190afa5a43748e43548a63ff82","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"Ie44508fe3d3b08190afa5a43748e43548a63ff82","subject":"Fix multiple CI failures","status":"MERGED","created":"2020-04-07 12:37:46.000000000","updated":"2020-04-08 13:58:15.000000000","submitted":"2020-04-08 13:58:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718088-1586354295747-0a348a98","meta_rev_id":"5b64a1635262222c2224b81d1a40b84520b69135","_number":718088,"virtual_id_number":718088,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 13:58:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-04-08 11:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2020-04-08 12:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-08 11:59:40.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2020-04-08 11:59:37.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 13:17:08.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-04-08 11:59:37.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-08 11:59:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-04-08 12:01:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-08 13:58:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4ea7d51e83a6b12b997433797ea39440cac4951","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-07 12:37:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abf4cc932cd32b05f15416cd57cf659109496129","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 13:12:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a4668ee586b61acd7cea725abefaf25020a1ae40","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-04-07 13:17:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4dcf8b3cd45c0635d630e1a38885392f69a35a59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:35:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a262ae7a7499420786cb01f17a453949 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c57aa20da24e0da4953fa1b116a8d7 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27914b333c5a4828aa0ed0ec50d5016f : SUCCESS in 8m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d93f2a4c8bc34eb8aff810e2be8bf2c1 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083d6510272e4d8d87359c1e1ecaae8b : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff2035f7ab5949588c928b3af3cebd33 : SUCCESS in 7m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64a0f8337dee4e519fab856f04a7bba0 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ad42d831d184891979db9e0ec74cc42 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94b6a0b11cba4ad5b652c74f455f4afa : SUCCESS in 29m 09s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4a6185fc96b44975977ce3903082fc1b : FAILURE in 51m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f9a548cf795941a282fdfb207bf82078 : FAILURE in 47m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/08d2c4333f71444788b065d0a966c644 : SUCCESS in 25m 48s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4eeb05baf5e84532ac2bbc7da70530a3 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3e126f74d6788e86ebe09967c6f0e6ca2993d69e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-07 14:05:42.000000000","message":"Patch Set 1: Workflow-1\n\nERROR: Package \u0027tenks\u0027 requires a different Python: 2.7.5 not in \u0027\u003e\u003d3.6\u0027","accounts_in_message":[],"_revision_number":1},{"id":"bba54350079a0a143ad9351a2dca15a5ec0dbc22","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-07 14:13:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c00f2c7e31c33d22cd8f2381b8adacd8155a2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:03:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/946799e4acc146be9cc5c399a7d8ee45 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5272be09f17a43e2a3fff2c465b26ea2 : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73a04d0f876f402c8e23a3b1917c0a5b : SUCCESS in 6m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f8ac454fb375482194df47dfb97a6895 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a9d26854b9e493eac4daab111d319c9 : SUCCESS in 7m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e5e487eef9e4df09fe0979e9ae418ea : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9ca21e24d8a4b41b1c617892da9ba47 : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48f93abe8a284456ae707e92f6c325ae : SUCCESS in 12m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b35495cb5a840d9ad5387dac2fe6e14 : SUCCESS in 26m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e5d5a202d5d64b04a27c4036e389ad54 : FAILURE in 47m 56s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1c7b57f62f8341a6be664c38d406a3e5 : FAILURE in 46m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9d9e2df56ff443b5b154aae40c002e92 : SUCCESS in 22m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/88ff559f47de4ca2aaf103f57b406067 : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":2},{"id":"967e92819a25cff04da9f29467a150975713fee8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-08 10:11:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d2ea73b846e22bd834d2d6d026065f58e1686343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:53:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5e1f682e094c4bb3b2c258ac5cba4f33 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07346d72bc0e441892a8b455db3ec6d6 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de535b6ea6154c7fac06ab970784e944 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/40c339a2ab8a474cac795a608a108dc3 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd79913fc0a4b57b4d387665a2f8f22 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71cb996828e240cb8b4d4d02f12c5f1b : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c111e93ac62447239a7a71791a3f71d6 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6d1c6e937f3438a9e5957972faebda1 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/49839584565f41e7bf3a2ea3ffe8ceb9 : SUCCESS in 24m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/cb198bfb9bb440fe82fa1cc278aaad21 : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/da64728a04114ad8ac7a997f6d0f0aa7 : SUCCESS in 1h 40m 16s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/587b9a52fb5149fe8ba4620aa53b8b0a : SUCCESS in 24m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/53f6fd2784324a5ebaff533960b8d2f0 : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":3},{"id":"ac578a49bc38f328626d0baccb19a5e39d74b447","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 11:59:13.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"dd76369546c91460da6fd24b7f37deb79d49a26c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 12:01:39.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cdd37d36319c5f5d456e1e5a5d289e4826f3f893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 12:02:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ced7fbefa888ac8e31a6f74622a6b688425ad2fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:58:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6c59e69fb3f4446aa2dcdc7cd925e93d : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e1e7480f90f4b8d8a313eab2371a389 : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/035e61b5b0014a4aac04457aa4d19dc6 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e5f933e5774eab82794eb05db6e296 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1cdd0048cd7a4daba9eedc733bc6f84b : SUCCESS in 6m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ca8940348ad486882b612a868e64b19 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0464d14af3d545c4855cdc588192ea5d : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/082a1203c57b4f109d8f7fa2c0860357 : SUCCESS in 32m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5d12fb4e9c48494da5d54dacbb40d516 : SUCCESS in 59m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2cb88c36a7354d1281e7fad92067d77e : SUCCESS in 1h 48m 09s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b6198d58f262498f9c2e06b9deb4c5b2 : SUCCESS in 21m 25s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bba483aff1cd4dcc8aa5c8be7c0e5e2f : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":3},{"id":"bbdc895f368784f08138ed69e35c4f6dde41f0c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:58:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e7971ec4479fb65ccdce26040a956716445059ab","revisions":{"bccab76d3840948de3d909480b80138abbe8b518":{"kind":"REWORK","_number":1,"created":"2020-04-07 12:37:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/718088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/718088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/718088/1"}}},"commit":{"parents":[{"commit":"d92c7303d108abe29fac0a50319d1511a2dba378","subject":"Merge \"Stop gzipping logs in get-logs.sh\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d92c7303d108abe29fac0a50319d1511a2dba378"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:58:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-07 12:36:42.000000000","tz":120},"subject":"Install python-openstackclient using upper constraints","message":"Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\nDepends-On: https://review.opendev.org/717208\n(cherry picked from commit ef33e6ecb770f852bb4433d6b0d2dec7778937a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bccab76d3840948de3d909480b80138abbe8b518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bccab76d3840948de3d909480b80138abbe8b518"}]},"branch":"refs/heads/stable/rocky"},"45fbd06c1adfed9bffa45a464e461eb47f0f6fde":{"kind":"REWORK","_number":2,"created":"2020-04-07 14:13:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/718088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/718088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/718088/2"}}},"commit":{"parents":[{"commit":"d92c7303d108abe29fac0a50319d1511a2dba378","subject":"Merge \"Stop gzipping logs in get-logs.sh\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d92c7303d108abe29fac0a50319d1511a2dba378"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-02 14:58:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-07 14:11:25.000000000","tz":120},"subject":"Fix multiple CI failures","message":"Fix multiple CI failures\n\n1. Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\n(cherry picked from commit ef33e6ecb770f852bb4433d6b0d2dec7778937a4)\n\n2. Use tenks 1.0.0 for Python 2 support\n\nTenks dropped Python 2 compatibility in its master branch.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\nDepends-On: https://review.opendev.org/717208\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45fbd06c1adfed9bffa45a464e461eb47f0f6fde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45fbd06c1adfed9bffa45a464e461eb47f0f6fde"}]},"branch":"refs/heads/stable/rocky"},"e7971ec4479fb65ccdce26040a956716445059ab":{"kind":"REWORK","_number":3,"created":"2020-04-08 10:11:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/718088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/718088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/718088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/718088/3"}}},"commit":{"parents":[{"commit":"d92c7303d108abe29fac0a50319d1511a2dba378","subject":"Merge \"Stop gzipping logs in get-logs.sh\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d92c7303d108abe29fac0a50319d1511a2dba378"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 10:10:58.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 10:10:58.000000000","tz":120},"subject":"Fix multiple CI failures","message":"Fix multiple CI failures\n\n1. Install python-openstackclient using upper constraints\n\nDetect current branch from .gitreview and use upper constraints to\ninstall python-openstackclient, to guarantee compatibility with the\nPython version in use.\n\n(cherry picked from commit ef33e6ecb770f852bb4433d6b0d2dec7778937a4)\n\n2. Use tenks stable/1.0 for Python 2 support\n\nTenks dropped Python 2 compatibility in its master branch.\n\nChange-Id: Ie44508fe3d3b08190afa5a43748e43548a63ff82\nDepends-On: https://review.opendev.org/717208\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7971ec4479fb65ccdce26040a956716445059ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7971ec4479fb65ccdce26040a956716445059ab"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
