)]}'
{"id":"openstack%2Fkayobe~718601","triplet_id":"openstack%2Fkayobe~master~I359b07abadc8665ff0a8c3407ca0fc5effc504cf","project":"openstack/kayobe","branch":"master","topic":"lldp-port-descriptions","hashtags":[],"change_id":"I359b07abadc8665ff0a8c3407ca0fc5effc504cf","subject":"Filter out switch port descriptions on Ruckus switches","status":"MERGED","created":"2020-04-09 06:37:43.000000000","updated":"2020-04-09 15:36:51.000000000","submitted":"2020-04-09 15:33:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718601-1586446421328-899b8fd8","meta_rev_id":"df3a65df7c9223e904fc7ebda88520a5bd69387b","_number":718601,"virtual_id_number":718601,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 15:33:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-04-09 13:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-04-09 09:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-04-09 13:32:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 09:52:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-09 13:32:00.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-04-09 15:33:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdee2afc7155086d0422205cde5326d51154b1c2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-09 06:37:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5998195250ac5c6b9f54a441346250d794203522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:09:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ee6dc61d65447918cb89e474ae90dbf : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa43196d7274b649534f5b918546cdb : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a003d04162184510b2b70c69ee8d9bee : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e369d0140f74c0ba2b9710b7c79394b : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c2adbc0bbb44e959d2cd41f545dcec4 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c33ecf083d949eb83f724a4aea74805 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd264b17d65747c3aa24794e1130818b : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dd85a52daaed48bba45f7adb4e7b47b3 : SUCCESS in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/26dd4084d7e1464c94a1c095014800a5 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b817aff2714143b5b80a3bf37e1aa846 : SUCCESS in 35m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e957dd4ae1d44268952d3acc34731ddf : SUCCESS in 1h 13m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d75fbb681008456b852206e99a18927d : SUCCESS in 1h 52m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9d46e6abbfb84cbdbf3bc2bb8861dba3 : SUCCESS in 21m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/08ec21d245b740248b9e140a85499cb3 : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fb57d3d9e6f65263c80b42c2b75912851fb76867","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 09:52:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed99cee131a47cd0fcc67f5e639c0aa1562f1a36","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-09 13:32:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d1a3614062d94c64d0aaaf4e526644fcf9f2e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 13:32:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8de99737b0d045b84fad4312f5bc628e0b5ca4a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:33:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1df650508b6867bbe035a96830e5d3d9fb85617c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:33:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c7fe82444f498582ef9b0b58c715eb : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/948cf2c1cfa8453999926472d544fb78 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/081b952b09674047b98a9837b6cfa648 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3882ccd28a449788466fb075c0c80b : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87b1cd5c146e4362bbbeca395cfc61a5 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25b9cbf261734fd08a958e297c4f8ed6 : SUCCESS in 9m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bca5e2f989da40708989109036e22054 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0f0b735aeed43e5a4e2fcf22104bb8c : SUCCESS in 28m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/38d7ae7f440d4c9b8be2ce8e14594da6 : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b9e3467e4ad4437c9f8d308d5e7ad946 : SUCCESS in 1h 55m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3d44959aa983464faff7b362544bdf41 : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/325284a6035848ad817ba01299141d06 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c7afdf54f30774003c23a2ed4e0f7539a4d7aeed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:36:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73ff6e79e6f41b8a0be50cffc7c736a : SUCCESS in 2m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afbea67265e641cab64ee93bfebb6742 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"010681d7d950c11b0f50ea63ff20a714798949b8","revisions":{"010681d7d950c11b0f50ea63ff20a714798949b8":{"kind":"REWORK","_number":1,"created":"2020-04-09 06:37:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/718601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/718601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/718601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/718601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/718601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/718601/1"}}},"commit":{"parents":[{"commit":"7110477bcbf535df644bea5a6e68402fdddecc3f","subject":"Merge \"Install python-openstackclient using upper constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7110477bcbf535df644bea5a6e68402fdddecc3f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-09 06:35:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-09 06:35:45.000000000","tz":120},"subject":"Filter out switch port descriptions on Ruckus switches","message":"Filter out switch port descriptions on Ruckus switches\n\nSome Ruckus switches, e.g. the Ruckus ICX 7150, advertise switch\ninterface names as switch port descriptions. Unlike Dell switches, there\nis no space character between port type and port number. For example:\nGigabitEthernet1/1/9.\n\nUpdate regular expression to match both styles.\n\nChange-Id: I359b07abadc8665ff0a8c3407ca0fc5effc504cf\nStory: 2007532\nTask: 39343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/010681d7d950c11b0f50ea63ff20a714798949b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/010681d7d950c11b0f50ea63ff20a714798949b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
