)]}'
{"id":"openstack%2Fkayobe~720040","triplet_id":"openstack%2Fkayobe~master~I555ee575ccec0cfbcc4c4bcb53677796c83227e3","project":"openstack/kayobe","branch":"master","topic":"story/2007555","hashtags":[],"change_id":"I555ee575ccec0cfbcc4c4bcb53677796c83227e3","subject":"Add support for encrypted block devices","status":"MERGED","created":"2020-04-14 17:46:51.000000000","updated":"2020-05-21 11:15:14.000000000","submitted":"2020-05-21 11:13:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720040-1590059590792-472850b6","meta_rev_id":"966bf8f4a107b6d8c4f37e7a19662dc1c576765b","_number":720040,"virtual_id_number":720040,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 11:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-05-21 09:19:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-20 16:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-05-21 09:19:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 16:02:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-05-21 09:19:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-21 11:13:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b5f1ee8555100f9784e7137841a16304b2b4952","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-14 17:46:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ce28947697fd79aab2971090f7d377a98ceeabe","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-14 18:58:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d684fa843a61c1db84dd8559c372bcfdb7b62c0e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-14 20:20:26.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eaf8f28d12d25420e05ed6306e7f36258cb473d1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-14 20:30:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)\n\nOther minor issues I noticed.","accounts_in_message":[],"_revision_number":2},{"id":"3bfdc82258133f8364c6670526a775b10a0d5d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 20:50:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d78feabb2e0483d9166b19ea0c06880 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08bd05f50c9d4b408273000e47756289 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eca837432120476d9d9cd7a6a711b991 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0eee78bf575147e5a40c2c582f403337 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8b9fcacbf8c4260b93877f7f75e26d9 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a9408467e94410ba7dfcda98e3ab86 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae67285c6d624635b3c8d7d6ee5f2ab2 : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/716888900740442a868c4cce7beed7df : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f85bd902335e418b94298b5913368530 : SUCCESS in 10m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d3be94e695b42cc8152db8052d98a82 : SUCCESS in 37m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/03c647e6b4dc4afb978e2d2d179fd944 : SUCCESS in 1h 12m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6f4a07dc89df46799ccdbe8aff185442 : SUCCESS in 1h 48m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1b5dffb0448f462f90b5472b160e9ee2 : SUCCESS in 18m 25s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f9cb7fcc18f34ac1be80644ca0e232c3 : SUCCESS in 28m 28s","accounts_in_message":[],"_revision_number":2},{"id":"fbf1982f4926422809843090ce18b80047d7f182","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-17 13:02:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5279c84a8b83e47faf6a0b62c506b4870c4f7874","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-17 13:02:55.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"955127fb05f27b21d00898e4f9389b19f1ed13dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 15:00:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/967fd9073c89481681e0ef4068d7737a : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2bc2612a53946f0ae67e886c2a96a6f : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b13894f3f8ba4de3a3f61358ef9c09d0 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bc259f9716a4d1391cab316c8ce4f83 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c846f02084044ec948c6b059f680829 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69fd98e91da14c7dbc84a1bb281ef1e5 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/975b76f10f0b4d589d49e8cf4f7b74ce : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fdc22ff4e18c4696b6f96ca6208a5f14 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/18b4b94e226d4a218df75a4b280e01cc : SUCCESS in 9m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/80966ea10e8046f7bf70823de7a90ef1 : SUCCESS in 36m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d5db6c9eb757433dbc47bcc7f6226099 : SUCCESS in 1h 10m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/61ccfac87164437a923ea3210a73cbb6 : SUCCESS in 1h 51m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e1ed616442f64043882ac9bdea438872 : SUCCESS in 18m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b429abd4f1c24d20b8c98e7dec13f32f : SUCCESS in 28m 54s","accounts_in_message":[],"_revision_number":3},{"id":"b9cb4cd6b7d5809ac8c6f0dce57b3de562f9a544","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:45:22.000000000","message":"Patch Set 3: Code-Review+2\n\nLovely stuff. It might be nice to add a test for this to kayobe-overcloud-host-configure-centos8.","accounts_in_message":[],"_revision_number":3},{"id":"629249c5889828f10452ff3bdc68f0082138cc7f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:45:42.000000000","message":"Patch Set 3: Code-Review-1\n\nActually, has a merge conflict. Test can be added later.","accounts_in_message":[],"_revision_number":3},{"id":"cc4662ffd980589eabcfe89f555c721a4435b2bb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-30 10:10:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"929b20bcf0ba6ae7fa11412199d6011b9bb20960","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 10:57:34.000000000","message":"Patch Set 4:\n\nIt conflicts with https://review.opendev.org/#/c/720818/10 which we want to merge, maybe we could rebase on it? Unless we want to merge this first, so we can backport it easily.","accounts_in_message":[],"_revision_number":4},{"id":"e2c4ed730bd15495c4f436d184f075cd8f8cc594","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-30 11:08:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bf9074f1c0be74b5ff80f6b629d6ceb13ea4c657","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-30 11:54:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"59021c5cfb8041f3025b10a60ab53cc308165755","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 11:57:44.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b0e91cf01b3254fb2d10b20786f2d6d58bee47d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 11:57:50.000000000","message":"Patch Set 6:\n\nLuksury","accounts_in_message":[],"_revision_number":6},{"id":"7e515571ad440611ba01f8e41ab9d1cb2fe3050e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 12:52:34.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0591b01ecae8a8a60b0f4dcb5bd2fede44040e1d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 12:57:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e925b9969fd20d13b715e87c89eca5408eb0a326","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 12:57:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"781de9780d3c06f7937434dd8643ec0371fb3b42","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 13:16:11.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"fac68d2f687a689bd796126df1f7c6ef65543381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:02:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a414d1f3e20647aba3853faec73078c6 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a12bc56480c8413382d3129fbfdcf629 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4c8455f4b284bd4972b94a2979308ef : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dfdc00105a540f49972974990e512f6 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc7ea55d854e476b9f1cdbeb42a6fe5f : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0bc9fe5454f4f3cace4915c5e5bfb0d : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2192ee9e930849a4b6ad5f339b85a05a : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e1aa889074054ef4b099df20c0818d21 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b8a07bb709f453b89f2e13cd508721c : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b39572f22f9c47f8985b9088ff0c5296 : SUCCESS in 34m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/223f9d12e7ca41dda6beae27b750ec0c : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/21f373bd5bca4b25b75e66a135f605a4 : FAILURE in 10m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7dd2985bd513433baf30b0e57d9138f4 : SUCCESS in 1h 59m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3d99f32f19dd4756a565e42dff7c700a : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/630a0eab96a14c42bdbd955045ab059e : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":8},{"id":"970797e60c8d4f495bcdd801b9399c214f832629","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 15:13:33.000000000","message":"Patch Set 8: Code-Review-1\n\nGood call on adding the test, it doesn\u0027t appear to work:\n\nfailed to open and parse crypttab file: [Errno 13] Permission denied: \u0027/etc/crypttab\u0027\n\nMissing a become? But should the role do it?","accounts_in_message":[],"_revision_number":8},{"id":"2fb3a5e67a8796da334872547d97dd827ba1e3d4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-06 09:51:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"49da94f8b4760c21983acec88ad0f0837c5fa51a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-06 09:56:31.000000000","message":"Patch Set 9:\n\nRequires 0.3.0 of `stackhpc.luks` which doesn\u0027t seem to have been imported into galaxy.","accounts_in_message":[],"_revision_number":9},{"id":"dc0c6a1b370a1d50a8b0222983ca0769f71c5b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:08:32.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3737ad41fea94032b682c1580244d5ed : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ccb37a3d8c4c8eb96fe18f1f17a6ff : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/483da95fe9ff4b0ab9e45ef2bdc481fb : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4250b9d2ee20402caa16ac9e9c24a9ca : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf938ef5830b45448bbddcb89a6de628 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/639a1fcf3fec438799e31430d57a5328 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8753ffc90fd4505aa2a13efef8894ff : SUCCESS in 5m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/112bd6d7b7374f29ac865395cbaab7be : FAILURE in 7m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a42c209ea35459d991d04db331297b9 : FAILURE in 5m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/acea9956366b4e2ba2e91ee2e554aca9 : FAILURE in 10m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5c38965590b74cf197d6564678c181e3 : FAILURE in 7m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f3f7cb17dd03477691730066f414f947 : FAILURE in 8m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d9b780e4868f489191448c77bec92e7c : SUCCESS in 2h 03m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7f6b51bb9478459cb19a3fb3a4f80bd0 : FAILURE in 9m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/25a273c92559482d9a2f50ffed1533a9 : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":9},{"id":"02d76e32e61fae9abe682a8af920504aa7fc34a1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-06 12:43:20.000000000","message":"Patch Set 9:\n\nrecheck: stackhpc.luks 0.3.0 now on galaxy","accounts_in_message":[],"_revision_number":9},{"id":"386b4f43f176acd10cff2c77cf54741a43e86577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 14:38:00.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/554ddc58a5374b609ee82e886eddd176 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b54361226864661b81ca97995fbdc74 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c2a43e2097c40f48f4c191d04c120db : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/867eaa6b6a024503a93d8d9dcaf1623e : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0519e57f9948475bb10fe52003441806 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe67fd349bb48cabe4deae254be3f5e : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/806d3b2a6331400c833891e29ffd1390 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0160b7a5a0ff4f2cb633e88472a5a33f : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/14ac0160fc5a4795b0ed359c3d477745 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0a299672f60f4cd69078d9fd55e6a906 : SUCCESS in 32m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/af493549f61f45a89d336a585120419b : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/02c119cf07f543b5b0d7b7411f28106b : FAILURE in 12m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/04c24ceda7b941a69f0242e0d119f8db : SUCCESS in 1h 53m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2cd987ee950246eb90cc5ab570e8a72a : SUCCESS in 18m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/935724e8e7c8403b8520df7393a2d8e6 : SUCCESS in 28m 21s","accounts_in_message":[],"_revision_number":9},{"id":"59cb17cc945ae06240da8f13e2bacce7bea83e2e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 10:39:40.000000000","message":"Patch Set 9:\n\nStill seems unhappy.","accounts_in_message":[],"_revision_number":9},{"id":"c0eb40f6f41d1c6c49642aa32d4682e318bdb05a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-13 09:55:02.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"3d193a794f0ed49a95fa7df1ad551d2123eec48b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 11:53:43.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b998d4c4ab00416eb894d11172a1f69c : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c04cf45ca5054edebe983f107d816740 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69df57316c3349f1859afd8ced67b15f : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc7a7c0b82c6479aa4764cb0cfb8ab32 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41fd2d7d0e19451d878396e631cd5bc4 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6bd8797ec984df0bc778644646e2f5d : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6be72e34e17345878f1969b86dbc857e : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73630e9a256847b7ae21353465338e1f : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/69196dfba11348249819f96b7f3f9805 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ce8ad50c7a54940b2bbada7d6097635 : RETRY_LIMIT in 7m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9116851c30a44d25aa0120405c4a2fdc : SUCCESS in 1h 13m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/429c3a9d89814f0190cd064327018e49 : FAILURE in 12m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/844e95ab75f848bdab229a71c22e8605 : SUCCESS in 1h 52m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ba36ad01815445c8ad7e776402781258 : SUCCESS in 18m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1338388a43c84c298a3fb3adad336193 : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":10},{"id":"d5b265426be67c031cc0e9e651fc4bee93c26dfa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-13 12:45:34.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"ba42c6f6956111ac2e9fb404e68247670df5f5e0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-13 13:10:24.000000000","message":"Patch Set 11: Workflow-1\n\nDamn - Still the same issue. Will need to try this on a centos8 generic cloud image to see what is is different between that an the vagrant image in stackhpc.luks CI.","accounts_in_message":[],"_revision_number":11},{"id":"b0ba30560d04a6a6a04a0d0155107c285a9ec0de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:42:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b51c89ff3a64e05af5fd1652f04ecdd : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9981e655b4141b497c3ccae418b76ad : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742e1d2258994575a6baef20c0616050 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf2726e348414b25bafabbc88896c6fe : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90f7240722f34d548bc9f7fc4c1870a9 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a0c198e6d2424789fcab323275ee48 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe595a6746ae43f486cc638832126df2 : SUCCESS in 6m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bfaa287401564675bdbf662b12e7a0fd : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f0699638e87491799d95d8ac7732f32 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c6c479f90be34fb184a0d21c7886d102 : SUCCESS in 26m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5a3a834a6ac447fa9dcf8f8237fd918e : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e804b2f90f204c74b5a09c434ed839d4 : FAILURE in 10m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/43a4a2496a7e46c993a240198688d5f6 : SUCCESS in 1h 50m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4871491d3d9d41db9db33c9973fbf7f2 : SUCCESS in 17m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b0dee67e26e245a1b36ef3860e58e953 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":11},{"id":"388e739e44868965ed7cae6c66c0badc985e5c79","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-15 15:35:52.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6e086675403089e1e78a5a35006e8a1594278de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 19:32:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22e6c29425e947459c9165f23231097f : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b302ffa9d274749866397f9f1b7efa5 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8dee92c5ab14ac0b9daf185caeb8a52 : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f0c468299254b1b9549ce6a44132526 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59f581b63fbb4177aa6a6febe4be3365 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad020599093431fa3bd0b601eae4d8d : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d77dde4c116c44548ff559657c59c60f : SUCCESS in 6m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0002ce003764437db398ab75457e6ca1 : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fee93fcf40574339a31ba47b7568c43c : SUCCESS in 9m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40c9d4b125544bc9be0431a20e009583 : SUCCESS in 34m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7729103e60da42ef97d54b5c595376d7 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a11269aed6184366a86de442c63e7c30 : FAILURE in 18m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9b9bfcc504de435cac921f9de8c1c4e6 : SUCCESS in 1h 54m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/faa9dac17e4a43758c23e85fd0a60c54 : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5133b9c27ed040afbba9caa09ba74d63 : SUCCESS in 30m 00s","accounts_in_message":[],"_revision_number":12},{"id":"5f0d955a5254fe1af8c669dbcd89d5c79ab43230","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-20 09:08:50.000000000","message":"Patch Set 12:\n\nrecheck - error looks unrelated","accounts_in_message":[],"_revision_number":12},{"id":"d52cde89f312a0affd3cd642fe3693848269c013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 11:04:54.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acfed05fc7cc421f84459e0bef64e7c1 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f14c289d9191464486d68d5510a62f3f : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7b930f173ff4a2fa0198dd32b215abd : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4bb89518a894209b6feffa25420626b : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3c7ff6e0f504a1fbfc9e5d0cf8ccada : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ce2481849142f7915a0a484829e709 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a28b121d3dfa457eb0498b3a46ff50f2 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58d684cf8d5a42929e9084b33fbba2ba : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ca03cd9c5624da287bd9107b588667c : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37113cf7e2294a00993230a4e09589c3 : SUCCESS in 26m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b21c04bb55c245b7875f3e41459ebdf1 : SUCCESS in 1h 17m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3c33a9dbd912497e887c86f13d472217 : FAILURE in 18m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a1dd41c5ed854ea2b9f31586ea622b26 : SUCCESS in 1h 51m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e49777f3688446a5898f3921c39658e5 : SUCCESS in 17m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1ea719da172d4554960323f99d8014f2 : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":12},{"id":"548ce12351bc61980b50da74082045cd89c500e1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-20 11:55:27.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"37e3160b5cd75c2c1f0fd9e77cfad73297134cf2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-20 11:57:08.000000000","message":"Patch Set 13:\n\nMy bad, misread the logs. The actual error was to do with with the flaky test:\n\n      def test_luks(host):\n          blkid \u003d host.check_output(\u0027blkid /dev/md0\u0027)\n  \u003e       assert \u0027TYPE\u003d\"crypto_LUKS\"\u0027 in blkid\n  E       assert \u0027TYPE\u003d\"crypto_LUKS\"\u0027 in \u0027\u0027","accounts_in_message":[],"_revision_number":13},{"id":"f5f4b7ae5ac7d5a2d5713a5d2a7f65f54789e2e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 13:53:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f22b0a2f53074c319c146f47961872d4 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405046a697794919b62810bb09e1f29d : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8bf4aed073dc4100bcb74b73300f3492 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2eac870179b64585b4db6aef14901049 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/400cabb3ce5848459ca93300e993d53f : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553c5c5ebeee440991d9fe214336dc97 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcaa36e6e24849e495b46a51d0bc35f9 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d31c6a748b44b2280ed95d5274f3042 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5dff53d0046449918764c124cf5e934c : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/23dff9d32eef44e8a5b2bd5d86f7a979 : SUCCESS in 35m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2561b83dd0ef4314b3bca6bff95f66eb : SUCCESS in 1h 14m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c281742198e94502903d58d5a4c1a75f : SUCCESS in 16m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/02fe7b1da7f145aaabaaab41a61ffc03 : SUCCESS in 1h 56m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e4e2750e41954c429337f17b16702986 : SUCCESS in 20m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d6a69e05170f4a5f8a1c2d9f7e3e50ea : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":13},{"id":"5d7b25cc138826e39c6b36d427ad08bdd46cd32b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-20 15:26:59.000000000","message":"Patch Set 13: Code-Review-1\n\n(5 comments)\n\nLooks good to me. Very minor comments about etc/kayobe style not matching the rest.","accounts_in_message":[],"_revision_number":13},{"id":"ea034004b1a89c3634f80639205616f64c84c283","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-20 15:49:58.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"59b4b7faff6925f48c077572f69d36e965bb757a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-20 16:02:25.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"d00637e3a67b169a7d6d258da7816083b5489116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 18:23:17.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d301d652dc3346c48be9c2850ddeacda : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a474f6c8884f06952abfaca030fc5a : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97ffe205880242daa778812c78703bc9 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb2909e338e04dcf8cfb2a4bd9abc638 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/138b0e1b979b48548417bf0ff6ceca31 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97b498128adc4ffeafc74173ae16c02c : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cf7a6a134434ffcbbcdf520e6263ccb : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85a3e5eb60704d6f803ad74b25eb1487 : SUCCESS in 8m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9c4e247db1b94b43a580e75fe992f833 : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/419e983a78c440dda5bba4ca89d80195 : SUCCESS in 29m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/134cdb8c4c4348909ee1a2ad747250f7 : SUCCESS in 1h 14m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/96c4b238a1c5450cb8d80203a4224a5b : SUCCESS in 20m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a3f92698be164dd7b5a79b9e861a9427 : SUCCESS in 1h 55m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9bb2466c54d942ea97b3231fd791c9c3 : SUCCESS in 18m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4ff21d51638f40b1adfe3ace0f59b491 : SUCCESS in 32m 18s","accounts_in_message":[],"_revision_number":14},{"id":"84bd9b3184c1919147a0b2cb522b57c903b45dc7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 09:07:51.000000000","message":"Topic changed from luks3 to story/2007555","accounts_in_message":[],"_revision_number":14},{"id":"46a56908567f359736cb8382ce87f3c3d3cf2662","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 09:19:27.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"8effa69ee61e7fac3eb42aaf609e00992a1e510a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 09:19:39.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"06f6873767e41adfbb0998cd32a204a3cb77f641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 11:13:10.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315267b2dc884eb7b4df49e86ae37382 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/034bab363ef649d58c8190e6e856d05d : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4260bcd9b5f94495a668ff8bd2f181a5 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c71110b4c93468daa8740d32b698f45 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/482bd3a5852747a39012b9dec8c0b935 : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c014a51b1c74a31bef9ab02cbfe3117 : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a157595ea7c4d518077ac7e91cfc7c6 : SUCCESS in 9m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df0974053c3d43d3b371a75d42a61c98 : SUCCESS in 31m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3c4f1cfc1ddb4b91b2015a172b37a624 : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f3d206f3bca44f1e8e26e45e142bee70 : SUCCESS in 16m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/553aa7e1f5d6424a93bda48258656c1e : SUCCESS in 1h 51m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d5d421db42164843b061da210e1dead8 : SUCCESS in 20m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9e02a38ee48d47e0818fbe1ef2b5bb52 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":14},{"id":"415aae7db6d3246cea2c58ee666b93b04fef448e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 11:13:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"966bf8f4a107b6d8c4f37e7a19662dc1c576765b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 11:15:14.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1828fb328e2441119d3a965cc1966713 : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/895f969499b24debb8193ddd54aa3867 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"5d2a46b2b07c93d982c7d2352c943f6b4728f3e9","revisions":{"b26e0644cfb556213456095ee8f357085aa1d378":{"kind":"REWORK","_number":1,"created":"2020-04-14 17:46:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/1"}}},"commit":{"parents":[{"commit":"9d5a671f3c94e8cd6b8a82bb1eb7b967c93e14d2","subject":"Merge \"Filter out switch port descriptions on Ruckus switches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d5a671f3c94e8cd6b8a82bb1eb7b967c93e14d2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-14 17:40:43.000000000","tz":60},"subject":"Add support for encryption of block devices","message":"Add support for encryption of block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b26e0644cfb556213456095ee8f357085aa1d378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b26e0644cfb556213456095ee8f357085aa1d378"}]},"branch":"refs/heads/master"},"34388eedf02c3fc6e75278685b05185c5c8015af":{"kind":"REWORK","_number":2,"created":"2020-04-14 18:58:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/2"}}},"commit":{"parents":[{"commit":"9d5a671f3c94e8cd6b8a82bb1eb7b967c93e14d2","subject":"Merge \"Filter out switch port descriptions on Ruckus switches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d5a671f3c94e8cd6b8a82bb1eb7b967c93e14d2"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-14 18:58:33.000000000","tz":60},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34388eedf02c3fc6e75278685b05185c5c8015af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34388eedf02c3fc6e75278685b05185c5c8015af"}]},"branch":"refs/heads/master"},"7ec3b97c8117ae26ee282adea66c4838b6cd5fb3":{"kind":"REWORK","_number":3,"created":"2020-04-17 13:02:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/3"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-17 13:02:27.000000000","tz":60},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ec3b97c8117ae26ee282adea66c4838b6cd5fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ec3b97c8117ae26ee282adea66c4838b6cd5fb3"}]},"branch":"refs/heads/master"},"0a5d1193867b268e34aba987fbf8c8c04f30c3b8":{"kind":"REWORK","_number":4,"created":"2020-04-30 10:10:00.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/4"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-30 10:09:49.000000000","tz":60},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a5d1193867b268e34aba987fbf8c8c04f30c3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a5d1193867b268e34aba987fbf8c8c04f30c3b8"}]},"branch":"refs/heads/master"},"98f2a3469acb87afcd33ced1f00d6dd670f13653":{"kind":"REWORK","_number":5,"created":"2020-04-30 11:08:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/5"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-30 11:07:14.000000000","tz":60},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98f2a3469acb87afcd33ced1f00d6dd670f13653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98f2a3469acb87afcd33ced1f00d6dd670f13653"}]},"branch":"refs/heads/master"},"8f353e4df0c40c5db78667771f0b47715e6b1c08":{"kind":"REWORK","_number":6,"created":"2020-04-30 11:54:12.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/6"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-30 11:53:57.000000000","tz":60},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f353e4df0c40c5db78667771f0b47715e6b1c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f353e4df0c40c5db78667771f0b47715e6b1c08"}]},"branch":"refs/heads/master"},"b511e1e63ddd5d76ccb6dcf4061fc29bf7d281ba":{"kind":"REWORK","_number":7,"created":"2020-04-30 12:57:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/720040/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/7"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-30 12:57:05.000000000","tz":120},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b511e1e63ddd5d76ccb6dcf4061fc29bf7d281ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b511e1e63ddd5d76ccb6dcf4061fc29bf7d281ba"}]},"branch":"refs/heads/master"},"e94cd1d3f46034653823735927e6bf1777a8f005":{"kind":"REWORK","_number":8,"created":"2020-04-30 12:57:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/720040/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/8"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-30 12:57:52.000000000","tz":120},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e94cd1d3f46034653823735927e6bf1777a8f005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e94cd1d3f46034653823735927e6bf1777a8f005"}]},"branch":"refs/heads/master"},"96d6a9d4cd241c2bc15d09acaaf8d7b1717df872":{"kind":"REWORK","_number":9,"created":"2020-05-06 09:51:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/9"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-05-01 18:47:39.000000000","tz":60},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96d6a9d4cd241c2bc15d09acaaf8d7b1717df872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96d6a9d4cd241c2bc15d09acaaf8d7b1717df872"}]},"branch":"refs/heads/master"},"3a863d1899567c2254bea3dfb5c62ff127c0d78e":{"kind":"REWORK","_number":10,"created":"2020-05-13 09:55:02.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/10"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-05-13 09:54:46.000000000","tz":0},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a863d1899567c2254bea3dfb5c62ff127c0d78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a863d1899567c2254bea3dfb5c62ff127c0d78e"}]},"branch":"refs/heads/master"},"2fcfe1d113a3f317834088521845ef617b802152":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-05-13 12:45:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/720040/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/11"}}},"commit":{"parents":[{"commit":"840c56167ac104ec18f7efde7b558be77274960e","subject":"Merge \"Fix multiple CI failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/840c56167ac104ec18f7efde7b558be77274960e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-13 12:45:34.000000000","tz":0},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fcfe1d113a3f317834088521845ef617b802152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fcfe1d113a3f317834088521845ef617b802152"}]},"branch":"refs/heads/master"},"b615451e66bc4857f4fcb65eec668bb481e7943e":{"kind":"REWORK","_number":12,"created":"2020-05-15 15:35:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/12"}}},"commit":{"parents":[{"commit":"840c56167ac104ec18f7efde7b558be77274960e","subject":"Merge \"Fix multiple CI failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/840c56167ac104ec18f7efde7b558be77274960e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-05-15 15:35:41.000000000","tz":60},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b615451e66bc4857f4fcb65eec668bb481e7943e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b615451e66bc4857f4fcb65eec668bb481e7943e"}]},"branch":"refs/heads/master"},"05542fea39098b06f939cde6682975345cb01d27":{"kind":"REWORK","_number":13,"created":"2020-05-20 11:55:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/13"}}},"commit":{"parents":[{"commit":"840c56167ac104ec18f7efde7b558be77274960e","subject":"Merge \"Fix multiple CI failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/840c56167ac104ec18f7efde7b558be77274960e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-05-20 11:53:58.000000000","tz":0},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05542fea39098b06f939cde6682975345cb01d27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05542fea39098b06f939cde6682975345cb01d27"}]},"branch":"refs/heads/master"},"5d2a46b2b07c93d982c7d2352c943f6b4728f3e9":{"kind":"REWORK","_number":14,"created":"2020-05-20 15:49:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/40/720040/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/720040/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/720040/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/720040/14"}}},"commit":{"parents":[{"commit":"840c56167ac104ec18f7efde7b558be77274960e","subject":"Merge \"Fix multiple CI failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/840c56167ac104ec18f7efde7b558be77274960e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-04-09 22:16:00.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-05-20 15:49:51.000000000","tz":0},"subject":"Add support for encrypted block devices","message":"Add support for encrypted block devices\n\nThis can be advantageous in deployments with a data security\nrequirement.\n\nChange-Id: I555ee575ccec0cfbcc4c4bcb53677796c83227e3\nStory: 2007555\nTask: 39410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d2a46b2b07c93d982c7d2352c943f6b4728f3e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d2a46b2b07c93d982c7d2352c943f6b4728f3e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
