)]}'
{"id":"openstack%2Fkayobe~720278","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ie0edf6a924a914395c6502e2d5cf1139bce14a48","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Ie0edf6a924a914395c6502e2d5cf1139bce14a48","subject":"CentOS 8: seed VM \u0026 bifrost","status":"MERGED","created":"2020-04-15 19:34:41.000000000","updated":"2020-04-17 11:13:01.000000000","submitted":"2020-04-17 11:09:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720278-1587121743991-767da42d","meta_rev_id":"63bc9e7544d0b8201e3eb6f6b7fb01b429130b32","_number":720278,"virtual_id_number":720278,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 11:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-04-16 12:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-16 14:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-17 07:25:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"date":"2020-04-16 09:27:19.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-04-15 19:38:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 09:27:19.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-04-16 12:47:28.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-17 07:25:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-17 11:09:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07506d1a83ab4b9ea9c82b349e62c26f3c0e5ff7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 19:34:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"541e4ab15ffe4d3137514d0ab9f81ccd141d7d94","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 19:38:09.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"208be625539a6df52f9383316a0f85e36805f31d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 21:34:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f3a1c01b4f84b62bd3b8f83ce6b1cc8 : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fb71acfded1244b484c7ffa5b4f0a01e : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd1cb743d572418f815ada05dfb3862f : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f067bb3dd7c47b89a68124fe0390cff : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5341371db50a4ce49d8b0eba68bf33c5 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6be5d2cae5f48ffa366dde0d103c3c0 : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e963add79c2d45939cbe92d3a373605b : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2beccccefd9e4952a4825344b1397485 : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4d9d991367bf4951b42261bd37a1860e : SUCCESS in 36m 18s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b62ca3a9a71446c484265653f28ca5f3 : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2c5edc9065bf49d3bdb5001a8a9b967e : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f97efe1bdd8d48d6a159a3d62a35aceb : SUCCESS in 1h 55m 19s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5500d92bd5434ce7b218901e63acf011 : SUCCESS in 26m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/996426a19beb416a832c80b87488c80f : SUCCESS in 17m 23s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/81ae7e6ed66a425c8a4237638778cec5 : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fe8a5f35b11d059866e8cdc1c7e1f89e4b3739aa","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-04-16 12:47:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6e1c263a949aec85bd6988b4a38e09a2deba5a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-16 14:47:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2175ea51c0d1d433a28c7882a41380fc6dfc998","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 14:47:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8adee7af770a0c78dba4cc9a82ca488deb5f6949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:08:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/514f61229dc84ed2813f27ecac7d0c01 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17eb8cbc889c43769f891c3342967659 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b89ff6c447b4213a57856f49488805a : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a00a0006554648989b684504fa44b0ac : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fef63bf9ec14bec87750f7fd2f6ee43 : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7581f549664d43a5a466c15c1c767806 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/822f9c0d019c4ac4897da99db63b0be6 : SUCCESS in 9m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ccd58cd481b4258b7cb51dd63abd117 : SUCCESS in 37m 06s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/43588fe1d7aa4310a11fea21a7ec8bac : SUCCESS in 1h 33m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/23124f3ceb9c427280cfc10ee24c63fe : TIMED_OUT in 2h 01m 56s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6c75506c8a2a40d19c7fb0b681233cf7 : SUCCESS in 2h 17m 33s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ea2a5134163346eaa37172fa6197763c : SUCCESS in 33m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/49c2ca06071a4050864bfb194cec8272 : SUCCESS in 20m 29s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/abc8983fb042463e99b3f63e8d93d939 : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b08c2e956759788dbf7ecf0b756546a4b9f227c4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-16 19:32:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50ff854047a5cf58a26802c0c3676a475f217c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 21:20:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac9dc5134da64736a5f901f4507c3b24 : SUCCESS in 6m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c0e3f51e50f44c5a849c8be0ce88877a : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17af35f1b4c434bb352d45af99879d9 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b8f686ba1f4ef8b12c59a64a97e5eb : SUCCESS in 7m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/187f0fb756a747549db8b06b56b629af : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1633d762ac540988a70af77c8ec2861 : SUCCESS in 7m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b3a962783994cc1b7357c1d171a312b : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a4d1855591041d295088cada150c970 : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00eac83a267c475fb86ed791b87fe474 : SUCCESS in 39m 16s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b20af35e7f984fa9b3b9135786da6ba5 : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1a6dc5d7b00b4a3f95d1268652f35345 : SUCCESS in 1h 04m 51s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b20265cfeb1546d99802afc3af2e2370 : SUCCESS in 1h 44m 15s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/d3ec9b243a564a9d99d3bfa7493df432 : SUCCESS in 22m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/077103e37379461db9f3d8f2eae3b2da : SUCCESS in 21m 20s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/304d80d2403f4db5b03855ed8c08b5e2 : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7ef4532ecf15ee85504434124244ba2c46221c13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 21:20:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"835ea0c8c61356c90529274647decec16a787954","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 23:05:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/29956e0ae2954876a839bb903bc19b2e : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bff9b635507349e79fdec67751aca009 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ba4491a78b4607aa66d1cf74a5a34d : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73287fd77d004da2baa4e5eb22079e8d : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c0739172e84363a9a2cb835b753604 : SUCCESS in 9m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a0ef6fe76114d2e9d185c605fe1c8b6 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8866e29f552948ffb2643bc099cab545 : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1068afc37f4f4ae5acb104d24a4f72e1 : SUCCESS in 36m 09s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/1d1c7cc8377d409fb58993c9ec4a0632 : SUCCESS in 1h 16m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9d5637de48b040a99bd947cf3abcdb3c : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2046447e1a3741409ace4aab6496433a : RETRY_LIMIT in 1h 39m 14s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/952f43b9bad9498484b7633ccadecfe9 : SUCCESS in 22m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2cfe8b353dbc48aebd64014a17e2393f : SUCCESS in 19m 33s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e5b068a1b0c24b6cb011893a6814d03a : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c17a06d8827da9779dd1907e31276abe5370b74c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-17 07:25:07.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"93ab17af3af4374651163d0cf84e8d5ea1ddba83","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-17 07:25:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d9dfe81173ff87128c129feaf6239fa4a0f00b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 09:18:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/effc80a98ee64811a440efcfd4c7965b : SUCCESS in 5m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9c5acea0bfa24455b76ed9ff2b756a7d : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f1b5ef02b284190bfcaa77e31f71d91 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671b8587f2a04b43ba3cb2c3fed18514 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e5cb8e9aff04a4bbbaff6c3c246b58a : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ac7b6bbeb64de69811c64b3d186627 : SUCCESS in 9m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/320a9a2619d445fe8853c4a90ac07b35 : SUCCESS in 7m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bf7ce05d41f46a3af9b39d451b725b2 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32862c57345e4a60b4e01789a6c32293 : SUCCESS in 36m 24s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/38a7068942684e1b94f08d953bc88799 : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/17f5997a0b794a57808a5a1421e574c5 : SUCCESS in 1h 19m 51s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7af3e1e65aa04431a23da41c12ed95b3 : SUCCESS in 1h 49m 01s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/c638b02b13444827a9e92ad8f9884883 : SUCCESS in 24m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2a66b27e7f3646eea59546035764c844 : SUCCESS in 19m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/add8a3151b9c4ca3872d90839ec0bf6b : SUCCESS in 44m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d4c1eff3bd6d47f4a40b549d1429e76838a8478c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 09:18:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12f153baa45599f0450d32f30a4cb5cf3e054509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 11:09:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4e429be8cf2b4451a03e4cb82b0437cc : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a312f9eebbf4d93ac36c9d331e5b854 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/429680809b294cf3a30ef70548ab07c5 : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dfb6097e47a444659810bd77abe965bb : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946dcce8814a4a54bef48fe18635890d : SUCCESS in 8m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/35b43f5ae88a43aa959912925d9fc32e : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/047a749397b44a5d83b49b1e20a1040f : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/88cd8ffd16d14d978ada08c7fa85643a : SUCCESS in 35m 46s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/6bde80345b454e86b08e5a86c4e8f3de : SUCCESS in 1h 24m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/93e81b0310954fa1914db6aa06f98329 : SUCCESS in 1h 19m 30s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f31707d24cde47a8a53bacaea2885db3 : SUCCESS in 1h 47m 17s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/6aea1f4235c84322a25cf2cf0557262c : SUCCESS in 24m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6a88f010553f4df88c466e4be26f819c : SUCCESS in 19m 44s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/cb96b46431af41b4b297f56da550b54b : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2ab22a6feaa105849f4b74984b761dd801dba51a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 11:09:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63bc9e7544d0b8201e3eb6f6b7fb01b429130b32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 11:13:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af253df46fde4975808473e2aa4fe29e : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"359d2b0d580f05fb10eaddb88ad6707476c5450a","revisions":{"359d2b0d580f05fb10eaddb88ad6707476c5450a":{"kind":"REWORK","_number":1,"created":"2020-04-15 19:34:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/720278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/720278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/720278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/720278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/720278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/720278/1"}}},"commit":{"parents":[{"commit":"5985e0ef315bdd9783fb019bfe510c9ac414d130","subject":"Merge \"CentOS 8: Support DNF\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5985e0ef315bdd9783fb019bfe510c9ac414d130"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 16:46:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-15 19:34:10.000000000","tz":60},"subject":"CentOS 8: seed VM \u0026 bifrost","message":"CentOS 8: seed VM \u0026 bifrost\n\n* Install iptables on seed for SNAT - missing on CentOS 8\n* Fix provider network MTU lookup for empty string\n* Bump stackhpc.libvirt-host to 1.7.0 for CentOS 8 support\n* Bump stackhpc.libvirt-vm to 1.13.0 for CentOS 8 support\n* Bump jriguera.configdrive for Python 3 support\n\nThe following changes from the master version of this patch are not\napplied here. These should be applied through configuration:\n\n* Change default seed VM image to CentOS 8\n* Change default bifrost deploy image to CentOS 8\n* Workaround DIB bug\n  https://bugs.launchpad.net/diskimage-builder/+bug/1866847 by setting\n  DIB_DISABLE_KERNEL_CLEANUP to 1\n\nChange-Id: Ie0edf6a924a914395c6502e2d5cf1139bce14a48\nStory: 2006574\nTask: 39000\n(cherry picked from commit 51b84b6001e90435aa76aa09a763edf8063ed085)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/359d2b0d580f05fb10eaddb88ad6707476c5450a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/359d2b0d580f05fb10eaddb88ad6707476c5450a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
