)]}'
{"id":"openstack%2Fkayobe~720818","triplet_id":"openstack%2Fkayobe~master~I368f6edafed9779658798fc342116b4c1b3ffd48","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I368f6edafed9779658798fc342116b4c1b3ffd48","subject":"Remove support for CentOS 7 and Python 2","status":"MERGED","created":"2020-04-17 19:02:24.000000000","updated":"2020-05-28 15:59:00.000000000","submitted":"2020-05-28 15:54:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720818-1590681266795-649d007f","meta_rev_id":"f8a4fd2e36acbaaa1fcf9022d4d9f294b5aec5f7","_number":720818,"virtual_id_number":720818,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 15:54:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-05-28 09:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-05-28 09:40:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-28 14:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-28 14:06:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":1,"date":"2020-05-28 09:31:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 16:27:15.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-19 10:59:54.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-05-19 11:33:03.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2020-05-28 09:34:22.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-05-28 09:40:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-05-28 14:06:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-05-28 15:54:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dce26f5fa73a16f2a7de678150d729b2d286602e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-17 19:02:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54a54833d38aa166b7ee82ee99345e75f7d69d5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:43:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52d7bb52dd154648a2fe09e7bf940041 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d35f0ceffc84a1cae8b2dcf715367de : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a51958ec1ac14093b7978e45033d4d73 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b43ceed83275469299c526d218fb42b2 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae4fc7a8ee3346fc9e3fbcd0b3a3aafd : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7eb78cd122f44079d8a8138d50e0ffd : FAILURE in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec391f55a21f45efb8d794682ee98d47 : SUCCESS in 4m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15082a2f1c8c4f2eba0566d57b55f077 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e825f8910204423b9c5862542a017dee : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eaefe134f65c439ba97120bd64980c0f : SUCCESS in 28m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/52c9c319e67e4b81b7bfe22876af439a : FAILURE in 6m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8ef270a5c72c4031904813c599db4d8b : FAILURE in 1h 19m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/26b48ee4d92b404aaacd005ea5958d0f : FAILURE in 8m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1e8f4b365cb447848d524e578f60a15f : FAILURE in 24m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3cd19eefd223ccbbea4a939642f9f49aa8d67654","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-17 21:12:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"289efb5558379b289e551ffe1703dff5cbaeb8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 23:17:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f1c1d5f135044049a92bef493044189 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48af94cd20c44bd6a0a48d62f243dbda : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2d7d64d490d426fae815e70cfe93ab2 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94e096e3a8894f3cb90aa3c4d5ec0bf2 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/237e1cd596664987a29d124fc8bf4223 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343571884bcc44c3b9d2ff54734b0668 : FAILURE in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20f2387323be438ab5ed88762fc46ea2 : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ccdbe26f32d4e2699e4fcba86db852c : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57c3062729d94256b47766939d754831 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/67bf1f3ff9bc47679ab8f15cc8b904da : SUCCESS in 36m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ed84169163424caaa0fe8be923052306 : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/da4aafc03170457790b97aa6b7614581 : SUCCESS in 1h 58m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9b81369eeff34d23b9f841a7107f3c9a : SUCCESS in 18m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/862c0d089cec49d9a94bc14482f2525a : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c695e85dc82b5ebc6b409331ad94164ea26e8d16","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-20 14:53:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"333fc2ad9db6606eba52ce18ca6823ef0890560c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-20 14:54:24.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a9c65361ddd94b83594735c381a2d62515e0cbd4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-20 18:09:14.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"bc1c284eac5c73c05cd88239e3e396c23f4bdb7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 20:31:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be6ff466de5340ebb1acb5da8b4f139c : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beef284b18ed4ee686d497fa0e6cf098 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f694081a380463397365fda805bd74c : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1815610c3bb0469286bd4b0ebe4b42a1 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ce6207c61ad46b2bec2003bc787a143 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ba9446fde54f7c8583d7fd8cecd86a : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bea0d541f624548b15a2320ac34f8cb : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f67a7bb76bc743b283a81bc9378ff8ec : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b59add59e762452e893ef09f77a2709a : SUCCESS in 9m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/87e68c475e204260a6b406a2c30bdcb5 : SUCCESS in 35m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9282c731eb674dd5aa81cbfe41aa637d : SUCCESS in 1h 28m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/097b0c82a0a04c249d2325433f7fb9af : SUCCESS in 2h 12m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4abe1adac11a420b8324d8cd8313955c : SUCCESS in 22m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db8de2d3ccd54131973bdb084d396382 : SUCCESS in 29m 30s","accounts_in_message":[],"_revision_number":5},{"id":"218811c1514daa189ec6f9e336a50ded1224f10f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-21 09:20:04.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"b1c6ce9188f4c33cbe574db1a183b2ff56069ef6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-24 16:57:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b0215b362fb4473548c9f3ef125b33011062bdab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-24 16:58:51.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"87e805b8b359ea93b9a53f5c6a0880b2e1ae5e1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:59:35.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25a3b4e9bbc941f6b1f8ef4c834b07f6 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b526b6601542beb3cadc7e18c0ff5e : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23408bfe4bd343b692c6ebd0a32ff8e7 : SUCCESS in 8m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be9c720cdca74ad4aea28d46b0d37fce : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c5eb92f90b9403d9bf19e7cc64e5562 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79def9f5172d4a74a5f722d6434eaeac : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37dea36917c54b608e0cc0fa1433f4ea : SUCCESS in 5m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db396c0101264f3d9e6241ccb5e873a2 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d767277766c0465fad7361637ace9275 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cbb45a95d96943e683b33fa6cd602827 : SUCCESS in 30m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cb4da3b55aa445379cbeb8c79275c9e1 : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fc6130a7c120467c95b00d81c6f4df35 : FAILURE in 18m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8d08def7e2974150bb1e70796f0ac9ee : SUCCESS in 1h 54m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/eb82241fa0b642e9a51b3ddd4354eef6 : SUCCESS in 19m 25s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/892f8e3fb74e462da8de118d1a5fdc22 : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":6},{"id":"e7270f4bb537abe04a06da78432c21e99ec27e77","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-24 19:20:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8aef9130979d36ec165cf704fb347273bc0066a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-24 19:21:41.000000000","message":"Patch Set 7:\n\nNew host configure test caught an issue, woop. We lost timezone support by removing ntp.yml.","accounts_in_message":[],"_revision_number":7},{"id":"09449eb731202735343c38527bb80931373e1c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:02:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36876ba5531c4975b518ea6d84472b07 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eae0c79ee444ac1bb08d9ad4a30f663 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed41c1d7f7954c77a011452a3349aad8 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0e4d9b447dc498e902891a44280cfaf : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/605fd8ec92084bbc8f7670f50227c8d4 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca99a3ebb63412ea571da4b635ec0e9 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/157ae57d228f4d1dbbbc90bab4a0aebd : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5c0340896ff840f78f93a9c6c57d6937 : SUCCESS in 7m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a989718d013a4a6f88539cd77bfebc57 : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/213564fcd7144f59b231de165c544054 : SUCCESS in 36m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/79866974683747ed94e59aad3eb5b690 : SUCCESS in 1h 10m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/73ba93bd38ae42018f1a56d1b8ff40a4 : SUCCESS in 19m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/82f8cf3ea59d440ba82564752ada0592 : SUCCESS in 1h 53m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/190c6b914f3e4f1baeb78ac9e690cb9d : SUCCESS in 20m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f2518f5557774bb7b63a4c4997406b5b : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":7},{"id":"10f3dee15a2dadeb720a3fefc1df00172352ef2e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:05:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a1c8d7295f4125d171ee08f33790dd6bc29d5413","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:05:35.000000000","message":"Patch Set 7:\n\nFixed merge conflicts","accounts_in_message":[],"_revision_number":7},{"id":"5a3d11f9658786ad272e45ec5c2224f28e4736be","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:06:09.000000000","message":"Patch Set 8: Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"e58f6b5fd3d6a1f99a33e7e21ff2d3fd2ec77709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 16:10:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5c6532f129243ddabe9b998aafacb73 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384b2ccd7df14127a92a560578e07a0e : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f30492428d8249a193d79d286fc88977 : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/833e90d8e1e54f9a9be94a027880a6ad : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ac8a24488ab4c5f978a768e266c7356 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac580a0b1c345a29476fee65d53eeb7 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0ddfdf4b03e49b49772b183f4831616 : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7532842fd0f04fcd95055cadd0214d67 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d10ad86fbb754ff0828df51b6e804aad : SUCCESS in 10m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e34819f730c94e5dae54e551efab76e8 : SUCCESS in 38m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3339095a7ec54f5580fd793e39f25dbf : SUCCESS in 1h 10m 29s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c108b32fb7fe4e98a2fa313cf05f61d6 : SUCCESS in 17m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e534263bf60e4599ae4d045616680e1b : SUCCESS in 1h 57m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5cd405d148e840f78db31a5d7db91246 : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fc2dd2009ffa4bdbb26a8f9dbd0d6be2 : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":8},{"id":"bcd22b4ca3d353d52921f9867500b08f942b57f1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-28 21:01:42.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\nImpressive cleanup. Some comments focused on docs / release notes. Can be in a follow-up patch to lessen the CI load, in which case I\u0027ll change my review to +2:\n\n- we should say explicitly that we are dropping support for configuring any kind of NTP client on the seed hypervisor. Unless you plan on adding a host chronyd or timesyncd playbook just for it?\n- some yum code left in Dockerfile.j2 files under ansible/roles/kolla-openstack/molecule/\n- a yum reference left in doc/source/administration/seed.rst","accounts_in_message":[],"_revision_number":8},{"id":"a0e0c4cefaf8b3190bc2965814508417e864eec2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 10:02:14.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"91bb807a19c66a2b8a9b4a3fff7bc06dd03cb9bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 10:02:44.000000000","message":"Patch Set 8:\n\n\u003e (1 comment)\n \u003e \n \u003e Impressive cleanup. Some comments focused on docs / release notes.\n \u003e Can be in a follow-up patch to lessen the CI load, in which case\n \u003e I\u0027ll change my review to +2:\n \u003e \n \u003e - we should say explicitly that we are dropping support for\n \u003e configuring any kind of NTP client on the seed hypervisor. Unless\n \u003e you plan on adding a host chronyd or timesyncd playbook just for\n \u003e it?\n \u003e - some yum code left in Dockerfile.j2 files under ansible/roles/kolla-openstack/molecule/\n \u003e - a yum reference left in doc/source/administration/seed.rst\n\nThanks for the review. The dropping of NTP on the seed \u0026 seed hypervisor is a bit of an issue. It\u0027s at least clearer in release notes for now.","accounts_in_message":[],"_revision_number":8},{"id":"b65b4efd7e386957da63c8dbd5a98d88bdc10fac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 10:31:29.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"68815061adfc50890cae5258c3e97f25d6d73157","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 10:33:59.000000000","message":"Patch Set 10: Code-Review+2\n\nJust added a missing word to NTP release note","accounts_in_message":[],"_revision_number":10},{"id":"b6bae8276926bddc7b38dd330b02aa23630ee2cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 12:29:05.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac7bc91726954558b1274034b6ed6d89 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0289768618674a76b0c088aa71b10e64 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cd17abd63bf466688952b6b37c0e818 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0edbef776434709ac4853f2d630dcc5 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e66bf6cce4c14a4598f7738c41fdd824 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0fc5d3ffa648aca7c84bc8d7055a1c : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f50f49ce250480cb8940130dd698124 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36ef51ea1ba34fa4929cb9e5f5491634 : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4844db56137447b9a92acf98d52b1fad : SUCCESS in 8m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a8abbeb7d8b46a4b999506285c6e73a : SUCCESS in 34m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c182fd0bcc39406eb103f70d3f2dc680 : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c2a77cabb68142b2b963a61db6d21192 : SUCCESS in 17m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/011055fc94f4487e9e1cf6b6d621af18 : SUCCESS in 1h 55m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bfa36fccde2247c9b91b382c5ec2f4eb : SUCCESS in 22m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f1291eb7e9524d7380fea6f2ee3b4427 : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":10},{"id":"18553b54167521b6645b75a4da30ca8d472d5b28","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-13 16:27:00.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"fe237b1779147f127a86974352b4d4bc4b03a802","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-13 16:27:15.000000000","message":"Patch Set 10: -Code-Review","accounts_in_message":[],"_revision_number":10},{"id":"be1009e0cabb29a2726050c71b3c6aabc495f745","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 08:12:51.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"01c5c5807b7758bdb442b483aeb84f021ca89f47","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-19 08:31:59.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)\n\nLooks good, waiting for CI","accounts_in_message":[],"_revision_number":11},{"id":"d96cd9848fe99e8bcbe28dec6a70b76591376903","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 08:40:11.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ade03ff11d1f872dafc34c56c5ee1e542385e77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 10:06:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d085fd99f50d4e2db19177d4720fe648 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd7f118b4c4464dbc6b06c0e3ff7162 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1182a4498ad04094a1aec6c264987002 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77deb2310ff54697bee19915b4dcb813 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ce652cacd3043f89110dc8505b543d9 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80566c2c86404268a777523e932a29cd : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdbf02e691fa4dafa57e45983f912ff0 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82ec60bbb3e84440b02d621572ba36ef : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eff2f590ffd143c598185481b22acd2a : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f596ccc48c5488682382eb03d9deead : FAILURE in 8m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9795ec37d1314bf08c985ad7b92ead2c : SUCCESS in 1h 10m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cbb0f0675e32470887b1885c52e726f0 : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e2d3606ce66a4a7582d4a065d95bc1a3 : SUCCESS in 1h 51m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/45f5f236088e4812bbecbd8191e33334 : SUCCESS in 19m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ee5549fdfa734aef9722b02aa7574972 : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":11},{"id":"1e8a0699ba5191f315911630e05542c40c82e89c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-19 10:59:54.000000000","message":"Patch Set 11: -Code-Review\n\nLooks like `ERROR: molecule 2.22 has requirement flake8\u003e\u003d3.6.0, but you\u0027ll have flake8 2.5.5 which is incompatible.`?","accounts_in_message":[],"_revision_number":11},{"id":"ff40d095eb5a6bc408a300c0faa2763d67d8d9e5","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-19 11:31:54.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"17fc8879c30580c481378b897170e1d1b2ff9da5","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-19 11:33:03.000000000","message":"Patch Set 11:\n\nAre you planning to move from deprecated network-scripts to NetworkManager too?","accounts_in_message":[],"_revision_number":11},{"id":"ba2d07ba8f21a3ef91b545b4ae5c4553e3ae7d48","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-19 12:45:38.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9e3df12cdee7daa7592d62dcbdcc44baa2117c3d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 12:53:00.000000000","message":"Patch Set 11:\n\n\u003e Are you planning to move from deprecated network-scripts to\n \u003e NetworkManager too?\n\nI\u0027m not planning to work on it. If we find problems we can\u0027t resolve then we might have to, but it\u0027s quite late in the release for a big change like that.","accounts_in_message":[],"_revision_number":11},{"id":"0fd62b441b48e4bf890fe102c1fbb7f3062c128a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-19 13:07:52.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"a9135974f7a8bffddb7cce6bd1382e2664d3c407","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 13:08:14.000000000","message":"Patch Set 11: Workflow-1\n\nNeeds a fix","accounts_in_message":[],"_revision_number":11},{"id":"27931c3d214460a549c11bb114f877f318389610","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 13:33:05.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a7824203eb5e0ff829aab0d7f46607ce05b06cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 15:38:11.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a03f312859d417394001d1436c64ee5 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9368457ec5a4d5c8807419d1e048831 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef92d6b3393b4d5baa5453eff0a40348 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ac1867e620c49afb8a12ffda83fa066 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/135e986ccf7545ea833067d8fa944840 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011a12bdcabc4b97b552060d772f1b1b : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f3d9762ee874eea85355097517a0645 : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85e881a138114e21ae77d8d3b08c5bcf : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d5e2f335c359412abe4b33e3796d1db8 : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa8ae1f5e82e42da8eccd5d964bcd665 : SUCCESS in 35m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/579b0d175c354da7be2caf84e0a0e31f : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4cc7936f513e4d99a0648dcf8a375cb6 : SUCCESS in 18m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/df9d678e1d99439aaa5c622a3814abf1 : SUCCESS in 2h 02m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8c130c39fd4842918b3a3936960e81c8 : SUCCESS in 18m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8afd0bce177d4e9e8b269753681cc046 : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":12},{"id":"7633c3562bde54253de383b73d73fd490faa3016","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-20 06:46:05.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)\n\nMinor comments, can be fixed in a follow-up if you prefer to merge this first.","accounts_in_message":[],"_revision_number":12},{"id":"20a25940ce7306cc421e820c66f5752c05921bb5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-20 11:40:55.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"fdf5fbc56930f10197d04fa5427719c17b38816f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-20 12:06:45.000000000","message":"Patch Set 13:\n\n\u003e (2 comments)\n \u003e \n \u003e Minor comments, can be fixed in a follow-up if you prefer to merge\n \u003e this first.\n\nFollowed up: https://review.opendev.org/#/c/729571/","accounts_in_message":[],"_revision_number":13},{"id":"8f6bdc0608477e6e3a76b46e7ae35cac25a95ee4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-20 13:25:09.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"a63e1eb604a5c00680c62dc89acb1e4179957d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 13:38:35.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3d94e8cae1e4fcf91128f98bcf49c95 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed14e4d437c435086636bbcfcbd1796 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cc227f4bb394680a28380c5bb84395f : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f56089718c748de8c61fc288bd67a56 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97506a7625c446ccb415118a9476e897 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36397b77d22d4561a6be12b4863226fb : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dc2d2035e224692ad44cf8cecfab717 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d309fc4069ff4bf8b8e73bd3defe1242 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be5cd98f4e594d98aed40ff23eca527f : SUCCESS in 10m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/561a400bee8449fdbd7f26761db5e117 : SUCCESS in 35m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1a921414355346569cf044c1ba4d8512 : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d1124d4823de4e87b604813755e95a86 : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c8df7bd2d7634c72a582fb068bf8c4f7 : SUCCESS in 1h 52m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/72ace604e56f44d9bf37329f805f5c6e : SUCCESS in 18m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3e60bf9cd8724a059ce0f2b1d1077437 : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":13},{"id":"e6ce51634ef3de96dfb282cf9d7348ff7e9e6d62","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-20 14:01:43.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"bdcdaffe1aa3cd10a411ae21a45849f3aff8db2e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-20 15:55:14.000000000","message":"Patch Set 13: Code-Review+2\n\nOK, it was no biggy anyway :)","accounts_in_message":[],"_revision_number":13},{"id":"80172f1755db338963061f88bbbb9e84fcbbe60f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 09:26:02.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f1abadd74959a453224ac7242d5cc0be9bd03f3e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 09:31:45.000000000","message":"Patch Set 14: Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"060038b0caf6823a12b8fd9598ea00fd997c0ef3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-28 09:34:22.000000000","message":"Patch Set 14: Code-Review+2\n\nJust looked at the changes since my last review. They all looked sensible.","accounts_in_message":[],"_revision_number":14},{"id":"ff95c5f52ae651cbbf6e21bb3fc7470afc6f32b2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-28 09:40:49.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"6a22016e610b3de70160e565e6c7ae2ba070c549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 11:58:29.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35ebfda27e8a48f2a8099ce41b79fd3c : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb420ca5e5b430f9711d56df77cf98b : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b71b8193c5534004872603e764e8dfde : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab1b7cad185a4db6932a7310d7abfc8e : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cdb8436965547e49e7aaff38618f29f : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76b6237541424d86bcbe73eb2a36adff : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e92deef0891f42809a12f86c485cbad0 : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36e9de306be547629949cc4f2457ef63 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d33f294eb7ed49ef99249cf665e25fd5 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d0c66a5e3194230a39f34fac68ef391 : SUCCESS in 29m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7a50beb89399437a9a7a6f3782965931 : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/efc09258983f43a794eee88b26bcfad2 : SUCCESS in 16m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/86e7f1bcb6c643d39e72e7c5cd0a2f8d : SUCCESS in 2h 30m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2f3d62fc86544e1b910566d66cbc0ccf : SUCCESS in 17m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1ba4999f7fa6445496c1c6fbb8b6dee8 : FAILURE in 42m 37s","accounts_in_message":[],"_revision_number":14},{"id":"61bf6fe9fcc9027efe0a6f8164fc40c1aa68df01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 12:08:37.000000000","message":"Patch Set 14:\n\nrecheck\n\nnetwork/opendev issues during seed upgrade","accounts_in_message":[],"_revision_number":14},{"id":"86fffdf50c9b2b0a95c48d6554b836fd1efe01bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 13:36:57.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c43f4e5b95540009a9ac12709c8c3c9 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a292fdfb954b7ba0fd4edbd34865ca : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96accfc443604da7b16bc1f7f5add0f8 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95085fdea4d14e668bf7d56fbf676bda : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d272189e836d4a80861e52274340bbe8 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beddb4f46c054427bc2825eac1fbe226 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38e2c50d22134eac8fd4d21859b07f7e : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/05cccc45b452480eb74ccde826792a56 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/30a6d3c1d94946e18c6d20027cc45ef9 : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a120f00f3aa4600a7b348a018c6e5cd : SUCCESS in 37m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/96320968a8f941bd8c051bf35428ef0d : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a03a99c68912466bbd16b8d7b9eb92a2 : SUCCESS in 17m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c82dce1b090b4f4fa2abcf3613e12690 : SUCCESS in 1h 19m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e8474bfa754449e9b36e8aec89789250 : SUCCESS in 18m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9633ddce737146b99d78df24a4a3a58d : SUCCESS in 29m 24s","accounts_in_message":[],"_revision_number":14},{"id":"ee93f2926a498eb00b318b5f05ca1d45a59d6681","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-28 14:06:54.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":14},{"id":"8c560a21787758795a53bde3101ac3ef9a5453eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:07:09.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"c5120514846e8f72e8401f883703155edc0a2092","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:54:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"1a77191d59b4e732dfdee3baeb871542e84cf73b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:54:26.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b222ef1dbad44d63ae1112e1c5fe70b3 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89b3ef2e690a4342b57ee85bca6c73a9 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ead4a86f6bef4993a1f65a00045366e8 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc3e301c902490886b9ff257e08da0c : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c955f89305b4ca7aaa7c4dcf71c8b68 : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ab25c71950d44c891f8941aa2487e42 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/072a97a38d584686bac609d3132d12c3 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25c9613006a246d4b253b89d930dba76 : SUCCESS in 37m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/55679fb0b1c74b38a4475c3ba6c91ef4 : SUCCESS in 1h 06m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b2ae1cf43a754636bd8aec9c07623a6e : SUCCESS in 15m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8632e0f61d3e491a90cf73185ba38249 : SUCCESS in 1h 45m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6fdb55cc4c014edd82ba1f659e4e2faa : SUCCESS in 17m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c2f68a2445394f85bfdd9ea1e5db7e43 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":14},{"id":"f8a4fd2e36acbaaa1fcf9022d4d9f294b5aec5f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:59:00.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c7b9809f2e407f90969d4b1fa42c9a : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be4ed54e93134401a29cffb0a674cca3 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"b9d76f6ef5e9ada68df47a417a50708a262005b1","revisions":{"f757275265869a4885f6f19bb6fc217114e4b8a7":{"kind":"REWORK","_number":1,"created":"2020-04-17 19:02:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/1"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"subject":"WIP: Revove support for CentOS 7 and Python 2","message":"WIP: Revove support for CentOS 7 and Python 2\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f757275265869a4885f6f19bb6fc217114e4b8a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f757275265869a4885f6f19bb6fc217114e4b8a7"}]},"branch":"refs/heads/master"},"57b8ede07ba4f7ae636fcc26edfce7923491ce37":{"kind":"REWORK","_number":2,"created":"2020-04-17 21:12:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/2"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 21:11:59.000000000","tz":60},"subject":"WIP: Revove support for CentOS 7 and Python 2","message":"WIP: Revove support for CentOS 7 and Python 2\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57b8ede07ba4f7ae636fcc26edfce7923491ce37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57b8ede07ba4f7ae636fcc26edfce7923491ce37"}]},"branch":"refs/heads/master"},"c63d35443812b5d458f1d96699e28d0eab5129b9":{"kind":"REWORK","_number":3,"created":"2020-04-20 14:53:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/3"}}},"commit":{"parents":[{"commit":"348ee0b5d34adfaaa83d3de6d9b1341c344bea56","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/348ee0b5d34adfaaa83d3de6d9b1341c344bea56"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-20 14:53:06.000000000","tz":60},"subject":"Revove support for CentOS 7 and Python 2","message":"Revove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Removes the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c63d35443812b5d458f1d96699e28d0eab5129b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c63d35443812b5d458f1d96699e28d0eab5129b9"}]},"branch":"refs/heads/master"},"b92d070fa8568ab599bd27bf953aff6a38684cf3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-20 14:54:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/4"}}},"commit":{"parents":[{"commit":"348ee0b5d34adfaaa83d3de6d9b1341c344bea56","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/348ee0b5d34adfaaa83d3de6d9b1341c344bea56"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-20 14:54:07.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Removes the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b92d070fa8568ab599bd27bf953aff6a38684cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b92d070fa8568ab599bd27bf953aff6a38684cf3"}]},"branch":"refs/heads/master"},"296b99df57df3a97ba95f7069113306c0609267e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-20 18:09:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/5"}}},"commit":{"parents":[{"commit":"83bc574ba525cba8e16a82c9fac9a7d6ee1c38b7","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/83bc574ba525cba8e16a82c9fac9a7d6ee1c38b7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-20 18:09:14.000000000","tz":0},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Removes the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/296b99df57df3a97ba95f7069113306c0609267e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/296b99df57df3a97ba95f7069113306c0609267e"}]},"branch":"refs/heads/master"},"f2ce6302e96a86fe07456e08c00c6a7a9fc94d9f":{"kind":"REWORK","_number":6,"created":"2020-04-24 16:57:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/6"}}},"commit":{"parents":[{"commit":"558276a8a6af4a7375bf58afd7ab033e11a93296","subject":"Merge \"CI: Add overcloud host configure job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/558276a8a6af4a7375bf58afd7ab033e11a93296"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 16:57:36.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Removes the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2ce6302e96a86fe07456e08c00c6a7a9fc94d9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2ce6302e96a86fe07456e08c00c6a7a9fc94d9f"}]},"branch":"refs/heads/master"},"78c8a70c5d61dcd0924298050329f58b944f9093":{"kind":"REWORK","_number":7,"created":"2020-04-24 19:20:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/7"}}},"commit":{"parents":[{"commit":"558276a8a6af4a7375bf58afd7ab033e11a93296","subject":"Merge \"CI: Add overcloud host configure job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/558276a8a6af4a7375bf58afd7ab033e11a93296"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 19:20:30.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Removes the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78c8a70c5d61dcd0924298050329f58b944f9093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78c8a70c5d61dcd0924298050329f58b944f9093"}]},"branch":"refs/heads/master"},"658731fa1aa96c4f9ef8f946cd9113bd0a25d948":{"kind":"REWORK","_number":8,"created":"2020-04-27 14:05:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/8"}}},"commit":{"parents":[{"commit":"5cfca2aa82dc33f86fe29af9f2b123430e6bccf4","subject":"Merge \"Run kolla-ansible bootstrap-servers as kolla user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cfca2aa82dc33f86fe29af9f2b123430e6bccf4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-27 14:05:09.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Removes the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/658731fa1aa96c4f9ef8f946cd9113bd0a25d948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/658731fa1aa96c4f9ef8f946cd9113bd0a25d948"}]},"branch":"refs/heads/master"},"4c785437d2972f8543d3f5b2f60b2a435113f51b":{"kind":"REWORK","_number":9,"created":"2020-04-30 10:02:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/9"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-30 10:01:58.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Remove the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c785437d2972f8543d3f5b2f60b2a435113f51b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c785437d2972f8543d3f5b2f60b2a435113f51b"}]},"branch":"refs/heads/master"},"9e7e6e0a7f6a1dc1c1d5f68cd658a51c56cc5d10":{"kind":"REWORK","_number":10,"created":"2020-04-30 10:31:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/720818/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/10"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-30 10:31:12.000000000","tz":0},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Remove the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e7e6e0a7f6a1dc1c1d5f68cd658a51c56cc5d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e7e6e0a7f6a1dc1c1d5f68cd658a51c56cc5d10"}]},"branch":"refs/heads/master"},"589f317eb0057dbd0fd0fb03904cc019c91df6fe":{"kind":"REWORK","_number":11,"created":"2020-05-19 08:12:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/11"}}},"commit":{"parents":[{"commit":"fc017249c23cb819fd44202e2cff80d563cab8ee","subject":"Install coreutils package before using configdrive role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc017249c23cb819fd44202e2cff80d563cab8ee"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-19 08:12:39.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Remove the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/589f317eb0057dbd0fd0fb03904cc019c91df6fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/589f317eb0057dbd0fd0fb03904cc019c91df6fe"}]},"branch":"refs/heads/master"},"38f1425d35bd2f6126123e0865542f20e684f0f3":{"kind":"REWORK","_number":12,"created":"2020-05-19 13:33:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/12"}}},"commit":{"parents":[{"commit":"fc017249c23cb819fd44202e2cff80d563cab8ee","subject":"Install coreutils package before using configdrive role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc017249c23cb819fd44202e2cff80d563cab8ee"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-19 13:32:45.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Remove the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38f1425d35bd2f6126123e0865542f20e684f0f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38f1425d35bd2f6126123e0865542f20e684f0f3"}]},"branch":"refs/heads/master"},"ba41ce861288a8d08bb930e882475c6f9480dd19":{"kind":"REWORK","_number":13,"created":"2020-05-20 11:40:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/13"}}},"commit":{"parents":[{"commit":"e8eb5f4128030cc43b92749dfc40956df6532ded","subject":"Merge \"Upgrade Vagrant VM to CentOS 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8eb5f4128030cc43b92749dfc40956df6532ded"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 10:46:29.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Remove the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba41ce861288a8d08bb930e882475c6f9480dd19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba41ce861288a8d08bb930e882475c6f9480dd19"}]},"branch":"refs/heads/master"},"b9d76f6ef5e9ada68df47a417a50708a262005b1":{"kind":"REWORK","_number":14,"created":"2020-05-28 09:26:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/720818/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/720818/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/720818/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/720818/14"}}},"commit":{"parents":[{"commit":"8fb30208279b1d578dc3291a362e51525964c177","subject":"Merge \"Add ability to run playbooks before and after a kayobe command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fb30208279b1d578dc3291a362e51525964c177"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-17 19:02:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-28 09:25:51.000000000","tz":60},"subject":"Remove support for CentOS 7 and Python 2","message":"Remove support for CentOS 7 and Python 2\n\n* Always use Python 3\n* Drop code paths for CentOS 7\n* Drop support for Yum\n* Remove support for host NTP daemon, always use chrony\n* Switch references from \u0027yum_install_epel\u0027 to \u0027dnf_install_epel\u0027\n* Remove overcloud host image workaround for tagged VLAN admin network\n* Remove the kayobe.utils.yum_install function, which is unused\n\nChange-Id: I368f6edafed9779658798fc342116b4c1b3ffd48\nStory: 2006574\nTask: 39481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9d76f6ef5e9ada68df47a417a50708a262005b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9d76f6ef5e9ada68df47a417a50708a262005b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
