)]}'
{"id":"openstack%2Fkayobe~721982","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Iaf8db47e70f023b446c17aa61fc8bb89cf7c2b28","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Iaf8db47e70f023b446c17aa61fc8bb89cf7c2b28","subject":"Add support for stopping overcloud services","status":"MERGED","created":"2020-04-22 14:22:22.000000000","updated":"2020-04-24 00:10:29.000000000","submitted":"2020-04-24 00:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721982-1587686921460-81340386","meta_rev_id":"7c1c75da8ec61a045ab7fd87452815005089fd86","_number":721982,"virtual_id_number":721982,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 00:08:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-04-23 12:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-22 20:05:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-04-23 12:27:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-04-23 09:21:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 20:05:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-23 12:27:04.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-24 00:08:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4bf7c97b4d2c5fe4a7e9c757680cae9053922ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-22 14:22:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"597896674fc2d4c65daa8a3880f07f2605083dd4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-22 14:22:36.000000000","message":"Patch Set 1:\n\nRequired for CentOS 8 migration.","accounts_in_message":[],"_revision_number":1},{"id":"58990163c6f8491db7549d06f781948be4071bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:10:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/26ed0dbb78bf4f9e97b0b8bfc26ceeec : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fe2297b18cc4714b54053da98e78345 : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d274cd06d90a4e7fa6a64692e6e756c5 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07ff50e5deca4afd9e1970f5a61ae946 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7f5101fe15458baa5b87981cfc4ae5 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30f7a62c443048d98db116229eaa9b10 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92dd9ca44ac42b2bd4b7d757a4f13d2 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/453138da7a1e41bd8705c2966d03a266 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e0d19379c9f1410d8298fd359439cc61 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ad0546eb7d1745ec8ce52f3fb8605bf5 : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e27bdb4125cc4b5ba12cff90d3cb5fb6 : SUCCESS in 30m 23s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/dfd5f7c7b12d407189bcaec2fc2c757e : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b1321346031d47f6971701b8029f4f32 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1d66ed28d08d49eaba8e4933a459680c : FAILURE in 2h 26m 09s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/12432f6e665441cd8820e033342b4c6e : SUCCESS in 22m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/94ab387712ba42e9bfda4ef7abf5c46a : SUCCESS in 19m 13s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/93e6f370f68b41f6ac726dee1bcbb32c : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e31b9334ce128954c1a30e100a16216ae218f21f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-22 17:18:06.000000000","message":"Patch Set 1:\n\nrecheck\n\nFailed to connect to the agent running on node 67afec68-fd77-4cbf-ac8a-f4f4eee96200 for invoking command iscsi.start_iscsi_target.","accounts_in_message":[],"_revision_number":1},{"id":"d97b798140014490cab85564eee408831258d4e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 19:28:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1d9d6a8b953746688a92fb4b96297855 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfbd913aa2d942e29d728c11bbe3ac3f : SUCCESS in 6m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/45ed84f48f984914a80cffb8f83ef81d : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13db328aebc448d099793d42187e54e9 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa99e89617c403686bb7a693706fb87 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f39322732b8c4247af8236173de63de2 : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed54de15923f4eef9a16219cdacddb82 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdac91ace4d9410082e1212fd05fb216 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40026909154144288cd2d37812ad8289 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/844d45e3618444a394784429580c27f4 : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c15a3a1cfadf4e5eb092f450d5499fe1 : SUCCESS in 31m 04s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/306969d461364145af25565cc3e2a63f : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fa47e9195fcc48d6a2c11aa2b4a42291 : SUCCESS in 1h 06m 38s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4cc5dd2cfb6445b8a57b339a1e0f8e12 : SUCCESS in 2h 01m 17s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/f32ec510147e484c92fa71496d9be96c : SUCCESS in 23m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1a4aa4a377f44959bb9e9aa52c3bb6bf : SUCCESS in 17m 23s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/65a49c1a24594d4fa653f4bc041f06cb : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"441f3332e89dcd2b08ce41f93c0619d618d408a0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-22 20:05:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed78bf7b1939320d8ddb69e3bcfd4d4bed5d0f53","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 09:21:01.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"35327facee3c8f9373767c6c2c9c019e78c40960","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-04-23 12:27:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"effe730551264fa4a152978ff6837749bd818b90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 12:27:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d2876b8790b2ce0e931e99704cc3b856ffc0a64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:43:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4523c9363ac944399499c87df625a3a9 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d89decfeac81488a9033de0159aa4059 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9000025d99141548c6ae4f0b4eee51a : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcdb758bfeed45d69eaf796dd05e1f16 : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e30dd2e7a37f49e4b6bb2659da4de05f : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bd7a31ae1e24b07ad502f1192edd216 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd9d7dd406c2403e98304bc4de0c137e : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b330d65740c435abcfb26d0f91b7add : FAILURE in 8m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d851d313d704770b76ce73554fe5c55 : FAILURE in 8m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/58b8843496404a62868ef5bbfccd7f42 : FAILURE in 8m 38s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/0d04823fc9a842609994bda944904484 : FAILURE in 7m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/57dac6fb105a491a8ada5935f6d1cc55 : FAILURE in 7m 24s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0ccfff2d8bb64aaf8a2940693a5468bf : FAILURE in 8m 39s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/c48739dee3d94acdb3746fc0bf7e43ac : FAILURE in 6m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b0f7d35c61b9469aa823a2c658c27dd8 : FAILURE in 7m 51s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/067d34f9f72746128b709581eb24d2cf : FAILURE in 7m 59s","accounts_in_message":[],"_revision_number":1},{"id":"55f0cfef1711ad27346e8c480fe2067e0bd6067c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 19:58:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1cab69d220d87a8c2b4f47510eaa61a2e55f00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:05:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb8039ffdc3e427ab8fbf5a8290bc69d : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e93925127ac42439bdf557938e31c5f : SUCCESS in 8m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ca590ec4e6874c40adaa47075b5eefca : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e35340d22894e13ad07f8de0904b80b : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebba935c96b49dfb7d81429589b1f5a : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03d70cc3c9f74b48a1b0f2cb4ec04db6 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caacff2693424cf1bc2edf516d1b0834 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3efd36303a3243ea85359cc49239c7a8 : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f123f80c08a843ab86ed57ec465706fa : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a56664d1ade46019bf5ee17d9a0e1d4 : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/34148981da1d40a38f3b4f5c208d0b4a : SUCCESS in 29m 15s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/efe191c855914aa5ac922617f94bf793 : SUCCESS in 1h 07m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a1e05d584cb14b5784601e302af1bd83 : SUCCESS in 1h 48m 23s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e27a42b83dd14bdab43634b0d9d7fca1 : SUCCESS in 2h 00m 20s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ee6c2a52ed6043bfbbb4601f7911ee48 : SUCCESS in 22m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f80a6e6be5804752b860d8ecba4750f6 : SUCCESS in 20m 53s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/308284d0413b4eb3829e6da3dfaa95ae : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b9f4b46bbbd90e56b96882832638102dd96a0bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 22:06:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95b7e1cc8dcb6a1dd19d0fbd94da17e0cace2f5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:08:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4791db0c9ce0430baf4912ed668900b7 : SUCCESS in 2m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6964c77f3e834d10b78684819f2eb782 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc44d39912844d70ab9de7d4074f4f6e : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b27960835153446d95ff6fa7aa69d543 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc431b59573a4722a74bc2ce781609bc : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c2df52ecd5346cebfb5fea6042141cc : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8b39f4dca354c4a9f02511c76fa32cc : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29a24fb8a87848568c183b1c92e361c5 : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/921374530a774a4eb1d2953efe3ea353 : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6fdf9234adfa4d948c9736e00530b0cb : SUCCESS in 36m 35s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/af62c7be051d4ed892ffcbf8367436b1 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7332021ab0544a338d6fc48bd69e084f : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ecda2d95693f4711bb0fc005bcdfeaaa : SUCCESS in 1h 59m 31s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/018f651a612e4022b7370126a87083c3 : SUCCESS in 17m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3caaa44828fb4e9194f672217c498b8b : SUCCESS in 17m 22s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/d44536b8a9fd43e8a548449d5737af16 : SUCCESS in 34m 41s","accounts_in_message":[],"_revision_number":1},{"id":"617df635389f5e422848733a47f8fdf935b5b56d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:08:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c1c75da8ec61a045ab7fd87452815005089fd86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 00:10:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d75452297cf4c419e519d68143b0f94 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2b7c44969f549919a33c54186fc6a7b : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"773ac2e4c6476c445b90a0c53779ffff61ab8674","revisions":{"773ac2e4c6476c445b90a0c53779ffff61ab8674":{"kind":"REWORK","_number":1,"created":"2020-04-22 14:22:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/721982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/721982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/721982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/721982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/721982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/721982/1"}}},"commit":{"parents":[{"commit":"23080784c44e3901c7835c56c0183c0200d6029f","subject":"Merge \"Prevent openrc files from using wrong OS_CACERT value\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23080784c44e3901c7835c56c0183c0200d6029f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-26 12:39:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-22 14:22:22.000000000","tz":0},"subject":"Add support for stopping overcloud services","message":"Add support for stopping overcloud services\n\nImplemented via \u0027kolla-ansible stop\u0027.\n\nChange-Id: Iaf8db47e70f023b446c17aa61fc8bb89cf7c2b28\nStory: 2007467\nTask: 39155\n(cherry picked from commit 71c2aa1d90e97403aba67d1a65c1860bd9337b1c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/773ac2e4c6476c445b90a0c53779ffff61ab8674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/773ac2e4c6476c445b90a0c53779ffff61ab8674"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
